)]}'
{"id":"openstack%2Fkayobe~892480","triplet_id":"openstack%2Fkayobe~stable%2F2023.1~I2085f9d1fd457e5b54b17f37a5e725990a60e8bc","project":"openstack/kayobe","branch":"stable/2023.1","topic":"docker_login","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-08-25 12:46:13.000000000","reason":"Change was submitted"},"34940":{"account":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"last_update":"2023-08-25 12:46:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2085f9d1fd457e5b54b17f37a5e725990a60e8bc","subject":"Fix for seed-containers being unable to use password protected registry","status":"MERGED","created":"2023-08-23 10:22:12.000000000","updated":"2023-08-25 12:47:18.000000000","submitted":"2023-08-25 12:46:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892480-docker_login","cherry_pick_of_change":892479,"cherry_pick_of_patch_set":1,"meta_rev_id":"38d8d239da323314d7f146f44f437fe44840e00b","_number":892480,"virtual_id_number":892480,"owner":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-25 12:46:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2023-08-23 10:55:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-08-24 08:34:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2023-08-25 11:01:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2023-08-25 11:01:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-23 10:55:25.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-08-23 12:30:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-24 08:34:15.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-08-25 11:01:42.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"db64f94f0290215b33c8ea4c0eaa023ba6711158","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2023-08-23 10:22:12.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"7cb2fc74d9bfff32e5d925a4137f678642b0d00f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-23 10:53:39.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"4a546449a869d91d170c93d9f6d5dce0325b7a88","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-23 10:55:25.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"391e2375fdd635befc732bdeaec1a6ab039ad61a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-23 10:56:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ede2f821cac511032e9bd6e216afb34b4c1b0d57","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2023-08-23 12:09:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"add1977dcbaf1b5ed2dd8a0e449466f971fd4f25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 12:30:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b78ea287474d4798a16ad6e510e0eb9d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19f1b6e74c3841d6948f702ddb65f3cf : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab0a9479b4df46a6ba40de94e6e47c42 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81d4c9d05f654e53864d91ecf9dcaa24 : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0291c1551fb4455da99f97309f61d3ff : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fe92692b58c43efb33fef53762934ed : SUCCESS in 8m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8d930cd22254db5915cb806bf8694b1 : SUCCESS in 3m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/726e827468a045598967b595c4c56ca2 : SUCCESS in 4m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/705a6fea19d04c8e88036dfd04d441e9 : SUCCESS in 10m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1c3a9cb8ad2742df93ca6884f25d0318 : SUCCESS in 16m 34s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0cf0dbe0985541e7b92d98f54bcd09e2 : SUCCESS in 55m 19s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/864ea88b497440558ac68f963f2529ca : SUCCESS in 55m 30s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b783f8e3576e426fbcdc40138db312d2 : SUCCESS in 44m 23s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/2203c2ebd2294c8481d8160205f5344e : SUCCESS in 52m 38s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5c2057aea71740ad9deb7febd65be3ba : SUCCESS in 15m 26s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/17a44fa9d85947c280a8570fedc8525a : SUCCESS in 16m 22s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/43b7256634d6495d9ff7017439942684 : SUCCESS in 14m 27s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3f0dc12cb6a64b30b1ab6f8846b66508 : SUCCESS in 1h 20m 19s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ecb7045276b40158ce89137c9300b51 : SUCCESS in 1h 31m 24s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5f123b71594540738fd048b99675010b : SUCCESS in 18m 00s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3eb5f6e3d74342b08f965527132b19f0 : SUCCESS in 12m 51s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f806a147c4104fcba48286e07276ff6e : SUCCESS in 38m 42s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b9b5f08835364abaaf7132c03d55e1f8 : SUCCESS in 27m 26s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/32abe7ca6c6c4b11b9963d8813d3becc : SUCCESS in 31m 10s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e3d9a14a71004ef981223766d4dcc62e : SUCCESS in 14m 07s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1cdcdaade45240cfad82a77ba19d5dbf : SUCCESS in 10m 29s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2c74dad7169340ffbc1b1a718b749fa6 : SUCCESS in 33m 33s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b68e15c93784406885f97e93ace4a6f : SUCCESS in 27m 31s","accounts_in_message":[],"_revision_number":2},{"id":"a51773c835a022d2f3719c90b0cee1dbf5e48381","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-08-24 08:34:15.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9b5c40c18ddd82261b976b096481ad88fa1bfc30","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-08-25 11:01:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cb029d6c101a406d3d4aba1aa86abcf4786c5981","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 11:02:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ff4021dd456ecb6089a48d638382229d393957c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 12:46:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df974054982843209b83312deb531bb6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52a9ce7bc6b54106b8f19760db4f5059 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/393abe2f9ea949dc822982661804e1cb : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf473338d9724fcebea290e341936544 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5092cd05be5473eb865e880cadfab8e : SUCCESS in 8m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d07e482177c941bdbfd76779fa0afbbd : SUCCESS in 3m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c19581ea31324aab955f598403cea756 : SUCCESS in 4m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/51c7b30bd45045f7b15e4f444322bd8d : SUCCESS in 11m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/51fc6a295677409d90072a47424136ac : SUCCESS in 44m 40s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/6a7a3176392446779b592e31d0735a43 : SUCCESS in 58m 27s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6296cf53f259405c9db1d3040590e851 : SUCCESS in 55m 11s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/b66fe5789145453da0c2cb7603594a7c : SUCCESS in 1h 03m 06s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6afdcc70ffaf414fba02a0191193365f : SUCCESS in 16m 28s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/721d259ed7964171b893339b7b092486 : SUCCESS in 10m 54s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a73525df30134c01842edeb0cc1b4a02 : SUCCESS in 1h 37m 46s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/138c230005434cfc868745280cd536e0 : SUCCESS in 1h 35m 19s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7058e5cb650345af857c629ace1e7590 : SUCCESS in 16m 58s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be657d6e103e47b48c8472dc3739823d : SUCCESS in 12m 26s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a4fcc5580b0e4a1fb18e24d96d94a26e : SUCCESS in 28m 53s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/17da32fce6934a0faff7fb66888919d0 : SUCCESS in 25m 56s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d803a984a21d4a66b925b097137bb757 : SUCCESS in 13m 32s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/23e4ce5e00ad4219bbcd6fbd14240770 : SUCCESS in 10m 02s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f3cea0619ac04f369a6af9d53a5ba182 : SUCCESS in 25m 52s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a2cea2651ace48e3afad8292ff0a3339 : SUCCESS in 28m 45s","accounts_in_message":[],"_revision_number":2},{"id":"3e06c457a02195a7a844a7f998d88c888c022ed4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 12:46:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"38d8d239da323314d7f146f44f437fe44840e00b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 12:47:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2cdcfd10d1845b891c83f40575f1e37\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b8ea1dfe7eb43dd8f3c4dfd1154b836 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/56fee27e543b4527bb50fb12df1c2e44 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ad4fc2a312ecac6a3dc85cbb0e98c31b8e405ef4","revisions":{"087ab1cd0cbe66d3a9830d537d096af3cc7a00e0":{"kind":"REWORK","_number":1,"created":"2023-08-23 10:22:12.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/80/892480/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/892480/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/892480/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/892480/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/892480/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/892480/1"}}},"commit":{"parents":[{"commit":"5d712da26b36b095cf7e0d26e148f92792083027","subject":"Merge \"Fix CentOS / Rocky route options\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5d712da26b36b095cf7e0d26e148f92792083027"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-07-06 11:08:49.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-08-23 10:22:12.000000000","tz":0},"subject":"Fix for seed-containers being unable to use password protected registry","message":"Fix for seed-containers being unable to use password protected registry\n\nUpdate old code to use truthy filter\n\nCloses-bug: #2026276\nChange-Id: I2085f9d1fd457e5b54b17f37a5e725990a60e8bc\n(cherry picked from commit ed263eeb4f8e85c2b33d3f92f5f711f725f5fdae)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/087ab1cd0cbe66d3a9830d537d096af3cc7a00e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/087ab1cd0cbe66d3a9830d537d096af3cc7a00e0"}]},"branch":"refs/heads/stable/2023.1"},"ad4fc2a312ecac6a3dc85cbb0e98c31b8e405ef4":{"kind":"NO_CHANGE","_number":2,"created":"2023-08-23 10:53:39.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/80/892480/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/892480/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/892480/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/892480/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/892480/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/892480/2"}}},"commit":{"parents":[{"commit":"5d712da26b36b095cf7e0d26e148f92792083027","subject":"Merge \"Fix CentOS / Rocky route options\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5d712da26b36b095cf7e0d26e148f92792083027"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-07-06 11:08:49.000000000","tz":120},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-08-23 10:53:00.000000000","tz":180},"subject":"Fix for seed-containers being unable to use password protected registry","message":"Fix for seed-containers being unable to use password protected registry\n\nUpdate old code to use truthy filter\n\nCloses-bug: #2026276\nChange-Id: I2085f9d1fd457e5b54b17f37a5e725990a60e8bc\n(cherry picked from commit ed263eeb4f8e85c2b33d3f92f5f711f725f5fdae)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ad4fc2a312ecac6a3dc85cbb0e98c31b8e405ef4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ad4fc2a312ecac6a3dc85cbb0e98c31b8e405ef4"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
