)]}'
{"id":"openstack%2Fkayobe~892501","triplet_id":"openstack%2Fkayobe~stable%2Fxena~I2085f9d1fd457e5b54b17f37a5e725990a60e8bc","project":"openstack/kayobe","branch":"stable/xena","topic":"docker_login-stable/xena","attention_set":{},"removed_from_attention_set":{"34940":{"account":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"last_update":"2023-08-25 12:56:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2085f9d1fd457e5b54b17f37a5e725990a60e8bc","subject":"Fix for seed-containers being unable to use password protected registry","status":"MERGED","created":"2023-08-23 10:22:44.000000000","updated":"2023-08-25 12:57:10.000000000","submitted":"2023-08-25 12:56:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892501-docker_login-stable/xena","cherry_pick_of_change":887792,"cherry_pick_of_patch_set":9,"meta_rev_id":"3252cef9e933414b547fc9348bbe78e54fbcd740","_number":892501,"virtual_id_number":892501,"owner":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-25 12:56:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2023-08-23 10:57:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-08-24 08:36:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2023-08-25 11:02:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2023-08-25 11:02:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-23 10:57:22.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-08-23 12:38:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-24 08:36:08.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-08-25 11:02:23.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"0dce86849992f9c45f21a3ae80f85283153d1ab8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2023-08-23 10:22:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d7009af04de4b69d417229903d65e7121a4f430f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-23 10:57:22.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c8a43009c7866e6138162276cd67a28fca1fa3d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 12:38:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17b2e818f258499ca99719d67a50441b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4432bab66084680a3cb07b08869a234 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/568e9133684b4cb6b4dea25f6bc19a8f : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71e955d001fa49a6b6a1ba25e7e40d0d : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab04149f8ca743aeb99bbef692ad26cf : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e0d91656a3c4f4f9a1022c05dfed4df : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58bab9c05ce4453692b58aee102e5e86 : SUCCESS in 9m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69b0eba7feab4b4d997ddd4add8cbea3 : SUCCESS in 2m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9408f44dec614f3e82f5e41fd63a1292 : SUCCESS in 5m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7fd7119dc3c2448abf79a64141d3d033 : SUCCESS in 13m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f314a8c5375346548beffaadbed51355 : SUCCESS in 20m 19s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a8f0a7d10d904462b60f156211544060 : SUCCESS in 1h 17m 10s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/7d51047feae24a469d93e68406d276d9 : SUCCESS in 57m 07s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d1d2abea057643d781b7323765cde0bd : SUCCESS in 1h 14m 20s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ee038696097044c78219324cd4466fa2 : SUCCESS in 29m 40s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/ddb9475ccdac46f99172d0c74edf7ea2 : SUCCESS in 20m 41s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1fa2536690c44a86bd6fcff6fe142ef0 : SUCCESS in 13m 33s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/38c33146a09e4169afe7c3ddd981a33b : SUCCESS in 2h 09m 01s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c0a61c02c3244c419a7ebfe7b9167f90 : SUCCESS in 21m 52s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/dc1f025c585b4a5a9e270091e87a2ce3 : SUCCESS in 18m 00s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d51f343580841dd8b03610cd04c5607 : SUCCESS in 13m 48s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/6637d95458144df382f5077aa814634f : FAILURE in 26m 08s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9a98441fdd7d4af29a2d5ae4c3cbe910 : SUCCESS in 29m 13s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fdfa789ff20843689057ae80e681cbf6 : SUCCESS in 19m 48s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/20eb4a3ae5104f8cae1035d38b94e33e : SUCCESS in 16m 02s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/4ff4b367cb4c4cd78558af6a768daafa : SUCCESS in 13m 19s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18e21c092e30459aaeada59e7be61c7c : SUCCESS in 12m 01s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/0a089b68be584203befe92374c77f3fe : SUCCESS in 20m 32s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/b1887656b0f1444b9996ea2df3a20de5 : SUCCESS in 16m 00s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f15c5441ccf4b228d80b6242720e26f : SUCCESS in 11m 20s","accounts_in_message":[],"_revision_number":1},{"id":"25f0b3bab0f137e160cf15df6e32bde2101666b5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-08-24 08:36:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"46b92ae76d9527f040733c3c040c08cb3d880f50","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-08-25 11:02:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"374a2162da2df068d3359bd292af35a5f23e1d57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 11:02:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0ceb3f027d6a198ca6766cbd58a139458d85b926","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 12:56:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2717ca119464a99a577e900d9be7cc1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b545e00175842758b1d8871753cc0b8 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15af7760cdc846a6985d730892ae2f04 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/446d730d6e4d422bb5ab04c30868f15e : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8419c7d1ed23418793f76098e991acfc : SUCCESS in 8m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfa8627cc54c42ca8ac5b605e00c16b8 : SUCCESS in 3m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5d65866855834f02bb3b0887359f6d51 : SUCCESS in 5m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/aedf143b9cb44ee19779fed84bb3fb65 : SUCCESS in 14m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/16fc77b0d6bb4f6590fcfdc1b87cf675 : SUCCESS in 52m 38s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e9ab2da439c44748b43b4e19668ea345 : SUCCESS in 1h 09m 27s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/3f1ebc02bac747c497b2cbc91eef5392 : SUCCESS in 1h 07m 55s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/bce1dee4351c4beb8c667232e8e1c818 : SUCCESS in 1h 04m 54s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d2ec1715ea4a41509c0da538f86aa28a : SUCCESS in 26m 27s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/524b615643014c5fb922a92f65a13352 : SUCCESS in 22m 27s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0317099440644e3aaf360bdb00c948b6 : SUCCESS in 15m 58s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/57c37c73a221430fb0f48266230b103e : SUCCESS in 1h 46m 39s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/94fe046403024917895dda7e7f76eb53 : SUCCESS in 21m 49s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/c3220b4727c443ad80a020866d81a830 : SUCCESS in 19m 31s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/19a6e73bfffc494f99c753056bce0324 : SUCCESS in 15m 37s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9afe40efab9a40c994593d3d58f40318 : SUCCESS in 29m 49s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55cc34b78d6d4614ad62926b291cabaa : SUCCESS in 22m 10s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/0688e87949b14b94b5027acdbf2c7542 : SUCCESS in 17m 53s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/5601be127109496f94fcf94b85d473aa : SUCCESS in 15m 46s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/374855d1a8c748b181ed0b7a6897d9e2 : SUCCESS in 12m 18s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/0365f7edd00c434a90f9fc300b5867bf : SUCCESS in 18m 45s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/73f80bc2289a4413b35f6e4a2b347292 : SUCCESS in 18m 43s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58a907291b7745aa8d6a2d5dbe97b3dd : SUCCESS in 11m 24s","accounts_in_message":[],"_revision_number":1},{"id":"67d3b1838372e2de4af0070babb720a9d9d0287c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 12:56:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3252cef9e933414b547fc9348bbe78e54fbcd740","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 12:57:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d07ce2244b2a4650a3e94ac08dc16f99\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/770b85bf8b704489b41a7f8ad8455620 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d208c84cc2d941238768d51dd01cd006 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f594ee5f256a49f2f451459fd3eb529154ff6f12","revisions":{"f594ee5f256a49f2f451459fd3eb529154ff6f12":{"kind":"REWORK","_number":1,"created":"2023-08-23 10:22:44.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/01/892501/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/892501/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/892501/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/892501/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/892501/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/892501/1"}}},"commit":{"parents":[{"commit":"0635ac49eddf11cda1a6e4e2154774033fd3cf1c","subject":"Merge \"Limit interface names to 15 characters\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0635ac49eddf11cda1a6e4e2154774033fd3cf1c"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-07-06 11:08:49.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-08-23 10:22:44.000000000","tz":0},"subject":"Fix for seed-containers being unable to use password protected registry","message":"Fix for seed-containers being unable to use password protected registry\n\nUpdate old code to use truthy filter\n\nCloses-bug: #2026276\nChange-Id: I2085f9d1fd457e5b54b17f37a5e725990a60e8bc\n(cherry picked from commit ed263eeb4f8e85c2b33d3f92f5f711f725f5fdae)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f594ee5f256a49f2f451459fd3eb529154ff6f12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f594ee5f256a49f2f451459fd3eb529154ff6f12"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
