)]}'
{"id":"openstack%2Fkayobe~892591","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~I4ea22dbe43cd4780335df5cf0eee5c4339855fac","project":"openstack/kayobe","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-08-25 12:10:37.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I4ea22dbe43cd4780335df5cf0eee5c4339855fac","subject":"CI: fetch systemd configuration including kolla-generated unit files","status":"MERGED","created":"2023-08-24 14:56:24.000000000","updated":"2023-08-25 14:06:18.000000000","submitted":"2023-08-25 14:05:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892591","cherry_pick_of_change":890185,"cherry_pick_of_patch_set":2,"meta_rev_id":"c14daed32c21ebfd753838a1e95bc5558dc8b6e9","_number":892591,"virtual_id_number":892591,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-25 14:05:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2023-08-24 14:57:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-08-25 12:10:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-08-25 12:10:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"CC":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-24 14:57:24.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-08-24 16:50:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-25 07:09:23.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"}],"messages":[{"id":"46e8b63898fad3d37e17b136c295bd95f901e880","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-08-24 14:56:24.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a833ebc1876664b62d48f914824905b3cf6121d9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-24 14:57:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3062ca51d182834627fdc768d9fa2c40f547b5ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 16:50:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e282574f1a3041ebaeee9a5e6d8a0225\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/378f936af454475ca36794255d41a476 : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53e7583eee3b49e3aeb2c66303d62e57 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12d664e73fbb48439a0e7a80a12a5ecc : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3bab1204c28476aa7f210e2215d18bf : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8af75a0c941e4086a77506df7fcb6795 : SUCCESS in 11m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4c525546c678487790e34a61d5f412bb : SUCCESS in 6m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e22641a555f54489ae2439dfef55b8e2 : SUCCESS in 12m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5c49899c46f145928927dbaf78edeb8f : SUCCESS in 17m 19s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/771614b6595f4f5d9983e44849b8e5af : SUCCESS in 1h 01m 31s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/6492deed362d444f97025c00ec26a0f8 : SUCCESS in 54m 32s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/89bafd5756ba496ea7fc114b26b98217 : SUCCESS in 1h 06m 38s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/cdc8d027407e4b54860cd1a4a020c0aa : SUCCESS in 51m 07s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/06cba47819924548a8643f316a830fb7 : SUCCESS in 27m 31s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/39675f28c85d49279867b4096ffa84f1 : SUCCESS in 21m 58s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/9b093da0f55048638210a2f2c33f798b : SUCCESS in 20m 15s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/07471076f7e74139b755193b0fe9bb15 : SUCCESS in 21m 46s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d90386d92c0d4579a8ba32503e4588a6 : SUCCESS in 15m 38s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b39b7726c89d4e258b9cea9210ceefb2 : SUCCESS in 17m 31s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/57dc4d2cebed4feab22bff3566ed67af : SUCCESS in 1h 06m 25s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/55e7bb90a200437da8de36b250c4571b : SUCCESS in 47m 39s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0dfe64e11ad8476783d5d944bca220b7 : SUCCESS in 1h 47m 56s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/49e5856e3a2c474d8e14c64acf6b03f7 : SUCCESS in 23m 35s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/4a71afbca7c04176a9f86272174ab218 : SUCCESS in 21m 40s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/81b0db8df9264ef4a35d28e58264ce3e : SUCCESS in 20m 54s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9131afdaa554343a864cb9314e995da : SUCCESS in 17m 28s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/881835e7395f4177b85a7a7ec9934dba : SUCCESS in 17m 59s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/eaf4203d692748b09a274946074c5d2c : SUCCESS in 27m 11s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e8412d9b34b43059b1afbd615e47ae9 : SUCCESS in 29m 37s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b7ec1f4ed5a34bb39b01335a6bc62e3e : SUCCESS in 17m 39s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/e9fa0b45174c4117bebfa75ea158fb8d : SUCCESS in 20m 00s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ad03629bad8c44d1ab335249806e0ad6 : SUCCESS in 14m 37s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bcc589cb2dfb431f9fab3bd2675b81ee : SUCCESS in 15m 09s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fd7677f7ba934cf0b316a2115f71d9bc : SUCCESS in 14m 30s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/3c33955d49c14983a7cc75ea3487cad8 : SUCCESS in 25m 15s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/2b35b5bc8b0c4242af2e2a25e4bd3b85 : SUCCESS in 17m 53s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8f2aca7381b646aa9dc21a58699c3763 : SUCCESS in 18m 49s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7c59bbd599648d7b54d3dec0f3b6678 : SUCCESS in 15m 31s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/039e92e6793842d3843aaa15204d8039 : SUCCESS in 16m 04s","accounts_in_message":[],"_revision_number":1},{"id":"01ac68310c5a941b8a4c2e47693e69a7e2392b76","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-25 07:09:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bce14f9a0ad3d853c7ba1fd04331b9cb64d22aa7","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-08-25 12:10:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"97ae725e35153213ebcd255a00c5297cd486092a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 12:11:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c36fe6104d2174a31719be8ba222e8b97fc0e72f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 14:05:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdc8cccdb4f34706946199a60df87efe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f43b87124d08405ebd777b526cc0dca0 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a22ba4c2d0c04f4f8cac42c9daca8186 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/792e2e7bb2054d40a7c156ff189f3315 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3ebb59627d34add9c47455ea15dea93 : SUCCESS in 9m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/eac73c426a284b6e9508cb9a2677937c : SUCCESS in 7m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/17fdcb65a1cf4ec4bd4ed11723ab8976 : SUCCESS in 15m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dc80476cfbe540ec90c1f94bc969c069 : SUCCESS in 37m 34s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6a9f11ae13194fbf98b8899651b9ece6 : SUCCESS in 1h 12m 34s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/8684f617947a419480e2e3828305aa69 : SUCCESS in 1h 27m 37s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/75d4876a26564ac49220cd3513acbc96 : SUCCESS in 58m 35s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ac173154d5bb473091f40c304a940875 : SUCCESS in 25m 00s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/6fb0a327c89a47678cc1189d14b0499f : SUCCESS in 22m 31s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/83b1ebf95f444042b11c33e3d8e3e13f : SUCCESS in 21m 36s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/efb158b7b7464cb68351c71794497a91 : SUCCESS in 17m 03s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d7ed770ac544a1a8e07b07eaa900645 : SUCCESS in 16m 44s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9a099259c0524b398bf3733b119df30d : SUCCESS in 1h 19m 58s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/2af4a3937e1445da9c08e31cba95d81a : SUCCESS in 1h 01m 14s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8d66a473d5854faeb162856e33e3580b : SUCCESS in 1h 49m 55s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/9c50f1abd149452e8885f3b0a80b48a5 : SUCCESS in 18m 16s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/a3a9d091575747b9b60b2483b448d65c : SUCCESS in 21m 57s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/bd945042b445418b831b2f2562568340 : SUCCESS in 15m 41s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b17220df56e94951a89bf69d95a20776 : SUCCESS in 14m 34s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b5a6155a2d5040f4b5bed959d47eb7b9 : SUCCESS in 16m 51s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/022eace273414efda381a1f0b8d11b4d : SUCCESS in 23m 45s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c7daae74a9b420e855d7194652555de : SUCCESS in 27m 36s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/367c032cfee54ddf9cff2ae17d5b2036 : SUCCESS in 14m 29s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/19953686e7444bddbffa6ba6b51a8e6d : SUCCESS in 15m 35s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bc4693d2248249fda4d727da6b56bcd4 : SUCCESS in 13m 11s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1473725b76b24e69b4bc2c0b045bd6f7 : SUCCESS in 13m 31s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb6ab353ab3c451483f3c85ecfa3d12c : SUCCESS in 11m 48s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/2a113efdc5f04f9ea3b0f017e54ac9fc : SUCCESS in 18m 31s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/edf4611189184df4ae61209d7391f9d4 : SUCCESS in 15m 40s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0cc9341f06f9465db01df92511cf48ea : SUCCESS in 14m 03s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10cfe914b10b4bf79c5734477887415e : SUCCESS in 11m 28s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf752ecb0e5a4a6ead5e837c20307c1d : SUCCESS in 13m 51s","accounts_in_message":[],"_revision_number":1},{"id":"6ee70cfb0099679a696c7a3d44ae6a35aa391cb6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 14:05:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c14daed32c21ebfd753838a1e95bc5558dc8b6e9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 14:06:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/337858775ba54fccb5ebf63e392478b9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/996b3b6d1d68405988bac743061677a0 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3794929b1747769ecdbae8dfb218e31ae9aeeb23","revisions":{"3794929b1747769ecdbae8dfb218e31ae9aeeb23":{"kind":"REWORK","_number":1,"created":"2023-08-24 14:56:24.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/91/892591/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/91/892591/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/892591/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/892591/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/892591/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/91/892591/1"}}},"commit":{"parents":[{"commit":"967c4beff86689faba679f55255f92d23505ced1","subject":"Add STP option for bridge interfaces","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/967c4beff86689faba679f55255f92d23505ced1"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-08-01 09:50:24.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-08-24 14:56:24.000000000","tz":0},"subject":"CI: fetch systemd configuration including kolla-generated unit files","message":"CI: fetch systemd configuration including kolla-generated unit files\n\nChange-Id: I4ea22dbe43cd4780335df5cf0eee5c4339855fac\n(cherry picked from commit 09507a38248ccdbd34d978aca2fa4936cec55fc3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3794929b1747769ecdbae8dfb218e31ae9aeeb23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3794929b1747769ecdbae8dfb218e31ae9aeeb23"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
