)]}'
{"id":"openstack%2Fkayobe~893098","triplet_id":"openstack%2Fkayobe~master~Idbbe0dd57acc9b7a9440a67c2b677e526a6be917","project":"openstack/kayobe","branch":"master","topic":"rabbitmq-ci-fix","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-08-30 09:59:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idbbe0dd57acc9b7a9440a67c2b677e526a6be917","subject":"Work around failing RabbitMQ precheck","status":"MERGED","created":"2023-08-29 18:02:21.000000000","updated":"2023-08-30 10:00:58.000000000","submitted":"2023-08-30 09:59:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893098-rabbitmq-ci-fix","meta_rev_id":"ef55e11b7400abd73359d69798307c625215b36e","_number":893098,"virtual_id_number":893098,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-30 09:59:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-08-30 08:04:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-08-30 08:03:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-30 08:07:55.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-08-30 08:04:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-29 20:00:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-30 08:03:59.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-08-30 08:04:47.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-08-30 08:07:55.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"8457bea46a91038907fe1f0202d059d8c6100065","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-08-29 18:02:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4d8b246db2115bfb00b4ee27d4db236681529c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 20:00:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/14810f9e8b534ac8ae7fff1249394592\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f8e28326a3e45c1beb55be03900e1f6 : FAILURE in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a5ade4d69844193bc354443037e1cb4 : FAILURE in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40aa2bdce9484def8a7b7b7890e9d0b2 : FAILURE in 3m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/704297e8f34c44f88ae1b43a784693d5 : FAILURE in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fc7681c1f142412a80789579260f4226 : FAILURE in 5m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/50da861e44144085a35d3a8c85504b11 : FAILURE in 4m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8beeff89f79b42189ac69e4c5f51f253 : FAILURE in 3m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ce3d7917c6b945fc995dc1f3fe9237d4 : FAILURE in 3m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7d82a0fec9a14e50a52ea643c6d538dc : FAILURE in 3m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7ce6e64d7e1e4c14abb66ab8ac6cdd2f : FAILURE in 5m 09s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/63fa94be8d5446e2b7926d1f336f2cd0 : SUCCESS in 58m 56s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/0ad82d4e71d34e4fb623466d624b6e32 : SUCCESS in 59m 27s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8ecb0ac287dc4f58a901517fe71f9c05 : SUCCESS in 46m 11s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c765a42536b54cc2a89a6e4ace0680ae : SUCCESS in 56m 23s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/cb22b45f3e064ad9b14fcbaa158b178f : SUCCESS in 16m 24s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/6bf2b0f05e944272b26db204ad81a325 : SUCCESS in 17m 22s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7230649a4cc4957a90f038e9a1d0fdc : SUCCESS in 14m 00s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4dfc4301cb6b4961b08badc78573693e : SUCCESS in 1h 51m 26s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97bca8f4dabd4b52884e7a81b1a9285d : SUCCESS in 1h 40m 25s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a6bb450b822b4312beb5c19530f04b02 : SUCCESS in 15m 58s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4725b028a99242df95c2d71e4728771e : SUCCESS in 13m 39s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/282f6a82af154f16a723a00292450a0d : SUCCESS in 35m 06s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/736bc18247254d48ade0294626e477d2 : SUCCESS in 29m 23s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ec0af0ee636460cbb1ee5b82d408541 : SUCCESS in 26m 11s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/705cd59e012b45cc94913ebca9cb7a46 : SUCCESS in 12m 46s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c18fac3d99bd49fd834e07f5e92f16bb : SUCCESS in 11m 53s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a47397cc170b462180f4059d470227c4 : SUCCESS in 33m 48s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/629d8a572db1418e9a3dc816864afdaa : SUCCESS in 29m 03s","accounts_in_message":[],"_revision_number":1},{"id":"b3f9de05718c4b1f46a534fc83bda798448688ad","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-08-29 21:26:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eb367e9dd308972b86f6f3e15026fd12bd94394a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 23:29:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d6374703e0146e3b48ad58ad291204c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6371a67b35b145e88fd39ca16e10cfa9 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23f7b66723524b7f9d747b6a48089889 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ed2f91532834f3c8f3518bcf56f2076 : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0fcbc87f7ce04dbc88ab413e249ae892 : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/117c6b1bd1f745508a723d21aaf95c39 : SUCCESS in 4m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/52df65656b044f8688497138a5815668 : SUCCESS in 3m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/972fd94b609142e18f25f08b28e1b15f : SUCCESS in 7m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6a82dde402b047dd85145d021f6ada62 : SUCCESS in 4m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/48983f8623a8412d8345ef040801a55a : SUCCESS in 9m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/67ce986dcaad4770a44743cc8067da13 : SUCCESS in 15m 43s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/da71bd5615f44e4b9f623b2c2c6e41ba : SUCCESS in 54m 20s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/044e308e8e1345c6b561375cf948057a : SUCCESS in 50m 43s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/64b3be53163049249623b2afae01ebe7 : SUCCESS in 47m 43s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/f914134543ea4110959f31f0a63e38d5 : SUCCESS in 1h 01m 28s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c64442f7dfb8481b941db5149c444312 : SUCCESS in 17m 04s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/e83985703dc7456faca945a125e7570b : SUCCESS in 18m 48s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/37ed355730004e07acb0fe5bc7586f10 : SUCCESS in 17m 32s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/59a55a2baeb949f490dd040c2abc2d51 : SUCCESS in 1h 29m 53s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/22a93ffa37f64a9e9b2b46d169ed7d65 : SUCCESS in 1h 57m 34s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/186ebc8cfff94fcab14ffccdf66e95f2 : SUCCESS in 12m 54s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/efe3f83a94034726990cc9fa9f8e89ae : SUCCESS in 14m 23s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/34024bb72fc6431da8eafaa38ea33287 : SUCCESS in 26m 50s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/86790d5d91d6416fa0c56ada3f0fcae8 : SUCCESS in 28m 17s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/81f4c9918a5640f5bff5f50743077670 : SUCCESS in 23m 55s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/233356807eb746c4b84e67a612aaa3f8 : SUCCESS in 13m 30s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7cc84b550804a8688fe329b88510653 : SUCCESS in 10m 48s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8ca3639eb9704d219595ef1323dfae3b : SUCCESS in 34m 31s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7629a919075421baa0ecb9fe766b210 : SUCCESS in 27m 50s","accounts_in_message":[],"_revision_number":1},{"id":"245e1bd82fe24f21a0df8106c4e50894d03e49f9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-08-30 08:03:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ba1919bca564c285586f212e3a64a8e9dee5f8fd","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-30 08:04:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e5184680c49daeee6d5935855924bb0b54d93de3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 08:05:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"da3d4de3d0ea2d20422f5d929907ecbcf7427076","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-08-30 08:07:55.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b8e01d3905b62775bc13fab4163ee67567fbe616","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 09:59:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dac09342bd8147999da81565af0437d1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5e83e83252d42a090ec81512bed05b2 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9cd1e128a284a9ba4aeb4f7ee23d423 : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/095900f2c53747ad8933e33c4b905d6d : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/15c8f6d15bc3407998707e404117f7e9 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9675d6070262420fba6eed96814c773c : SUCCESS in 7m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6138da86cc2b4d288ea41e528e5d49a0 : SUCCESS in 4m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2ff69229e8c6459eb10b8783770fd1cb : SUCCESS in 11m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e264ee2fa80b40829fe4e4c00216ca6b : SUCCESS in 17m 12s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/1c747ad5151c4d478954f9da9b8e8e59 : SUCCESS in 57m 16s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d1b7c5d14634b7694c9b578bf14d80f : SUCCESS in 55m 54s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/834b424f70574386a7abd129bb8f890d : SUCCESS in 57m 19s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c251b2df98b04beca50d59c9889846ec : SUCCESS in 17m 23s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a942f87c0ab24d789bef1fabc771ddd6 : SUCCESS in 13m 16s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2ce6cd2363db45dcb9a30e1354ee8a05 : SUCCESS in 1h 50m 13s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7e57b8123454ff39af80e125590fb85 : SUCCESS in 1h 45m 49s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4a655151c0b941b5804f4a839daf331d : SUCCESS in 18m 13s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3729f197f9cd403090466f6e9b194cb8 : SUCCESS in 14m 10s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6c443c0f4624430c8a95bb7972a0f400 : SUCCESS in 29m 56s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5ed06f69e7947d9b4218d3df5738202 : SUCCESS in 25m 13s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/77416b353f2548a8a2bbc3b5817d1ef4 : SUCCESS in 13m 35s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ad3f07800d949f3969757e487431993 : SUCCESS in 11m 03s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ba2eff5cc2c1446aaa8449453559fcae : SUCCESS in 36m 46s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b5761b54c5e4bb68a670d8d325f5684 : SUCCESS in 31m 04s","accounts_in_message":[],"_revision_number":1},{"id":"3dcfa6c5a10a548c0edef246e76715b9d32e8e94","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 09:59:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ef55e11b7400abd73359d69798307c625215b36e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 10:00:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc92e106e1c34df5bbb5c16a1d7086dc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2864caa21c04074b767df138706830a : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"014446251d54091c174815865e1cda925f31ffe2","revisions":{"014446251d54091c174815865e1cda925f31ffe2":{"kind":"REWORK","_number":1,"created":"2023-08-29 18:02:21.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/98/893098/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/98/893098/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/893098/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/893098/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/893098/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/98/893098/1"}}},"commit":{"parents":[{"commit":"7110d834551f3a4ae299476ffea13f7a187f4a15","subject":"Merge \"CI: fetch systemd configuration including kolla-generated unit files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7110d834551f3a4ae299476ffea13f7a187f4a15"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-08-29 18:01:50.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-08-29 18:01:50.000000000","tz":120},"subject":"Work around failing RabbitMQ precheck","message":"Work around failing RabbitMQ precheck\n\nKolla Ansible enabled RabbitMQ HA queues by default, which require a\nmanual migration step [1]. Work around the failing precheck by enabling\nHA queues in the previous release until we implement migration code in\nKayobe CI jobs.\n\n[1] https://review.opendev.org/c/openstack/kolla-ansible/+/882825\n\nChange-Id: Idbbe0dd57acc9b7a9440a67c2b677e526a6be917\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/014446251d54091c174815865e1cda925f31ffe2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/014446251d54091c174815865e1cda925f31ffe2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
