)]}'
{"id":"openstack%2Fkayobe~893137","triplet_id":"openstack%2Fkayobe~master~I82c286fd17e3a1d7f31952442fa281302cda7ee4","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-09-21 10:06:43.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-09-21 12:36:00.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2023-09-21 12:36:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I82c286fd17e3a1d7f31952442fa281302cda7ee4","subject":"CI: Migrate RabbitMQ queues during upgrade tasks","status":"MERGED","created":"2023-08-30 08:39:04.000000000","updated":"2023-09-21 12:37:14.000000000","submitted":"2023-09-21 12:36:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893137","meta_rev_id":"78addf2a98e7ef8bc67289e9117c53dc7b0808d7","_number":893137,"virtual_id_number":893137,"owner":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-21 12:36:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-09-21 10:01:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-09-21 10:06:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-21 10:02:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-21 10:02:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-30 09:19:51.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2023-08-30 09:55:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-30 10:08:32.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-08-30 13:11:25.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2023-09-21 10:01:32.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-09-21 10:02:49.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"0a3497227a15fab2a71719b8ab0e6d4f8bd974f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-08-30 08:39:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b7880c481d38e390844bc9f9eeee33e0af28002a","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-08-30 08:39:14.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"39b033cf1d899e6d7bdbe3d93e908770b2054483","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-08-30 08:42:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8dd24df66ea810fdf4a5baeec59530af81145ad9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-08-30 09:19:51.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"00aa812ab2fdca02c08c7e05c6cffee4fe1bcc98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 09:55:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3689b8c7f81a47b3a7b07b9cba1ab954\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c317a1663264ffb805272c5328309b4 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8dc98f6e8fd417a80e1e46fdf907aad : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23ffaf99a67d462fa5d0a76fd944a152 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c8dadae7b4074f07853deb806c0c70e7 : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8f46113425bb48a2b91e8960fbd90a6d : SUCCESS in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/72a333f2ae2f4ce29d6cc4f6ab2e8947 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/443891dc4f0345feaafd58a2ae1057f2 : SUCCESS in 7m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cc875667082f482ab1185b5ec09668a6 : SUCCESS in 3m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/73fcfb2a9b0b4a92a74d58a9690694a2 : SUCCESS in 12m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7681abe8eb394d69b2478ecb3812b602 : SUCCESS in 16m 34s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ff940d265fd44adf94fd9c05dc4f2a26 : SUCCESS in 1h 01m 49s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/890faa660e1442acbd6909d8e528256c : SUCCESS in 53m 38s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d739860c7b2f493c99060481d19f8131 : SUCCESS in 52m 59s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9ee3c12e7ab146f0b14ae3bcbb0f5052 : SUCCESS in 1h 04m 02s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b54850dfef0c43ed820a4330cf147945 : SUCCESS in 16m 24s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/8b44b65cd3d64dce924490f596be54ee : SUCCESS in 20m 35s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd542a7b84d8415bb72fddc13d113f8b : SUCCESS in 11m 46s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/34e827bc895d4d3f8891b46b66892d1e : FAILURE in 1h 07m 58s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de63fa3d802b4c54aecb8465ba752bea : FAILURE in 1h 00m 25s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/cf97147bb9144e9690c14732f87f2f70 : SUCCESS in 16m 08s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/17acab967c1b476f96321fcf2198cd50 : SUCCESS in 14m 14s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/6f0aac65d5b74a3185f1ab590f41b131 : SUCCESS in 28m 20s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/56ddd397de5b4da4b2a2fcdf7c0c7226 : SUCCESS in 23m 59s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5adec90b9ac345dc895042f2d697be5f : SUCCESS in 25m 43s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1f5a6feea56c4f9eafa9bcbe6df2f195 : SUCCESS in 14m 32s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d03fb5523de3436ea225b5a712017d38 : SUCCESS in 11m 05s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c9d76ac5eeae46a18c0a587ad77f3a3d : SUCCESS in 22m 36s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/132f34c4ba2646128d8c2eab7def5b25 : SUCCESS in 27m 36s","accounts_in_message":[],"_revision_number":2},{"id":"056727d46503591413984f59493b920d4ff56ce5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-08-30 10:06:04.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"8d50c5396786aecb9e2ca13d6e5dc24baa62835a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-08-30 10:06:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b361e4d21057e2d5cdbe430e40198dc65f923863","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-08-30 10:09:09.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a9aab146f7dda2413c0a13f6aab50cc3d5f73736","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-08-30 10:10:22.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"17abbeb79865a5ec85099aea46610bb89144d20f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 11:41:09.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/87b620b7ee3e4774b03aa4c4cce68889\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b35edac4de444bca8d30ea4e10d854a3 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65dc8c5b29574fba908a186b8b3936e5 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff0df39a57f449e185ca4b659169043c : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/faac2d0111644710b0b557fe4b69df96 : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f55ad42f4f84052adf51f6387865285 : SUCCESS in 4m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/25c7b2f4e8da45b8908c05457d019448 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9318b8181d04b4899fd53c93f044024 : SUCCESS in 8m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/663ccbef76f34c849c8b0b7f4a008a55 : SUCCESS in 5m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0aaf1f040b49455c880fadc7fe6a3e6c : SUCCESS in 11m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f84be2aa9a7a42908df778ba44ca2a27 : SUCCESS in 14m 50s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d10f4ab82b754df386f71eef4e6d828d : SUCCESS in 54m 29s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/cdfa7dcff62343b7948ba9737f7b8d84 : SUCCESS in 1h 03m 11s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4f5291082b2a4661a0f83e4a48780ebb : SUCCESS in 56m 28s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/950bda16952d49b98cf356300036d023 : SUCCESS in 1h 00m 47s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/504267db59554806846b7dd9e1cf8f5d : SUCCESS in 17m 21s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/7c740e200adb4847aae0e468af83d769 : SUCCESS in 18m 35s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/45c4a3266ac44d458c25da8464a2ca3c : SUCCESS in 13m 46s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3725c53adefc472b8b865d25605a140e : FAILURE in 1h 17m 11s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/36f40ed8beb548b5bc4241ee0f5aa253 : FAILURE in 1h 24m 09s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/9c81129228ca4dfe8479b4ab98033c8b : SUCCESS in 18m 44s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e75b5fa86042454ea973756cfd7ae34e : SUCCESS in 16m 23s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/0c73da62703a49cab4d33b015e445c98 : SUCCESS in 36m 33s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c9040ed66390454c9b6d247f9570e68e : SUCCESS in 30m 25s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/56068ff4f9b445b087929b9e37b41e09 : SUCCESS in 25m 29s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/cfdb85e1582f4a49beae67aaa3b75343 : SUCCESS in 12m 54s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04ca4535f6204654a2b255a641d103a9 : SUCCESS in 12m 32s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/febe06e0d7834018880e964240147833 : SUCCESS in 29m 12s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc3dfed6904a4bf99fd873910a8644b1 : SUCCESS in 26m 35s","accounts_in_message":[],"_revision_number":5},{"id":"49225a853969e607cd3d752b4d444974c8fa2840","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-08-30 12:25:27.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"876346a5bebea5f69ff2e4d427210c94648088c6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-08-30 13:11:25.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c00ee6134f7479b95a411a8816b0a9b223e8b7f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-08-30 13:24:20.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":7},{"id":"84b56308f70766c0252bb3bebafe88aa49e3096c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-08-30 13:25:15.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":8},{"id":"018ece12cc65887d286bd55c0bc4ab12d8ad85a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 15:43:12.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2126c70f9bd04753874fd31d94740cb8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da3f260c1a7348c1baf8a6951e4b1b25 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48a8beb94dc44a96bd43b115dfe8353a : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/380231c56d2f4f46ab81319eec730f1e : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f76078bbbd004c5286f232313665c803 : SUCCESS in 5m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/555a9f7ed38a4009b9de4a86973a7881 : SUCCESS in 5m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/23baf0be99c6453ebbddf8dbd9133520 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b73b6862906045adad13b9d293813364 : SUCCESS in 7m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/39e6f8071db84653ac689da23fed985b : SUCCESS in 4m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2118c9c3d15245199b9cdd766bb43cfc : SUCCESS in 11m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e5319058c1504b27b3a24ce340642246 : SUCCESS in 12m 23s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/497301b565004c72932da3b84ce8556a : SUCCESS in 54m 28s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/74df0139b8d9454f8d380b396816019c : SUCCESS in 1h 01m 10s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1103afd8b69d406bb5c390f0c46f690a : SUCCESS in 59m 21s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/84e2abfea5784d2d8e01007b5cdfbcd7 : SUCCESS in 1h 01m 11s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5dcab7ace3714633aa129ceae27412c1 : SUCCESS in 16m 41s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/d1ffd9cbf8364f29b5a366f44a00dd92 : SUCCESS in 18m 26s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ed69af3c52254e8681d3f29c3ad5dfe8 : SUCCESS in 14m 23s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/908b20edebd54a76af843cb154cd8422 : SUCCESS in 2h 15m 59s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a15f5a28d8b45eb8bd234d5c1e1eb2e : SUCCESS in 2h 03m 00s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e4dd90a5b901417798ece73492a2c5dd : SUCCESS in 15m 21s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef36f479efc9406797180e7212ea410d : SUCCESS in 16m 58s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/3d55e3d07f6141c59b557d4b05704c8c : SUCCESS in 34m 13s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e4a182ffd09d4b67b4c0d8d4b9428f46 : SUCCESS in 32m 39s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3fbca198c49f41cda8e13f6f33281a4c : SUCCESS in 23m 23s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/54a7ee25f0f142e5806c09bd466b9dcd : SUCCESS in 12m 28s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/54cd5edeea5547e583a08d05c6011bf5 : SUCCESS in 11m 04s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7ed1dc212d0f49198ac6d1ab04dd3b24 : SUCCESS in 30m 50s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dded6a33b06b4159b5513d2a7aa06e1d : SUCCESS in 34m 10s","accounts_in_message":[],"_revision_number":8},{"id":"78ada8ea592858c382b0aaf0b31af170c2d4668c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-21 10:01:32.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"efc0f0207ca2c29d0cd62c480dda2f414c411325","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-09-21 10:02:49.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"ee4e5d23159e172f9860de45c59436ed72036066","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 10:03:16.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"f0aa92babc87bd4c9a52bb5ef7fef4c338420e99","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-21 10:06:43.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"73800c4e93c0a846b527083d0f722bda9b5c925d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 12:36:00.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b780662455cf4ccbba91fcf0fea8ddd4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71d2079178794b67925bcdb9b28d037b : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18771543fa5643d4becff04fc844673b : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a2e24d61e2784e92a59ff38ad9da5ef1 : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c09f59158e2e4fc7ba871c307c4da807 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f40584480124ae896ad4273b0abf3ee : SUCCESS in 7m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/12ab3def0f7b4c748f1ac8916ddf3f2a : SUCCESS in 4m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7947e5b589a74ff89dedfc3f4a59add4 : SUCCESS in 12m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3c80267c1921444f82d6c131db1d81b3 : SUCCESS in 9m 49s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ae7110c9295a4e8f8647c12f5abf82f5 : SUCCESS in 53m 25s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/79cb8223b11948e79fd1ed8fa46b1984 : SUCCESS in 47m 40s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/14829671b46940d38dc6cca7029ba8bb : SUCCESS in 1h 01m 20s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/bc9d5d52425b42199bcd6998b4f2efb0 : SUCCESS in 16m 59s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff89526dc09a484cbff4433565005d3d : SUCCESS in 15m 29s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/66c36a99c6d9454ba31439669ed6c713 : SUCCESS in 2h 13m 30s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/567429f155454c1f95aa2b11dd3aa37e : SUCCESS in 2h 09m 15s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/270283b59f154e7c85a52afee386ef99 : SUCCESS in 18m 39s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5cfc7d34b5774b92bfd7f2f1e1c19200 : SUCCESS in 15m 05s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/44161e2bccf1496cae95fccee31bb453 : SUCCESS in 29m 23s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a51a91d46e3435ea568833aea4f785a : SUCCESS in 24m 16s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e27719bb0b9d4cda8ad718110930daac : SUCCESS in 11m 43s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/825cfe3190e146cb9aad873c966ef54e : SUCCESS in 11m 48s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2b4783eba49a4d3b8792c47b9f739362 : SUCCESS in 27m 12s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04492b6e3e2d4cb9b5a749cdfeb22752 : SUCCESS in 22m 24s","accounts_in_message":[],"_revision_number":8},{"id":"d3b374f448abef5ef56678c6d44e7db1dd30abbb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 12:36:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"78addf2a98e7ef8bc67289e9117c53dc7b0808d7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 12:37:14.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/900a2108b76c48958c1e3661c0583166\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41fcfa79e11c4d63b945f92953d6aff4 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"e10446837113bbd65d6d84b38082797953282297","revisions":{"8254bca8069a0d677d169df0ddcb7fda036389c7":{"kind":"REWORK","_number":1,"created":"2023-08-30 08:39:04.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/37/893137/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/37/893137/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/893137/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/893137/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/893137/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/37/893137/1"}}},"commit":{"parents":[{"commit":"7110d834551f3a4ae299476ffea13f7a187f4a15","subject":"Merge \"CI: fetch systemd configuration including kolla-generated unit files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7110d834551f3a4ae299476ffea13f7a187f4a15"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-08-30 08:37:52.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-08-30 08:37:52.000000000","tz":60},"subject":"CI: Migrate RabbitMQ queues during upgrade tasks","message":"CI: Migrate RabbitMQ queues during upgrade tasks\n\nKolla Ansible enabled RabbitMQ HA queues by default, which require a\nmanual migration step [1]. Adds these to the Kayobe upgrade CI.\n\nChange-Id: I82c286fd17e3a1d7f31952442fa281302cda7ee4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8254bca8069a0d677d169df0ddcb7fda036389c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8254bca8069a0d677d169df0ddcb7fda036389c7"}]},"branch":"refs/heads/master"},"8f1558c78c0c28fbae3c883aecc67290d730ace5":{"kind":"REWORK","_number":2,"created":"2023-08-30 08:42:57.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/37/893137/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/37/893137/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/893137/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/893137/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/893137/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/37/893137/2"}}},"commit":{"parents":[{"commit":"7110d834551f3a4ae299476ffea13f7a187f4a15","subject":"Merge \"CI: fetch systemd configuration including kolla-generated unit files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7110d834551f3a4ae299476ffea13f7a187f4a15"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-08-30 08:37:52.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-08-30 08:42:49.000000000","tz":60},"subject":"CI: Migrate RabbitMQ queues during upgrade tasks","message":"CI: Migrate RabbitMQ queues during upgrade tasks\n\nKolla Ansible enabled RabbitMQ HA queues by default, which require a\nmanual migration step [1]. Adds these to the Kayobe upgrade CI.\n\nChange-Id: I82c286fd17e3a1d7f31952442fa281302cda7ee4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8f1558c78c0c28fbae3c883aecc67290d730ace5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8f1558c78c0c28fbae3c883aecc67290d730ace5"}]},"branch":"refs/heads/master"},"836878f30dbd58952154b45f258cb680ac0780a3":{"kind":"REWORK","_number":3,"created":"2023-08-30 10:06:04.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/37/893137/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/37/893137/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/893137/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/893137/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/893137/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/37/893137/3"}}},"commit":{"parents":[{"commit":"7110d834551f3a4ae299476ffea13f7a187f4a15","subject":"Merge \"CI: fetch systemd configuration including kolla-generated unit files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7110d834551f3a4ae299476ffea13f7a187f4a15"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-08-30 08:37:52.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-08-30 10:05:13.000000000","tz":60},"subject":"CI: Migrate RabbitMQ queues during upgrade tasks","message":"CI: Migrate RabbitMQ queues during upgrade tasks\n\nKolla Ansible enabled RabbitMQ HA queues by default, which require a\nmanual migration step [1]. Adds these to the Kayobe upgrade CI.\n\n[1] https://review.opendev.org/c/openstack/kolla-ansible/+/882825\n\nChange-Id: I82c286fd17e3a1d7f31952442fa281302cda7ee4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/836878f30dbd58952154b45f258cb680ac0780a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/836878f30dbd58952154b45f258cb680ac0780a3"}]},"branch":"refs/heads/master"},"aa8034c78b18bde6502ecd0f88c4610444cc7c99":{"kind":"REWORK","_number":4,"created":"2023-08-30 10:06:47.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/37/893137/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/37/893137/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/893137/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/893137/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/893137/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/37/893137/4"}}},"commit":{"parents":[{"commit":"7110d834551f3a4ae299476ffea13f7a187f4a15","subject":"Merge \"CI: fetch systemd configuration including kolla-generated unit files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7110d834551f3a4ae299476ffea13f7a187f4a15"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-08-30 08:37:52.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-08-30 10:06:39.000000000","tz":60},"subject":"CI: Migrate RabbitMQ queues during upgrade tasks","message":"CI: Migrate RabbitMQ queues during upgrade tasks\n\nKolla Ansible enabled RabbitMQ HA queues by default, which require a\nmanual migration step [1]. Adds these to the Kayobe upgrade CI.\n\n[1] https://review.opendev.org/c/openstack/kolla-ansible/+/882825\n\nChange-Id: I82c286fd17e3a1d7f31952442fa281302cda7ee4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aa8034c78b18bde6502ecd0f88c4610444cc7c99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aa8034c78b18bde6502ecd0f88c4610444cc7c99"}]},"branch":"refs/heads/master"},"6aba7cb382a06bca03ab1f7ad6d5a3e5be74b37f":{"kind":"REWORK","_number":5,"created":"2023-08-30 10:09:09.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/37/893137/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/37/893137/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/893137/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/893137/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/893137/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/37/893137/5"}}},"commit":{"parents":[{"commit":"7110d834551f3a4ae299476ffea13f7a187f4a15","subject":"Merge \"CI: fetch systemd configuration including kolla-generated unit files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7110d834551f3a4ae299476ffea13f7a187f4a15"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-08-30 08:37:52.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-08-30 10:08:38.000000000","tz":60},"subject":"CI: Migrate RabbitMQ queues during upgrade tasks","message":"CI: Migrate RabbitMQ queues during upgrade tasks\n\nKolla Ansible enabled RabbitMQ HA queues by default, which require a\nmanual migration step [1]. Adds these to the Kayobe upgrade CI.\n\n[1] https://review.opendev.org/c/openstack/kolla-ansible/+/882825\n\nChange-Id: I82c286fd17e3a1d7f31952442fa281302cda7ee4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6aba7cb382a06bca03ab1f7ad6d5a3e5be74b37f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6aba7cb382a06bca03ab1f7ad6d5a3e5be74b37f"}]},"branch":"refs/heads/master"},"75f66a08048d11ae950f8c33c97bdab16c976ec7":{"kind":"REWORK","_number":6,"created":"2023-08-30 12:25:27.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/37/893137/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/37/893137/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/893137/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/893137/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/893137/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/37/893137/6"}}},"commit":{"parents":[{"commit":"7110d834551f3a4ae299476ffea13f7a187f4a15","subject":"Merge \"CI: fetch systemd configuration including kolla-generated unit files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7110d834551f3a4ae299476ffea13f7a187f4a15"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-08-30 08:37:52.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-08-30 12:25:19.000000000","tz":60},"subject":"CI: Migrate RabbitMQ queues during upgrade tasks","message":"CI: Migrate RabbitMQ queues during upgrade tasks\n\nKolla Ansible enabled RabbitMQ HA queues by default, which require a\nmanual migration step [1]. Adds these to the Kayobe upgrade CI.\n\n[1] https://review.opendev.org/c/openstack/kolla-ansible/+/882825\n\nChange-Id: I82c286fd17e3a1d7f31952442fa281302cda7ee4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/75f66a08048d11ae950f8c33c97bdab16c976ec7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/75f66a08048d11ae950f8c33c97bdab16c976ec7"}]},"branch":"refs/heads/master"},"b712f07156948fdbfed8474972259cf67a7c2765":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-08-30 13:24:20.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/37/893137/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/37/893137/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/893137/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/893137/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/893137/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/37/893137/7"}}},"commit":{"parents":[{"commit":"014446251d54091c174815865e1cda925f31ffe2","subject":"Work around failing RabbitMQ precheck","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/014446251d54091c174815865e1cda925f31ffe2"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-08-30 08:37:52.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-08-30 13:24:20.000000000","tz":0},"subject":"CI: Migrate RabbitMQ queues during upgrade tasks","message":"CI: Migrate RabbitMQ queues during upgrade tasks\n\nKolla Ansible enabled RabbitMQ HA queues by default, which require a\nmanual migration step [1]. Adds these to the Kayobe upgrade CI.\n\n[1] https://review.opendev.org/c/openstack/kolla-ansible/+/882825\n\nChange-Id: I82c286fd17e3a1d7f31952442fa281302cda7ee4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b712f07156948fdbfed8474972259cf67a7c2765"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b712f07156948fdbfed8474972259cf67a7c2765"}]},"branch":"refs/heads/master","description":"Rebase"},"e10446837113bbd65d6d84b38082797953282297":{"kind":"REWORK","_number":8,"created":"2023-08-30 13:25:15.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/37/893137/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/37/893137/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/893137/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/893137/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/893137/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/37/893137/8"}}},"commit":{"parents":[{"commit":"014446251d54091c174815865e1cda925f31ffe2","subject":"Work around failing RabbitMQ precheck","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/014446251d54091c174815865e1cda925f31ffe2"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-08-30 08:37:52.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-08-30 13:25:02.000000000","tz":60},"subject":"CI: Migrate RabbitMQ queues during upgrade tasks","message":"CI: Migrate RabbitMQ queues during upgrade tasks\n\nKolla Ansible enabled RabbitMQ HA queues by default, which require a\nmanual migration step [1]. Adds these to the Kayobe upgrade CI.\n\n[1] https://review.opendev.org/c/openstack/kolla-ansible/+/882825\n\nChange-Id: I82c286fd17e3a1d7f31952442fa281302cda7ee4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e10446837113bbd65d6d84b38082797953282297"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e10446837113bbd65d6d84b38082797953282297"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
