)]}'
{"id":"openstack%2Fkayobe~893201","triplet_id":"openstack%2Fkayobe~stable%2F2023.1~I341075115442a38c327e3ade74b81b1b9e7e85c6","project":"openstack/kayobe","branch":"stable/2023.1","topic":"bifrost-speedup-stable/2023.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-09-04 11:26:06.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-09-04 09:55:11.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-09-04 09:55:40.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-09-04 11:26:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I341075115442a38c327e3ade74b81b1b9e7e85c6","subject":"Speed up calls to Bifrost","status":"MERGED","created":"2023-08-30 14:33:24.000000000","updated":"2023-09-04 11:28:18.000000000","submitted":"2023-09-04 11:26:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893201-bifrost-speedup-stable/2023.1","cherry_pick_of_change":882951,"cherry_pick_of_patch_set":4,"meta_rev_id":"747ccb171dec8defae9e9abb853f95ffdee050fb","_number":893201,"virtual_id_number":893201,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-04 11:26:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2023-08-30 15:32:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-09-04 09:55:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2023-09-04 09:55:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2023-09-04 09:55:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-30 15:32:41.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-08-30 16:07:05.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-08-30 16:07:05.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-08-30 16:07:05.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-08-30 16:27:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0def2ecd503e23192831f980226925ddfd32f531","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-08-30 14:33:24.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"cf50e97136f8753cb0d9bce1445cb2b181864724","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-30 15:32:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f22aeabca1c94c6b0a6d09a66f6622b406e96725","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 16:27:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20d6aed6cd4f4184b95bac87487b0b5f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81dca94b91684a70aa0d8d892c1c0477 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4faa449153394da88e76848fc51a81da : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e4b50bfb355446a86b4dc28b6825e4d : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5a87cf567ffd46e59fc8080247330063 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e77b58d290cb4cb49539e1d2f75a4e5d : SUCCESS in 8m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78d6d491db4d493091ce6da196b6a02f : SUCCESS in 3m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6a119ea0dfff4ffb98370b0839af6527 : SUCCESS in 4m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/774e87daf3b047d8af645c19cbe2780e : SUCCESS in 11m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c5ac49a1e2ce41f080ad041f91c029e5 : SUCCESS in 14m 39s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/29b8cdd2e0054c378f4b9aa963ec5abe : SUCCESS in 54m 17s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/b3e6603af0174cf4ba81a6eae016ebe2 : SUCCESS in 58m 19s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/065f6ce6f87c440598b6bb0f9db8cf66 : SUCCESS in 56m 52s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/040e07e7597547829554b9e6cf0f079f : SUCCESS in 1h 06m 06s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a2b9d330eb844977924d2cd31f4ef6b1 : SUCCESS in 18m 35s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/284d2d9f0034487e8f03f1ca750b874e : SUCCESS in 18m 39s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/325f27cd0fee44baa36ea2e0db302cab : SUCCESS in 16m 02s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/361ffc97c0b540ceb1d19371086ffb82 : SUCCESS in 1h 34m 38s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c669383c97ed46dc99b734e98191e774 : SUCCESS in 1h 33m 41s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/9d8f45d33ce44702b8d2a0bfba30f01f : SUCCESS in 18m 48s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/96693ec7f497451f96da8b28b2505d6b : SUCCESS in 15m 05s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/fe536c23f9d14fc5ae97c90c59bf3443 : SUCCESS in 31m 25s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/184f798302894f5c9ae50c750a05836b : SUCCESS in 27m 17s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/506a93f5ecb8412b8316c3d08e77b122 : SUCCESS in 25m 58s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/911485a30a964ac9b6395f85f736b499 : SUCCESS in 12m 46s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c71af0d57bad4838a6fb0f892174f6a0 : SUCCESS in 10m 36s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/35a0b911d7834d2aaa5a82fce202dd6f : SUCCESS in 30m 03s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/25ba36b87f7b468fa8a10d380dbe3124 : SUCCESS in 35m 18s","accounts_in_message":[],"_revision_number":1},{"id":"790d95620d6d92c10997cff34a12e6db4e41fad6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-09-04 09:55:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"27148af2644afb26975273bf0f75b3b1f9c5bc90","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-09-04 09:55:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"363c4e3b25d235dcd7b903f1b70bfe40f9e33cee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 09:56:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"55d6b08b11341d4ca2d2a3091907580929b83b33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 11:26:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f14e369ee2f44b1bbd0f7a2bf4bf1fcd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34e231452bba4c9fb3745077e46b600d : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a6b749b96204503b22ea07dc53027bf : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b518ed7479d0492cabffb416863a9fab : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abfd38e6bc7448b38d11c662af83ed36 : SUCCESS in 7m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb5591336cec495389d9285cfc352726 : SUCCESS in 3m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/eef8d84808fb42ec8f1f68ad66c3df88 : SUCCESS in 5m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/952d0aa7033d4e5ab4ae8da9fcd3c77d : SUCCESS in 11m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/95134970504e440dbeb1656bae3cf1fa : SUCCESS in 13m 46s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/7b54264f444440dfb5dd85417cfbca98 : SUCCESS in 57m 27s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/708ba222e60f490b93b1b035521cb4b2 : SUCCESS in 59m 12s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8eb4b6f415bb45d6870f7dab1f95ca67 : SUCCESS in 1h 02m 58s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ed780ba924ce4ba0b60d18326f2188e2 : SUCCESS in 18m 41s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82c2740a00844d80b48227ea299671e3 : SUCCESS in 15m 44s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ffeccde0a95f4e2f90179e9e08e1d67b : SUCCESS in 1h 27m 26s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c242da4f00ea4e0b9b8a8478bd4ba84c : SUCCESS in 1h 28m 08s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4ce6215a571e4169b8326e4783f4d5f3 : SUCCESS in 17m 39s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4563dcb212a24c3e9f10b473479912c9 : SUCCESS in 15m 07s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8f892bf5192c4b5e99d327fc3a1e280d : SUCCESS in 28m 00s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e8ce98c6887a427a8c64c74df760d6b1 : SUCCESS in 26m 57s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/cb3a09c2c6274854925617c2aaaf69cc : SUCCESS in 14m 02s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d53e7b24cf94692bde8d1e013a62a79 : SUCCESS in 12m 49s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b4a881c6cae842e6babea0669da3ea61 : SUCCESS in 28m 04s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1085c8a238f04815b719ac70d447dead : SUCCESS in 29m 21s","accounts_in_message":[],"_revision_number":1},{"id":"ea6ae2a9a861b8f07db343cc8aeb9b44c7385586","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 11:26:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"747ccb171dec8defae9e9abb853f95ffdee050fb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 11:28:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58f52a43a0644a4794ae51eaeaf79690\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b674f4e80d043a394e0527d79ac5105 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fcfee159b2814515a410af693efc4068 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c7aaecfc38cc53c15dac63e860e208e1902c47e6","revisions":{"c7aaecfc38cc53c15dac63e860e208e1902c47e6":{"kind":"REWORK","_number":1,"created":"2023-08-30 14:33:24.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/01/893201/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/893201/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/893201/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/893201/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/893201/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/893201/1"}}},"commit":{"parents":[{"commit":"4ae19a7d8438e361d434ca943cdba51c6624972a","subject":"Merge \"Fix for seed-containers being unable to use password protected registry\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4ae19a7d8438e361d434ca943cdba51c6624972a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-08-29 14:53:49.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-08-30 14:33:24.000000000","tz":0},"subject":"Speed up calls to Bifrost","message":"Speed up calls to Bifrost\n\nFor overcloud commands involving Bifrost (e.g. overcloud provision), we\nuse the Bifrost dynamic inventory. This runs bifrost_inventory.py which\ngathers node and port information for all bare metal hosts. Because this\ngets executed for each node in our inventory, it produces a large number\nof API requests to Ironic, which grows as the number of hosts increases.\n\nHalve the number of calls using the BIFROST_NODE_NAMES environment\nvariable, which restricts fetching port information to just the node\nbeing queried. bifrost_inventory.py still performs one call for each\nnode, which must be fixed separately in Bifrost [1].\n\nTime to query overcloud nodes\u0027 hardware introspection data for 44 hosts:\n\n* before patch: 823 seconds\n* after patch: 415 seconds\n\nWith the patch to Bifrost [1], this is further reduced to 46 seconds.\n\n[1] https://review.opendev.org/c/openstack/bifrost/+/882950\n\nChange-Id: I341075115442a38c327e3ade74b81b1b9e7e85c6\n(cherry picked from commit 3e03884c476604a717c1fa5fb35771f60e39dd8f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c7aaecfc38cc53c15dac63e860e208e1902c47e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c7aaecfc38cc53c15dac63e860e208e1902c47e6"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
