)]}'
{"id":"openstack%2Fkayobe~893203","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~I341075115442a38c327e3ade74b81b1b9e7e85c6","project":"openstack/kayobe","branch":"stable/yoga","topic":"bifrost-speedup-stable/yoga","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-09-04 12:05:37.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-09-04 09:55:54.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-09-04 09:56:31.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-09-04 12:05:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I341075115442a38c327e3ade74b81b1b9e7e85c6","subject":"Speed up calls to Bifrost","status":"MERGED","created":"2023-08-30 14:33:43.000000000","updated":"2023-09-04 12:06:44.000000000","submitted":"2023-09-04 12:05:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893203-bifrost-speedup-stable/yoga","cherry_pick_of_change":882951,"cherry_pick_of_patch_set":4,"meta_rev_id":"9946920afd8c03a58003f894b3c1e8234f9460a6","_number":893203,"virtual_id_number":893203,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-04 12:05:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2023-08-30 15:32:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-09-04 09:55:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2023-09-04 09:56:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2023-09-04 09:56:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-30 15:32:54.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-08-30 16:08:03.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-08-30 16:08:03.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-08-30 16:08:03.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-08-30 16:20:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"13fcec79ae5bc2d0cde0209c52aa996f6415ef01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-08-30 14:33:43.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3beaeaa2bf6beb4369b63005a76abe86d2170c6b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-30 15:32:54.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fa41df4d6a5286bffe716d9cf51ea4dcdbadfec9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 16:20:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1426ef747f0a4f639daaa5090d4945f2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1fd2d6d8ca8446f1a832eafac9a5ac2d : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22e18727211b48298da02c2034242b22 : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68b159a902bc4d71b838e76a750b9451 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/36798bb5ebe9470f99a4c1e41d24deb0 : SUCCESS in 7m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0d3adf3c6c846f8ae220a79e5b388ac : SUCCESS in 8m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53245a7a057e4f23874a7a712056716b : SUCCESS in 3m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6325f13406bb4cd8a0719b082cf49775 : SUCCESS in 5m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cc9452075e7f40b9bd2e7cde23100789 : SUCCESS in 15m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9a26b8a76499457eafad37027846d42a : SUCCESS in 17m 33s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/00b280e34c944a4c9dca62b4a45d56c9 : FAILURE in 1h 07m 03s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/4e6001a73ca8462e824d13c3a21b5394 : SUCCESS in 44m 52s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/49fe3badb1d6402c9ae95554f353fa7c : FAILURE in 59m 35s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/377b1c4857cd4a199c2255c5f05a4bbe : SUCCESS in 52m 27s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/29ac6480f04443aa809d383380e5a1a3 : SUCCESS in 26m 12s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/b8baf04f76d740e4ac0824d5cfa95d86 : SUCCESS in 22m 53s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/58a8af80ac5041009d6031c2c5fa2991 : SUCCESS in 24m 35s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/626390dc60c64683bed5e49bf2d606b6 : SUCCESS in 22m 18s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/551afb10a7354f86b0e7d6cbd2da0528 : SUCCESS in 15m 19s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55986b9ee1c34a0cbeaf22a98e1466e4 : SUCCESS in 15m 06s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6a35613265e34939b7e7d233d2b24b93 : FAILURE in 1h 22m 23s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/68eccd16698c411e88f8f88bfb540f60 : SUCCESS in 1h 11m 24s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a2cd5a95828b47639112dcfa8d2a6317 : FAILURE in 1h 09m 55s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/afed54be53d4496ba6e45a48606ccd3b : SUCCESS in 23m 41s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/b3570730c7904466b5752961577ef083 : SUCCESS in 20m 22s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a86f3ea5084e4514a404073509e860ab : SUCCESS in 19m 09s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b4468b249a842eabe0e121c6d479ac1 : SUCCESS in 15m 13s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/26d61033428b476596aba69306d702f8 : SUCCESS in 20m 12s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e1adae0a82ff4c118691b886dcc7b536 : SUCCESS in 31m 27s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eed080c7e7264cbc846de8a82715bd0c : SUCCESS in 24m 26s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/cb40daac5afb4e45a012d595f0931f47 : SUCCESS in 19m 44s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/50b9b53218d14e288fd08ed6f2925dfd : SUCCESS in 19m 29s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d79c928de863436b9edffbd9c2a6d533 : SUCCESS in 16m 14s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d50c8b71d05d43f5ab09fbb0d2ad5651 : SUCCESS in 12m 12s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/03632544c3334ee5a677a6b05646458d : SUCCESS in 12m 43s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/9e3cfd968cd54d309f06da21175e723c : SUCCESS in 19m 32s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/b48909c4ad3248cc8ca1971ef509711d : SUCCESS in 21m 13s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/21fd77c762f84bba8c05127cc3dd7a73 : SUCCESS in 18m 08s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b68af7da6a39416b9587b04e95bcf01c : SUCCESS in 16m 08s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/867b4c6372d94e4aa1e686f13afcfce0 : SUCCESS in 15m 44s","accounts_in_message":[],"_revision_number":1},{"id":"dd81364aa555949ab39bf943fa741d4328728dcf","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-09-04 07:38:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"34e3a673bb061ca6f546fb85f0c5f8739606d3bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 09:22:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6780e456b1224af4babf1f4bb97e37a9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc2a38520c274508b33aa0f62affd4e6 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5ff4094ad3646649b9d89e4d889caf4 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/973c5fd1d7c54b878d0f20f66d099d1d : SUCCESS in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e92bc6c85654c939da5ce8f7f29f49f : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c71ab7e7cba14e799cb034e962f231bf : SUCCESS in 8m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9e158f36c9a46b59ebcdc80ef5d2e34 : SUCCESS in 3m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9087d05f63b845bbb469374eea5238de : SUCCESS in 5m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/862226d1b10c4352936097c232ad0aa6 : SUCCESS in 13m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/727390a7ac8f40e584177a6ba61b5f30 : SUCCESS in 16m 26s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/1d56a71f3a3b43e6a95513e78e4525ae : SUCCESS in 1h 00m 07s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/12d9e9ffa1e84aad827a27b6476f731c : SUCCESS in 40m 18s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/e4c152ecee784dbc9cca1b0a79b95526 : SUCCESS in 1h 06m 15s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f35a6fa9d4c64bb68636901e434c10f0 : SUCCESS in 55m 33s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a1ac26947e424c54af1932575f7d9cee : SUCCESS in 21m 36s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/290d438e04154a9180bdf64191a646c3 : SUCCESS in 21m 49s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/4d1b30fed92e4ac393fdc5a525a16ce3 : SUCCESS in 22m 25s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9ccfe5826dad4442b490edabf94cc979 : SUCCESS in 20m 32s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eddd1a1ab970480bb9297f87c3d5520d : SUCCESS in 17m 24s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cf4be2bc177646c680568f1de0976716 : SUCCESS in 15m 59s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9cea347e970843bc8dee53bd4defe647 : SUCCESS in 1h 20m 41s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/90269d8dec0142a393b6840950d2720f : SUCCESS in 1h 01m 38s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cad020857e614df48b9b3c8244e115a2 : SUCCESS in 1h 37m 13s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3707ba99f30147e488eb9ff4f09cca75 : SUCCESS in 20m 21s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/e9ad9da506a94e778abe122214dd6d6f : SUCCESS in 18m 49s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d3b8b7868a7a46e5880df8fb5e54a418 : SUCCESS in 18m 49s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e8cebfc8aa148329a6b5bc448207fd1 : SUCCESS in 15m 00s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ffa2297bd8e0454e9bd51e2be1e64301 : SUCCESS in 15m 46s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/43fcf302589b46b6840f0113c85090b4 : SUCCESS in 23m 50s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5de04631f89049fc89a13c62ce3076f7 : SUCCESS in 25m 27s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b68fa75e884a4b01a626f5fd1807bbd0 : SUCCESS in 15m 40s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/4407b6059d134eb6830beef658801199 : SUCCESS in 15m 26s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/589ba864ab9e4663802d33979e566443 : SUCCESS in 14m 41s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/166c0549acb648ed8739447c92d1287b : SUCCESS in 13m 31s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30eae979e7004c73a65a5a4e330e7d5f : SUCCESS in 12m 49s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/2e8850589d1b4379ae700cabfea22620 : SUCCESS in 19m 22s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/dd1a74dbba5248a6a9f72fff5bfd05df : SUCCESS in 18m 17s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0178ec5a6a274038867abea4b6cb1e9d : SUCCESS in 18m 08s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8a468ef0bc045b2a80be1fdd57fc0d7 : SUCCESS in 13m 16s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff2bd5da18aa453db4a0bdfc6c93fbe7 : SUCCESS in 14m 44s","accounts_in_message":[],"_revision_number":1},{"id":"c50f0368a6077cd6a7496de71501f56f6e290863","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-09-04 09:55:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b8c76c35c96adf139366e5f1ec5a8af4f5df7116","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-09-04 09:56:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cd28f171489863d498e2acf660b840063f491dfa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 09:57:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"568fe33c4ad07bed1ed28986bcc9f21a0fefdde3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 12:05:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/887cee67413e4a65bfdf4201dbde6b07\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b79ccd7aa7da4749acbc126af8777e9a : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9b823d6ea3f49cdb0228b814836119b : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c842ed566bd4c199ee30a20b5c50ae6 : SUCCESS in 7m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/695c552f67af432e81458a909fc80ab0 : SUCCESS in 10m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bdc2fffa40e44950ab592d3e41d5db3f : SUCCESS in 3m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0629f0f7ff2248529bf8e562f9c93252 : SUCCESS in 6m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f5ff47d388654395b3fd3b7c4a404262 : SUCCESS in 15m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3caed974eb3242408988067df332a346 : SUCCESS in 17m 18s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/077782b84bbd47b2b7acd4378eef1913 : SUCCESS in 1h 12m 47s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/f81b7181ac6f4d58bea8e88fdf3de3ef : SUCCESS in 1h 09m 27s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0c862385dcaa467c880be46e752899ee : SUCCESS in 53m 40s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/711f936d81fe46aeaa2b76435da5c301 : SUCCESS in 24m 06s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/e6943ba832a04e84b725df53226501f7 : SUCCESS in 21m 40s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b05d34b06f5f49ceaa9dbbba8451e225 : SUCCESS in 19m 47s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba6e987a0bfd4722822cb51a6977340e : SUCCESS in 16m 11s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2fa51d468ed4193b0f9124a8dd61423 : SUCCESS in 16m 31s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/fee90db8519d4811b93933c763a7cc4a : SUCCESS in 1h 13m 02s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c0e7cba9e3c04234b308095c1c84c639 : SUCCESS in 52m 48s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4059e6dea3394926ab179428f6513a55 : SUCCESS in 2h 02m 52s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/f215f562ed5048438faa48bf780cf7ae : SUCCESS in 23m 43s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/e88b2af6e77b4a86b51ce4df0d7e6304 : SUCCESS in 20m 53s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/711585e6350d42bea439f9d5b66e8800 : SUCCESS in 20m 44s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a369256b4543480db57676a6bc40fd4e : SUCCESS in 18m 59s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8889f66f374d497a82f2771973c59734 : SUCCESS in 17m 45s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0b257071a4d645d38a094164d037f2a3 : SUCCESS in 32m 01s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/acad01dd30804ba4b9391f2ea2ddd195 : SUCCESS in 24m 33s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/1b3ba0d133fa4cb9a125cabe5f6f211c : SUCCESS in 20m 05s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/a2fcd052b12949f8af9d3a9f0ee37843 : SUCCESS in 16m 01s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3fc59389d7c4464cbe7374a9503e9e43 : SUCCESS in 16m 07s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1bad7265bf4648bc9de53d88c2d07a28 : SUCCESS in 13m 10s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/567994b0489b449183a90726410bbb3d : SUCCESS in 14m 07s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/7f578614e19a4e7790465fb516727026 : SUCCESS in 21m 14s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/2158d22c5ee448bdb7b658bcc0202892 : SUCCESS in 20m 49s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c7401bee512c47918c668e94c5821c69 : SUCCESS in 16m 43s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4de212bb60514102b1584fa14a72a18b : SUCCESS in 14m 44s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a08283ba1734662a97d75cc7950fec8 : SUCCESS in 16m 39s","accounts_in_message":[],"_revision_number":1},{"id":"0e599e2db8a19d1ee7d3f3bc4f1eedc4c0d7b9be","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 12:05:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9946920afd8c03a58003f894b3c1e8234f9460a6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 12:06:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84305176040c465188112873af1556b7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fd1325645c04c30adf928d958bb7e8e : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f386186e0164187a556f4281283cd49 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0582864d4839268e8fa781d7960a653b998f282d","revisions":{"0582864d4839268e8fa781d7960a653b998f282d":{"kind":"REWORK","_number":1,"created":"2023-08-30 14:33:43.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/03/893203/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/893203/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/893203/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/893203/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/893203/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/893203/1"}}},"commit":{"parents":[{"commit":"19aebf9cd7634a828d6e428cebcbf6c827f835e0","subject":"Merge \"CI: fetch systemd configuration including kolla-generated unit files\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/19aebf9cd7634a828d6e428cebcbf6c827f835e0"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-08-29 14:53:49.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-08-30 14:33:43.000000000","tz":0},"subject":"Speed up calls to Bifrost","message":"Speed up calls to Bifrost\n\nFor overcloud commands involving Bifrost (e.g. overcloud provision), we\nuse the Bifrost dynamic inventory. This runs bifrost_inventory.py which\ngathers node and port information for all bare metal hosts. Because this\ngets executed for each node in our inventory, it produces a large number\nof API requests to Ironic, which grows as the number of hosts increases.\n\nHalve the number of calls using the BIFROST_NODE_NAMES environment\nvariable, which restricts fetching port information to just the node\nbeing queried. bifrost_inventory.py still performs one call for each\nnode, which must be fixed separately in Bifrost [1].\n\nTime to query overcloud nodes\u0027 hardware introspection data for 44 hosts:\n\n* before patch: 823 seconds\n* after patch: 415 seconds\n\nWith the patch to Bifrost [1], this is further reduced to 46 seconds.\n\n[1] https://review.opendev.org/c/openstack/bifrost/+/882950\n\nChange-Id: I341075115442a38c327e3ade74b81b1b9e7e85c6\n(cherry picked from commit 3e03884c476604a717c1fa5fb35771f60e39dd8f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0582864d4839268e8fa781d7960a653b998f282d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0582864d4839268e8fa781d7960a653b998f282d"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
