)]}'
{"id":"openstack%2Fkayobe~893204","triplet_id":"openstack%2Fkayobe~stable%2Fxena~I341075115442a38c327e3ade74b81b1b9e7e85c6","project":"openstack/kayobe","branch":"stable/xena","topic":"bifrost-speedup-stable/xena","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-09-04 11:26:10.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-09-04 09:56:07.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-09-04 09:56:24.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-09-04 11:26:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I341075115442a38c327e3ade74b81b1b9e7e85c6","subject":"Speed up calls to Bifrost","status":"MERGED","created":"2023-08-30 14:33:51.000000000","updated":"2023-09-04 11:28:24.000000000","submitted":"2023-09-04 11:26:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893204-bifrost-speedup-stable/xena","cherry_pick_of_change":882951,"cherry_pick_of_patch_set":4,"meta_rev_id":"dd1382f476772c6149432135d7e91b445492b367","_number":893204,"virtual_id_number":893204,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-04 11:26:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2023-09-01 08:35:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-09-04 09:56:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-04 09:56:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-04 09:56:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-30 15:33:00.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-08-30 16:07:45.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-08-30 16:07:45.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-08-30 16:07:45.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-08-30 16:13:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"09e94b54d763ec9b7a59905fee30d05783559281","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-08-30 14:33:51.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"bbe62461c0a1ff1c13b59718a8fb4f29ebcc0717","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-30 15:33:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bc79ee66e8e41cc50d5b0e06268899e708e3ba23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 16:13:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/04b2f946f5334195b07cfd2ae8042f08\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fdedca2ff15b4c009a2cfbfcd31cabfe : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43b891331815456c915d7d46cbecc475 : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb5a468b5bb044489f81836e47c85b92 : SUCCESS in 7m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/907ef1bb1f8547748e60bcc9fe9c787c : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/646f1ba4bf354cd0b460ea8a2cfa0f25 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52807cd758c1437695af99e0dac87e80 : SUCCESS in 7m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4689e4a9edf04112872d07021c202229 : SUCCESS in 2m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/967cdc4db9dc458e9ca3bb1416d23c2d : SUCCESS in 6m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4f143c27bf06476da757dd3c6cd0f702 : SUCCESS in 13m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/414c57ecb2dd440c997e71bcff9a6d86 : SUCCESS in 18m 45s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9951bd6c35874c799a67e10c4853980c : FAILURE in 1h 17m 51s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/29fd7851ae034f4588a4da5546ed47a8 : FAILURE in 1h 18m 23s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ae1932176c6643e088274ec11542a02d : FAILURE in 1h 05m 36s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/8f759b88d2e5478e88855350d1829b7b : SUCCESS in 28m 56s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/5158ea072b8a49b0aa06d4fd9ed3f093 : SUCCESS in 24m 37s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11235486b1c34ff99d2f8a862cc74c16 : SUCCESS in 14m 51s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1280ba4138b94d21a8462666305c7b51 : FAILURE in 1h 02m 08s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8b84f69be68442858a5c3e6292545c04 : SUCCESS in 19m 51s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/33ba42a7181f49a79ac582838d072584 : SUCCESS in 25m 17s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92a4d3270a15407a86e1015336debb6b : SUCCESS in 16m 44s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/a6146e31a41846b3addae2ac6735ca89 : SUCCESS in 48m 26s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f69a13ed38744a3da6702860a40bea35 : SUCCESS in 36m 06s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7be64314c72b4408b4e73df970edb36d : SUCCESS in 25m 49s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/caa1197756924b2bbcaf00effef67432 : SUCCESS in 21m 24s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/3030a9c7862e4bdd8688c021dd3a8a1c : SUCCESS in 16m 28s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/428ddd64aeeb4da8b009fbddb1da741e : SUCCESS in 11m 43s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/0dfbe19b80754f7a8e8f6a577552e2eb : SUCCESS in 22m 05s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/1245d63a6c894f46a6d3a05c3d85ed55 : SUCCESS in 19m 12s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62f28f6d8f90427ea90b96e657bfbcce : SUCCESS in 13m 55s","accounts_in_message":[],"_revision_number":1},{"id":"166494ddfd92f8e57f4564f7b4e742b9f468a7c6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-09-01 06:55:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ee69edd85b3b8aa68ad308a543e4302269a35f17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 08:31:32.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/48e09f0bb36b4576883bce310ddffc51\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c88ab43d34b44781b6b7cab87047c241 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dac49dccbecc42cda9729e2928244b15 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f543658d05f44584a959c5fd5d0eeddd : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4cf037a3a16e4c78ac0d8b58b1e95ed8 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/014b4a3d36f94fc4bc73c9f52b1a7c06 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f87e235ed8742c5a0757aebc681966e : SUCCESS in 9m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ac7e5e89dd3409d814d047ad10e5310 : SUCCESS in 4m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a81ff13e817244058f768ab727fd7b57 : SUCCESS in 5m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e387be764c6b4916974c910b1eb8cd28 : SUCCESS in 12m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4c2f842b47f14976995e433250a886ce : SUCCESS in 17m 23s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3f05cbbc0ec04fde95add4aca64a73ce : SUCCESS in 1h 28m 38s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/61eeb9e80c52407b86b9b39bf33811c2 : SUCCESS in 1h 14m 37s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e2bc79e492c542259585d3de215b179a : SUCCESS in 1h 18m 58s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/9452c2c15fff47d9bc6855ebd80ea125 : SUCCESS in 27m 41s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/1a82a76687084938b6a2e2fc5cb2ec7e : SUCCESS in 23m 57s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4196b115021446db81a62b086263490 : SUCCESS in 16m 58s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/431b5a5da8ea47daa074b371c9d4e8cc : FAILURE in 1h 15m 57s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3d0601f2d3b9453b8ca54a8b395c6789 : SUCCESS in 21m 59s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/e6e9cf0651d24e64876866b9893dc743 : SUCCESS in 20m 00s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a46cdbba11684d4a90eba69cbde32012 : SUCCESS in 17m 30s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/d9aecfb7f4d644d4b2e99efae46fa117 : SUCCESS in 38m 53s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/aa99cb01450a4ffab1c7a141649610a5 : SUCCESS in 34m 21s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f633501cde94ef9a235c0aad001a05a : SUCCESS in 22m 25s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f19b683ccdec448da313ee550d3373b1 : SUCCESS in 15m 44s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/3443c7b854d84bd0943702a44d079476 : SUCCESS in 16m 47s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95be73178312474f872372f7327075b6 : SUCCESS in 11m 00s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/cc0c5537a7f94637821ce2a10ffb552a : SUCCESS in 18m 23s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/302f277a791343d4bd6cdf2967a55b0d : SUCCESS in 15m 59s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc98f9785a1c4d6ab04b7455892bbb6f : SUCCESS in 13m 06s","accounts_in_message":[],"_revision_number":1},{"id":"28a477de38d42c6d5c83298b0343503709e7d61f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-09-01 08:35:26.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6a45e5515cdb83e68990df6b73278ed84c3425fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 09:59:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56bc1da6073c4c86a2401d8d765ae7bc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/197a25e11a6d403ba8a6b9b86c004359 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/991dea44fa534d34900a599fb01e59bd : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef7c748ac2ed4e5d8ef3f25d4b62cc80 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ecc8f21e2364fec8459644fec381799 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad273e20ec36409a93c501ac0fbc6968 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c218badc71f74d8e8838df92348259f5 : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/941f8aed995342ac9c0ee0ebc19b39d4 : SUCCESS in 3m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4a0161c2c4c24e018591d813a66560d7 : SUCCESS in 4m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6c125c826c004ffd9c50ad898d8d31fb : SUCCESS in 13m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ce4cd51e7b4640b28afea9894a1fb35e : SUCCESS in 17m 18s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4347451ee04443128f8b1a553f2549bb : SUCCESS in 1h 17m 01s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/684a3209108341b28c7bf1ce55e21e9d : SUCCESS in 1h 04m 52s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/135a10c5a2d04ceb92f1c7dd481c2c3e : SUCCESS in 47m 09s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f47e21b65a794815a2abb9449b8db61b : SUCCESS in 1h 03m 23s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/69f3d51c2ff94971829bb09b738dca59 : SUCCESS in 27m 28s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/4a605b5efe004c748a63273cf04950f9 : SUCCESS in 23m 11s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4cb6b85e22824acb931bcde4251e90d7 : SUCCESS in 15m 33s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/950d3133b1d9483f8651abcab45ffbb6 : SUCCESS in 23m 20s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/b5ef947ffe9f42a3afa6c562acce8339 : SUCCESS in 21m 09s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e51079edbcd4bf38c196f132b2bdd4f : SUCCESS in 13m 14s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/6f866396d0404f2494cf7af1740b62cc : SUCCESS in 34m 46s (non-voting)\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d8bb1f270e7b4fa79bd8ecf136269837 : SUCCESS in 17m 21s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/efbc7418d0c84a2f9168caadfc894eff : SUCCESS in 12m 41s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89fe203d30714f6eadfb5bb24f89f54e : SUCCESS in 11m 20s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/be7a1b701fcb467b8c10362b2020d3ee : SUCCESS in 21m 19s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/4618d69bf0bd4b0085d3dcf34aecc552 : SUCCESS in 19m 30s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b7ebaa44d1a44709ad7d866e6244e92 : SUCCESS in 11m 55s","accounts_in_message":[],"_revision_number":2},{"id":"df44e2a449a768bd8349698eecea7522e6a03c89","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-09-04 09:56:07.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3fc0352ace0498a9ee7bf7b3543d32b94b34e41c","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-09-04 09:56:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0700f4852c98cbd05dfec39b008f28be99de1327","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 09:56:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"dfeb4e8b40107920faadb70d2a88cf6a19cebf01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 11:26:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1bbdb89c5d74f01ac01028f12464afc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/514dac6c519f4dbcb96b10586610d571 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6645717076845928cc4ee60e195792f : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ca3dfb4716d4b4c87f5d2b011bfb0c1 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12a6e9bfb884475682a9a5900ea8acad : SUCCESS in 8m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/660e1632ae0649d5b70e5468a8f31263 : SUCCESS in 3m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5009808c53ca427caba06e7765682286 : SUCCESS in 5m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0f6a58b4458e4378b34110ecb5936b0d : SUCCESS in 11m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1716025cea2e4c7db8fa26f8593234cc : SUCCESS in 17m 31s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b2b0c348b3b54419a04c9b2a5c12c96c : SUCCESS in 1h 15m 14s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/07cc1033f9734697b4d7e6021b4b743d : SUCCESS in 1h 10m 01s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18711b65d537445f959a89f08e0a998c : SUCCESS in 1h 01m 28s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/358ee6bb01d94392a8d7d091d7130900 : SUCCESS in 1h 08m 31s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0b88400184bd4d9c96ee2b28df3efae9 : SUCCESS in 27m 10s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/f8bba075dde64c63a2d78def3e9f3b5d : SUCCESS in 24m 35s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3394d6b015d741ec8b6dc3ae6823091b : SUCCESS in 14m 42s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/409c5ee5b24e4aab88cf408fd83c9144 : SUCCESS in 21m 10s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/3d0f12a0b1074628b4517454c93b06ad : SUCCESS in 16m 37s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ade70fa3957c49d89b7a3968aa0ff58c : SUCCESS in 16m 06s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/03cb6298661b4217973317a4b68893a9 : SUCCESS in 20m 45s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/503592c587a94f7ba6224dd7f78f6426 : SUCCESS in 17m 49s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8e1e2cfb3ea4a0394e4cff006281c0c : SUCCESS in 12m 43s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/4e2783d3cb08487baa20d441cecdd8d2 : SUCCESS in 21m 01s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/c3725065d69c48e5b2fc3adaa88feaab : SUCCESS in 19m 02s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88cf4bdf549f49afb50ef41810e9be96 : SUCCESS in 14m 14s","accounts_in_message":[],"_revision_number":2},{"id":"43c683493c280bee9f1d0a5a41833aece0be6e1d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 11:26:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"dd1382f476772c6149432135d7e91b445492b367","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 11:28:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fbaa92fa09e4c3299947e2ae52c9508\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/472a7068e6a34ac490ca57c68b7b3a7d : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f92448a9439442a1aadb73cd8dd02440 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c27a4362a1b01d83efce43e654aabc4c956e8540","revisions":{"51e1244d9240b790f3275321abaf45db39d6f985":{"kind":"REWORK","_number":1,"created":"2023-08-30 14:33:51.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/04/893204/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/893204/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/893204/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/893204/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/893204/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/893204/1"}}},"commit":{"parents":[{"commit":"f594ee5f256a49f2f451459fd3eb529154ff6f12","subject":"Fix for seed-containers being unable to use password protected registry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f594ee5f256a49f2f451459fd3eb529154ff6f12"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-08-29 14:53:49.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-08-30 14:33:51.000000000","tz":0},"subject":"Speed up calls to Bifrost","message":"Speed up calls to Bifrost\n\nFor overcloud commands involving Bifrost (e.g. overcloud provision), we\nuse the Bifrost dynamic inventory. This runs bifrost_inventory.py which\ngathers node and port information for all bare metal hosts. Because this\ngets executed for each node in our inventory, it produces a large number\nof API requests to Ironic, which grows as the number of hosts increases.\n\nHalve the number of calls using the BIFROST_NODE_NAMES environment\nvariable, which restricts fetching port information to just the node\nbeing queried. bifrost_inventory.py still performs one call for each\nnode, which must be fixed separately in Bifrost [1].\n\nTime to query overcloud nodes\u0027 hardware introspection data for 44 hosts:\n\n* before patch: 823 seconds\n* after patch: 415 seconds\n\nWith the patch to Bifrost [1], this is further reduced to 46 seconds.\n\n[1] https://review.opendev.org/c/openstack/bifrost/+/882950\n\nChange-Id: I341075115442a38c327e3ade74b81b1b9e7e85c6\n(cherry picked from commit 3e03884c476604a717c1fa5fb35771f60e39dd8f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/51e1244d9240b790f3275321abaf45db39d6f985"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/51e1244d9240b790f3275321abaf45db39d6f985"}]},"branch":"refs/heads/stable/xena"},"c27a4362a1b01d83efce43e654aabc4c956e8540":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-09-01 08:35:26.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/04/893204/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/893204/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/893204/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/893204/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/893204/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/893204/2"}}},"commit":{"parents":[{"commit":"55e83aff27ccf64cb41874c6d90dc2aabf1c1936","subject":"Remove upgrade jobs following Wallaby EOL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/55e83aff27ccf64cb41874c6d90dc2aabf1c1936"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-08-29 14:53:49.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-09-01 08:35:26.000000000","tz":0},"subject":"Speed up calls to Bifrost","message":"Speed up calls to Bifrost\n\nFor overcloud commands involving Bifrost (e.g. overcloud provision), we\nuse the Bifrost dynamic inventory. This runs bifrost_inventory.py which\ngathers node and port information for all bare metal hosts. Because this\ngets executed for each node in our inventory, it produces a large number\nof API requests to Ironic, which grows as the number of hosts increases.\n\nHalve the number of calls using the BIFROST_NODE_NAMES environment\nvariable, which restricts fetching port information to just the node\nbeing queried. bifrost_inventory.py still performs one call for each\nnode, which must be fixed separately in Bifrost [1].\n\nTime to query overcloud nodes\u0027 hardware introspection data for 44 hosts:\n\n* before patch: 823 seconds\n* after patch: 415 seconds\n\nWith the patch to Bifrost [1], this is further reduced to 46 seconds.\n\n[1] https://review.opendev.org/c/openstack/bifrost/+/882950\n\nChange-Id: I341075115442a38c327e3ade74b81b1b9e7e85c6\n(cherry picked from commit 3e03884c476604a717c1fa5fb35771f60e39dd8f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c27a4362a1b01d83efce43e654aabc4c956e8540"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c27a4362a1b01d83efce43e654aabc4c956e8540"}]},"branch":"refs/heads/stable/xena","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
