)]}'
{"id":"openstack%2Fkayobe~893205","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~I341075115442a38c327e3ade74b81b1b9e7e85c6","project":"openstack/kayobe","branch":"stable/wallaby","topic":"bifrost-speedup-stable/wallaby","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-08-31 13:09:03.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I341075115442a38c327e3ade74b81b1b9e7e85c6","subject":"Speed up calls to Bifrost","status":"ABANDONED","created":"2023-08-30 14:34:20.000000000","updated":"2023-08-31 13:09:03.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"cherry_pick_of_change":882951,"cherry_pick_of_patch_set":4,"meta_rev_id":"90698a1753b29a0ab1082fc735b69d5a0c8ab1f8","_number":893205,"virtual_id_number":893205,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-08-30 16:12:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":-1,"date":"2023-08-30 15:34:29.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-30 15:33:09.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-08-30 16:12:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c41fa4c7253251377aa8752aa5d6a1247a61a362","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-08-30 14:34:20.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1e73b656b9befff29ea21dfac68c7ed607e33a17","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-30 15:33:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"62423af435419fdc455e1375a5ebf667174809f0","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-30 15:34:29.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7a19d4a453679ab3b6925cb2985611ad44457d47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 16:12:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/92286c37bb544ecdb5dc739b64d235be\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0acbd0d6c694cfeb55eadde7139cc12 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c6994053af14718aa975f5d4300a200 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97df8fc3ac444b9f824ccaed29bbd666 : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5e0e618ddc944c29a652ad5db651c29 : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aada4079e74f4fe8953422e3835f1c8e : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66e5e27502a84052b0627745594201aa : SUCCESS in 8m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad7dcde055aa4c1a91c4976c611f847c : SUCCESS in 3m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6e8e12c8092e47a0be42bd9e6bb1dbeb : SUCCESS in 6m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9628909f15d749f399a010ea4698ad1b : SUCCESS in 11m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c957775d90024943be0d6a17792ef431 : SUCCESS in 17m 00s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8eb5a9a0e4c3487a8a861583cc99d029 : FAILURE in 1h 19m 56s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ab3ab426b8924628a34e14556ce2448d : FAILURE in 1h 10m 22s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/69a8c5c515024ef89d8a564b2e2f1f33 : SUCCESS in 21m 12s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/462feafedbf24e67bd7e872b5f65272c : SUCCESS in 14m 32s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/47575ce517be4701b83b671c45181602 : SUCCESS in 20m 44s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/515c42d453b1438387c6e6e07d84a060 : SUCCESS in 17m 48s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/1c4c4a482cc54086b9d70a0e1f18a2e7 : SUCCESS in 43m 15s (non-voting)\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f7aad7fae0b24b7abe0b46170e71fdc1 : SUCCESS in 17m 59s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ffa84ebee24949c6ae59f1b61eafb675 : SUCCESS in 12m 34s","accounts_in_message":[],"_revision_number":1},{"id":"90698a1753b29a0ab1082fc735b69d5a0c8ab1f8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-08-31 13:09:03.000000000","message":"Abandoned\n\nstable/wallaby is going EOL.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"18b5fa92d6bc9f616e922bdb793464c3190e2e92","revisions":{"18b5fa92d6bc9f616e922bdb793464c3190e2e92":{"kind":"REWORK","_number":1,"created":"2023-08-30 14:34:20.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/05/893205/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/893205/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/893205/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/893205/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/893205/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/893205/1"}}},"commit":{"parents":[{"commit":"b4b70123790037eea71d50a0c6d42397888a67a0","subject":"Correct name for role mrlesmithjr.manage_lvm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b4b70123790037eea71d50a0c6d42397888a67a0"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-08-29 14:53:49.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-08-30 14:34:20.000000000","tz":0},"subject":"Speed up calls to Bifrost","message":"Speed up calls to Bifrost\n\nFor overcloud commands involving Bifrost (e.g. overcloud provision), we\nuse the Bifrost dynamic inventory. This runs bifrost_inventory.py which\ngathers node and port information for all bare metal hosts. Because this\ngets executed for each node in our inventory, it produces a large number\nof API requests to Ironic, which grows as the number of hosts increases.\n\nHalve the number of calls using the BIFROST_NODE_NAMES environment\nvariable, which restricts fetching port information to just the node\nbeing queried. bifrost_inventory.py still performs one call for each\nnode, which must be fixed separately in Bifrost [1].\n\nTime to query overcloud nodes\u0027 hardware introspection data for 44 hosts:\n\n* before patch: 823 seconds\n* after patch: 415 seconds\n\nWith the patch to Bifrost [1], this is further reduced to 46 seconds.\n\n[1] https://review.opendev.org/c/openstack/bifrost/+/882950\n\nChange-Id: I341075115442a38c327e3ade74b81b1b9e7e85c6\n(cherry picked from commit 3e03884c476604a717c1fa5fb35771f60e39dd8f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/18b5fa92d6bc9f616e922bdb793464c3190e2e92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/18b5fa92d6bc9f616e922bdb793464c3190e2e92"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
