)]}'
{"id":"openstack%2Fkayobe~893688","triplet_id":"openstack%2Fkayobe~stable%2Fzed~I28f5d73d414d405d67f5fc92ab371aa2e28a4ce3","project":"openstack/kayobe","branch":"stable/zed","topic":"merge-configs-stable/zed","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-09-12 08:46:03.000000000","reason":"Change was abandoned"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-09-12 08:46:03.000000000","reason":"Change was abandoned"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-09-12 08:46:03.000000000","reason":"Change was abandoned"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-09-12 08:46:03.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I28f5d73d414d405d67f5fc92ab371aa2e28a4ce3","subject":"Pass through kolla inventories as is","status":"ABANDONED","created":"2023-09-04 20:06:59.000000000","updated":"2023-09-12 09:29:39.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"94aba8cec5b065db52c115169e573e2cc02c89c4","_number":893688,"virtual_id_number":893688,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-09-07 11:51:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"CC":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-04 20:06:59.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2023-09-04 20:08:33.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-09-04 20:08:33.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-09-04 20:08:33.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-09-04 20:08:33.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2023-09-04 21:11:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0b2731bd6ecde76aec0e13910af93837d88a4d8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-04 20:06:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d57fc760a2f353a83a20b20d099ab2b2337fa394","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 21:11:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb68177089fc443ab701bb6dec249efe\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f23701d01ca447eae46be8096428f8a : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1589c811eb141f78f0c52a7d0304a62 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ea2284f877c4e6587f180e945610220 : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d0b7eb27cf646e88d739fdb8c318f90 : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dbd37ff85307490f8b91d479b4286406 : SUCCESS in 3m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ea21c489b934e4aa4bdbbf837b60e34 : SUCCESS in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3296675e36da461a998b3aaa45826280 : SUCCESS in 2m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8d19917e0aea4d908a679901af81c24e : SUCCESS in 3m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/06de3dda1c8c45d581be154045883006 : SUCCESS in 8m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2b74323acd114f14943c1f6be8570fa4 : SUCCESS in 10m 27s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/af181c6600ec417b9f2ee66f1655c1ba : SUCCESS in 46m 02s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/8ab6595c3edd4c1683617e4986bc7f2f : SUCCESS in 37m 48s (non-voting)\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a584ecebd2af45aa8fa782b06d8170fa : SUCCESS in 59m 11s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/73b06c5164c14aefb537d3088959deaf : SUCCESS in 18m 10s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/51f14b484f1a4da999817f752aef9ce9 : SUCCESS in 14m 08s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/83d1a4e93dc44dca835d99c66ceec131 : SUCCESS in 13m 57s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7cd7e7aecf96449097a1fe8363f85171 : SUCCESS in 18m 28s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5396ac00faa8415e90eb574c7d2dc612 : SUCCESS in 14m 08s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/6d046ce67f08425aa5b7c6a2db3a9a94 : SUCCESS in 34m 33s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3fb9777b97e445acba2a93e2f3f1e07e : SUCCESS in 22m 57s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5012bc33254f404287a0a5d7076ec71e : SUCCESS in 13m 18s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e4ff8d3681148519cc925400f2bd0de : SUCCESS in 9m 19s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f070717b26c94e33a98fa77f8e574aa4 : SUCCESS in 15m 34s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c967b854e97c44f490209a78902fc466 : SUCCESS in 12m 36s","accounts_in_message":[],"_revision_number":1},{"id":"ca079d5381d19d48630645896093ae628a45183d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-07 10:53:23.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"de1c65cee167ecc37004c8f5d68d4ce027a29521","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 11:51:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afd2c3a9d3f04fff83064d79a85d8208\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6afcea294cf4c91a66aa8b63aaa23a4 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/370a0ad917f845b49226965ad8fbe1e5 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e16c48401cb24fc186c1d9bebf13dea9 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b417c45e3c064a93ae43aff30d505310 : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/096baee59bf846c3a252943229cf653f : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10b131fe44b948a28daac42a964be364 : SUCCESS in 8m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/025818811cad4cd29457b5be4c55a9a8 : SUCCESS in 3m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fd9a25d6401b44779f2ae863f28cf184 : SUCCESS in 3m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b3d078b5cfea40eb93c69bcc6684476a : SUCCESS in 10m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0aaa0d99d21d432b981d0c3c5a310099 : SUCCESS in 13m 23s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/05a1c98fc93f4bba845a1ad72cf44cc9 : SUCCESS in 49m 14s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/77349a85d6b24ce4a66201c787ba6c8b : SUCCESS in 42m 36s (non-voting)\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/4709c8f0417d4a998b93fae2a7f4bd73 : SUCCESS in 53m 24s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b3a9c352592d4ed7b5a96df540a7641c : SUCCESS in 17m 35s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/30f7b72c78d64c49a063ef62b89a3ce3 : SUCCESS in 15m 30s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/26d624175db948e987ac1c30dff80a7a : SUCCESS in 13m 58s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/dd948c073f02468990e0b1bb5aae6e1e : SUCCESS in 17m 28s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/25659b174ccd4acfacfd5864c197cfcc : SUCCESS in 16m 49s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/71279423c1144830b44fa4cb0822d3bb : SUCCESS in 33m 06s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5512a8debe9c44f8b3d556733bfe8841 : SUCCESS in 27m 06s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/de8a310c9d81472eb6836875c0c35b9f : SUCCESS in 12m 13s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e009a8ba9f8c422a86427f22453c7b8e : SUCCESS in 10m 44s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9b31c40e8764496b8feef6929f752010 : SUCCESS in 13m 58s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5676fbb0e95c4eba99563b00a912b29b : SUCCESS in 14m 39s","accounts_in_message":[],"_revision_number":2},{"id":"4c9f8f5b06ea127ae080495b80b919dadd239682","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-07 14:32:13.000000000","message":"Topic merge-configs removed","accounts_in_message":[],"_revision_number":2},{"id":"bd97be61a60ff316cc344e7a85839830727191ac","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-07 14:32:29.000000000","message":"Topic set to merge-configs/zed","accounts_in_message":[],"_revision_number":2},{"id":"28190c9182356983fe1acb47f886a416e640f57b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-07 14:32:52.000000000","message":"Topic merge-configs/zed removed","accounts_in_message":[],"_revision_number":2},{"id":"fe79780a22fa2ff654992b9cbda772aafd395d60","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-07 14:33:03.000000000","message":"Topic set to merge-configs-stable/zed","accounts_in_message":[],"_revision_number":2},{"id":"2c3af033c8234e0125efe2d4198f09966f01ba9e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-09-12 08:46:03.000000000","message":"Abandoned\n\nSorry, this feature is too large to backport beyond Antelope.","accounts_in_message":[],"_revision_number":2},{"id":"731a788f384e6561af5012f976d5bf90d83b7458","tag":"autogenerated:gerrit:restore","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-12 09:24:56.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"09e7022ef8bcffc81990c1e1026748ddeb553633","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-12 09:25:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"94aba8cec5b065db52c115169e573e2cc02c89c4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-12 09:29:39.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2710871e1271063419b8343d6e7dd63dc978bcdd","revisions":{"f365b0d2221b313c51bc2e9889f1b6b60a84b73a":{"kind":"REWORK","_number":1,"created":"2023-09-04 20:06:59.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/88/893688/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/88/893688/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/893688/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/893688/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/893688/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/88/893688/1"}}},"commit":{"parents":[{"commit":"50d6605cd541884d0442ce1f0c0e781666029bd3","subject":"Merge \"Speed up calls to Bifrost\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/50d6605cd541884d0442ce1f0c0e781666029bd3"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-07-23 18:03:11.000000000","tz":0},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-09-04 20:05:45.000000000","tz":180},"subject":"Pass through kolla inventories as is","message":"Pass through kolla inventories as is\n\nPreviously, we only supported passing through group_vars. Passing\nthrough the inventory as is allows you to use other features of ansible\ninventory such as host vars. It also simplifies the logic of merging\nmultiple inventories as we can just pass the inventory to ansible and\nlet ansible take care of the rest. This is useful for the multiple\nenvironments feature.\n\nChange-Id: I28f5d73d414d405d67f5fc92ab371aa2e28a4ce3\nStory: 2002009\nTask: 42910\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/802863\n(cherry picked from commit c75a32e72fa2d31e494b83bd406e42040e8de15d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f365b0d2221b313c51bc2e9889f1b6b60a84b73a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f365b0d2221b313c51bc2e9889f1b6b60a84b73a"}]},"branch":"refs/heads/stable/zed"},"2710871e1271063419b8343d6e7dd63dc978bcdd":{"kind":"NO_CHANGE","_number":2,"created":"2023-09-07 10:53:23.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/88/893688/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/88/893688/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/893688/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/893688/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/893688/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/88/893688/2"}}},"commit":{"parents":[{"commit":"50d6605cd541884d0442ce1f0c0e781666029bd3","subject":"Merge \"Speed up calls to Bifrost\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/50d6605cd541884d0442ce1f0c0e781666029bd3"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-07-23 18:03:11.000000000","tz":0},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-09-07 10:53:09.000000000","tz":180},"subject":"Pass through kolla inventories as is","message":"Pass through kolla inventories as is\n\nPreviously, we only supported passing through group_vars. Passing\nthrough the inventory as is allows you to use other features of ansible\ninventory such as host vars. It also simplifies the logic of merging\nmultiple inventories as we can just pass the inventory to ansible and\nlet ansible take care of the rest. This is useful for the multiple\nenvironments feature.\n\nChange-Id: I28f5d73d414d405d67f5fc92ab371aa2e28a4ce3\nStory: 2002009\nTask: 42910\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/802863\n(cherry picked from commit c75a32e72fa2d31e494b83bd406e42040e8de15d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2710871e1271063419b8343d6e7dd63dc978bcdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2710871e1271063419b8343d6e7dd63dc978bcdd"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
