)]}'
{"id":"openstack%2Fkayobe~894172","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~I28f5d73d414d405d67f5fc92ab371aa2e28a4ce3","project":"openstack/kayobe","branch":"stable/yoga","topic":"merge-configs-stable/yoga","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-09-12 08:46:11.000000000","reason":"Change was abandoned"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-09-12 08:46:11.000000000","reason":"Change was abandoned"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-09-12 08:46:11.000000000","reason":"Change was abandoned"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-09-12 08:46:11.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I28f5d73d414d405d67f5fc92ab371aa2e28a4ce3","subject":"Pass through kolla inventories as is","status":"ABANDONED","created":"2023-09-07 11:47:57.000000000","updated":"2023-09-12 08:46:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c080a0268c591db5441a408b8640dcc818f33115","_number":894172,"virtual_id_number":894172,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-09-07 14:12:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-07 11:47:57.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2023-09-07 14:12:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-07 14:31:04.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-09-07 14:31:04.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2023-09-07 14:31:04.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-09-07 14:31:04.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"ae3aa26c8806556c3d41fe26ba79651d1095f603","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-07 11:47:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b6cd9dba9689de59ec07f42920d9e621de41cbe7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 14:12:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e602f447fcc4a468534f3fcc9ce45dc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32c953aaa5bc4bb5bd63fe0fa29663cd : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90a4175c025b41d9bbd45310bc2151eb : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e8c86e33a414072bd78f3dd10f18667 : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d10a84bcb97141429a470bbca208f3e8 : SUCCESS in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64472f9d97e746d5917b594f8e39d678 : SUCCESS in 9m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aaa827fc89f34d72aaeec1143d70d2cc : SUCCESS in 3m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c0dd1340609b4e488082dad895dcf10e : SUCCESS in 6m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/90940a0cc14c4111bb743d6b3ccbe4fe : SUCCESS in 17m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f6b34c0610d5493483927477189243bd : SUCCESS in 16m 06s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f8169da6bfd542f3b91e07a35b404361 : SUCCESS in 1h 11m 32s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/92fb5f33f20045738e09853ade45e518 : SUCCESS in 57m 27s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/dbc9d9f4abe64e7db2b5f270a0da2b98 : SUCCESS in 1h 09m 45s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a2ea0525285a4626be7a68679807e26c : SUCCESS in 56m 07s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/bd171d7b01dd4ca2abc5ea7eec3d561e : SUCCESS in 25m 20s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/146153d11cbe410284b905f29cdfe427 : SUCCESS in 17m 49s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/c50cf4987ee94d8c9b69fbc0f09820aa : SUCCESS in 24m 26s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ca87ffc6199f446bbbbbadf938ca968f : SUCCESS in 20m 49s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67dc9705e8bf45c89b1aebdcd4f61bc2 : SUCCESS in 16m 02s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/406164bb6c614bf5ac3fd96e4454fbe2 : SUCCESS in 18m 06s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/0e72447721484115a2d20ba1baf61a4e : SUCCESS in 1h 09m 51s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/1a876b40af2344ed91c9f5d5d0ea2ce5 : SUCCESS in 54m 04s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bff740770b444ed5b388044d6a45f7ac : SUCCESS in 2h 23m 24s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/42cbebe06f874ee8bb745e2213f1ef67 : SUCCESS in 22m 38s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/2b62628029084d8ba2cf48944682c595 : SUCCESS in 21m 22s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4b0d4bc3b5964a9b9c76d322a0298384 : SUCCESS in 18m 43s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1e6ecf7db5d46be8d19a2b8d04f218e : SUCCESS in 20m 00s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0d0642685c0d471192c02abdb0c50c13 : SUCCESS in 15m 37s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c020e1d9f3cb4fc893abcd919d5df721 : SUCCESS in 30m 06s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db8fe8c594584a91b46a877cf1f5079c : SUCCESS in 29m 05s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/65936171fdce44d989c76d98f49d6d4c : SUCCESS in 19m 23s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/55f7d180633648e7852c146683f99115 : SUCCESS in 17m 28s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2e650473a1b243a28bf941649a0987b9 : SUCCESS in 14m 40s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa0f20b2a4f645c9995cd057588f0c9b : SUCCESS in 13m 25s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9024f6be14974151a65592176e7d0875 : SUCCESS in 13m 11s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/25561fa76d93470a8333a9cb2cc19b4e : SUCCESS in 22m 24s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/44b2dfa9a43c4f75a07fe943d16d49a7 : SUCCESS in 20m 27s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/818d88c2ba3e471c9bcef64d1a62c7ed : SUCCESS in 18m 30s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b39ab30ccb4d479c8826299c2fef47c2 : SUCCESS in 15m 08s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c819acf003be4419a21646962677cb1a : SUCCESS in 14m 07s","accounts_in_message":[],"_revision_number":1},{"id":"564cf4c55f9a8d071d951f8a1279499f282bee4d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-07 14:33:21.000000000","message":"Topic set to merge-configs-stable/yoga","accounts_in_message":[],"_revision_number":1},{"id":"c080a0268c591db5441a408b8640dcc818f33115","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-09-12 08:46:11.000000000","message":"Abandoned\n\nSorry, this feature is too large to backport beyond Antelope.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3cd6014f88610d247be1587edec5deddbc6c5d90","revisions":{"3cd6014f88610d247be1587edec5deddbc6c5d90":{"kind":"REWORK","_number":1,"created":"2023-09-07 11:47:57.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/72/894172/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/894172/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/894172/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/894172/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/894172/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/894172/1"}}},"commit":{"parents":[{"commit":"0582864d4839268e8fa781d7960a653b998f282d","subject":"Speed up calls to Bifrost","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0582864d4839268e8fa781d7960a653b998f282d"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-07-23 18:03:11.000000000","tz":0},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-09-07 11:47:46.000000000","tz":180},"subject":"Pass through kolla inventories as is","message":"Pass through kolla inventories as is\n\nPreviously, we only supported passing through group_vars. Passing\nthrough the inventory as is allows you to use other features of ansible\ninventory such as host vars. It also simplifies the logic of merging\nmultiple inventories as we can just pass the inventory to ansible and\nlet ansible take care of the rest. This is useful for the multiple\nenvironments feature.\n\nChange-Id: I28f5d73d414d405d67f5fc92ab371aa2e28a4ce3\nStory: 2002009\nTask: 42910\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/802863\n(cherry picked from commit c75a32e72fa2d31e494b83bd406e42040e8de15d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3cd6014f88610d247be1587edec5deddbc6c5d90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3cd6014f88610d247be1587edec5deddbc6c5d90"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
