)]}'
{"id":"openstack%2Fkayobe~894174","triplet_id":"openstack%2Fkayobe~stable%2F2023.1~Id4588f4529a4522e68e22ce58711cb927fa68a9d","project":"openstack/kayobe","branch":"stable/2023.1","hashtags":[],"change_id":"Id4588f4529a4522e68e22ce58711cb927fa68a9d","subject":"Use merge_configs and merge_yaml to generate Kolla custom config","status":"ABANDONED","created":"2023-09-07 11:51:27.000000000","updated":"2023-09-12 08:47:59.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"de6cc9abc2584687552586a0fc1ee9f7eb6814b7","_number":894174,"virtual_id_number":894174,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-09-07 14:34:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-07 11:51:27.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2023-09-07 14:34:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"37abcf0fd08c4a409b27c62fa8d25a232ba4da8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-07 11:51:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62f6fc409e3aa9fc6c116faa10df4f79079fa399","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 14:34:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/534f964d06ba4245a36b617888359247\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3bc8b2cee3854e6286f8519e7adfb578 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f5689ccd58549458ec0ca9c7f4a4d9f : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc1c0c2e00534953894b0e7135d1c121 : SUCCESS in 5m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c51b90539c6e4a8a8d8696bdde9086de : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fd0114ed2e14b23b275e98ce9e1375f : SUCCESS in 8m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e35351fb32241928b231d0c04ba737c : SUCCESS in 3m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3670d786444d46dc9669663730503120 : SUCCESS in 4m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d4660a35529e480c915381faf184f996 : SUCCESS in 12m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/409b67353f244ad3b7f9a04477479957 : SUCCESS in 9m 36s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0a5e0687994d46df92b7e1d4586ef633 : SUCCESS in 46m 25s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/173df313d7604896bad178a65a9efcbe : SUCCESS in 57m 05s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0a3549da36b4bbb8829cd21b7245d36 : SUCCESS in 55m 46s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/04d916d800ef486d8d3ec2a9a87444e1 : SUCCESS in 1h 05m 16s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b642b5ed70244c6ab9eed4773bd41acf : SUCCESS in 14m 35s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/2bc96a730b5c46b9b09b3a7c0e8f7548 : SUCCESS in 18m 43s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/42ed76b1863549c599612dc5c3ddcfa9 : SUCCESS in 14m 25s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9cb9d4b9c71f433b99ac8a2b9bfad10a : SUCCESS in 1h 26m 11s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0d9a0fa2e4444a6978a7c4281a8bdf8 : SUCCESS in 1h 24m 13s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/0ffa155a66b8459fb51d3d1d7d42d0e9 : SUCCESS in 19m 00s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0db675bf82de4da295f7d59f390486ee : SUCCESS in 17m 31s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/1e526c4a053843029739b3d214c8144f : SUCCESS in 32m 25s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c71cbe6c933a4f119c73f3647498f31c : SUCCESS in 24m 54s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa1f440655d74a869a72fc038095484a : SUCCESS in 27m 09s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/656727994c494ad19ca4d5084bd3ce64 : SUCCESS in 14m 14s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b8a053a56274aae9bb4c69d93a731ba : SUCCESS in 13m 47s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e0ad8b72429644348c78161d8af5deb0 : SUCCESS in 24m 15s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b21a3923c9964e4eb67c65f67c2e04a5 : SUCCESS in 24m 41s","accounts_in_message":[],"_revision_number":1},{"id":"de6cc9abc2584687552586a0fc1ee9f7eb6814b7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-09-12 08:47:59.000000000","message":"Abandoned\n\nSorry, this feature introduces backward incompatible changes and cannot be backported.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3da7b06711342f426d54538c1d699f054db3012a","revisions":{"3da7b06711342f426d54538c1d699f054db3012a":{"kind":"REWORK","_number":1,"created":"2023-09-07 11:51:27.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/74/894174/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/894174/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/894174/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/894174/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/894174/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/894174/1"}}},"commit":{"parents":[{"commit":"109d40629c484bb7f8c5cc984f2c52aec5016264","subject":"Merge \"Pass through kolla inventories as is\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/109d40629c484bb7f8c5cc984f2c52aec5016264"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-08-10 10:53:16.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-09-07 11:51:11.000000000","tz":180},"subject":"Use merge_configs and merge_yaml to generate Kolla custom config","message":"Use merge_configs and merge_yaml to generate Kolla custom config\n\nThis patch adds new functionality - merging base \u0026 environment specific\nkolla config. This allows you to place common settings in the base\nconfiguration and only keep environment specific settings in the\nenvironment directories.\n\nChange-Id: Id4588f4529a4522e68e22ce58711cb927fa68a9d\nStory: 2002009\nTask: 42903\n(cherry picked from commit 428ef10fa4d4a281f0c690876fc3cbfff41c2bbb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3da7b06711342f426d54538c1d699f054db3012a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3da7b06711342f426d54538c1d699f054db3012a"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
