)]}'
{"id":"openstack%2Fkayobe~894176","triplet_id":"openstack%2Fkayobe~stable%2Fzed~Id4588f4529a4522e68e22ce58711cb927fa68a9d","project":"openstack/kayobe","branch":"stable/zed","topic":"merge-configs","hashtags":[],"change_id":"Id4588f4529a4522e68e22ce58711cb927fa68a9d","subject":"Use merge_configs and merge_yaml to generate Kolla custom config","status":"ABANDONED","created":"2023-09-07 11:55:06.000000000","updated":"2023-09-12 09:29:29.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"64c959a2bd8477414fc5858aef748107612e1c88","_number":894176,"virtual_id_number":894176,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-09-07 12:56:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-07 11:55:06.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2023-09-07 12:56:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e9d64d5d471932d52c9e1c2076ab463978cf97a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-07 11:55:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b0c08744696705eb2403c584d0c61cf3bff81d2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 12:56:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb225d68eb264394bcaedc04afcc094d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19618567fd18457abfc830f765aac3ad : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47e33933f1a54403b7fb2a5e7eebdf35 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00bb5fb087164642a5acfc31bb0cbd09 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42f8d47e3233456791d7bd1ff8c76652 : SUCCESS in 5m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/247193652c0f42688162c74569c210bf : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaddbfe6375549eca7b8a2ea9c8dbda7 : SUCCESS in 7m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a97289ff8e645d18386e767fa3a92ff : SUCCESS in 3m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/81d2ec87975743bfba36ce2638f86dc9 : SUCCESS in 4m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c2e7ee508c3848aeaa654dc6e5f48d4d : SUCCESS in 10m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e4c4d8578e1a4c4a9d964b3157a1cead : SUCCESS in 10m 05s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a17a1b14cce44ef48c135aa4385811ad : SUCCESS in 37m 39s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/e077a8f89f9743d5b3bb3e414936007a : SUCCESS in 53m 39s (non-voting)\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/50c9e58305714a0fa49bcb7f24696c75 : SUCCESS in 47m 53s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7a708a48537443e68f17130f6b78b48b : SUCCESS in 16m 55s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/564fb7a0ea49430487a1c1c2be22e052 : SUCCESS in 17m 12s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6acf3591594e4376955d1ec35756341e : SUCCESS in 15m 24s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/abba191924ef419ba97a60f58f7f03d2 : SUCCESS in 15m 05s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5610138bc5c1454589aeb89dd3df8897 : SUCCESS in 14m 55s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/8aa0825295e5434895d2f6acbd0e0211 : SUCCESS in 36m 21s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa99d6787f304e6986a40ce171a1cc15 : SUCCESS in 26m 19s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b46b998d7f0d429f8ab5976279ad7c1f : SUCCESS in 13m 50s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/add0229a46964bb1b893831267f2edd5 : SUCCESS in 11m 04s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/49c4d9bb1c2a493a81cbd53ea535100f : SUCCESS in 17m 44s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd7cc0f34b2d4fd0acdce299cb9b2676 : SUCCESS in 14m 44s","accounts_in_message":[],"_revision_number":1},{"id":"6dbb77f2a5b3215bd4d5d0e5c185354b0f350499","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-09-12 08:47:41.000000000","message":"Abandoned\n\nSorry, this feature introduces backward incompatible changes and cannot be backported.","accounts_in_message":[],"_revision_number":1},{"id":"b8d34eb38c040d4c2c58dc09e9972050e1c2d597","tag":"autogenerated:gerrit:restore","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-12 09:25:52.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"95edfa08112ec26d08d75e08bce863843015ef51","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-12 09:26:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"64c959a2bd8477414fc5858aef748107612e1c88","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-12 09:29:29.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c090da7491f074c8e47f7b435256ac301aad6f33","revisions":{"c090da7491f074c8e47f7b435256ac301aad6f33":{"kind":"REWORK","_number":1,"created":"2023-09-07 11:55:06.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/76/894176/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/76/894176/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/894176/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/894176/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/894176/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/76/894176/1"}}},"commit":{"parents":[{"commit":"2710871e1271063419b8343d6e7dd63dc978bcdd","subject":"Pass through kolla inventories as is","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2710871e1271063419b8343d6e7dd63dc978bcdd"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-08-10 10:53:16.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-09-07 11:54:49.000000000","tz":180},"subject":"Use merge_configs and merge_yaml to generate Kolla custom config","message":"Use merge_configs and merge_yaml to generate Kolla custom config\n\nThis patch adds new functionality - merging base \u0026 environment specific\nkolla config. This allows you to place common settings in the base\nconfiguration and only keep environment specific settings in the\nenvironment directories.\n\nChange-Id: Id4588f4529a4522e68e22ce58711cb927fa68a9d\nStory: 2002009\nTask: 42903\n(cherry picked from commit 428ef10fa4d4a281f0c690876fc3cbfff41c2bbb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c090da7491f074c8e47f7b435256ac301aad6f33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c090da7491f074c8e47f7b435256ac301aad6f33"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
