)]}'
{"id":"openstack%2Fkayobe~894178","triplet_id":"openstack%2Fkayobe~stable%2F2023.1~I4d9f96ec4cf3c6cd0d28dfe5ddb239d863498a72","project":"openstack/kayobe","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-09-07 14:29:21.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I4d9f96ec4cf3c6cd0d28dfe5ddb239d863498a72","subject":"Kayobe environment dependencies","status":"ABANDONED","created":"2023-09-07 12:11:32.000000000","updated":"2023-09-12 08:48:50.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"ff3f9f13bc2823eb062c91f9a8082b95855590c9","_number":894178,"virtual_id_number":894178,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-09-07 13:06:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-07 12:11:32.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2023-09-07 13:06:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8fd87d6cf46c0d36b10948b61c0228c19e11edc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-07 12:11:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ded807fb5a8ae2abdcdc3621f3632983d53575a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 13:06:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eee8802cb15a4f3798be037a8294a48a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8cc03b7c0454ab79c099bde9b471d10 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e51c1a1164534aa1b9b2bfff3606649d : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b534f47f00347f7af6fd12f960ba5c8 : FAILURE in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/512e55050bf3417bb447850562eaafe1 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a9c77cc669c4ba996a8b92cd640ff74 : SUCCESS in 8m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d26eedfa272441da4f38eab1d513dea : SUCCESS in 3m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2fc42954e7b04ac284c16d7c711a9277 : SUCCESS in 4m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9300fa851fd04156b7cc88a7396f18f4 : SUCCESS in 11m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a64bd0d37b434149a1ad8a54a21cdb3e : SUCCESS in 12m 03s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/8f6c72a2864444069b68902ea895f497 : FAILURE in 13m 23s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/c8c9a9162342440c9ed07125bd4c12d7 : FAILURE in 17m 02s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b334e2abb614afdb0b3c170ff055b04 : FAILURE in 11m 19s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/2dc9c755944e423a99d6c20c579f4f47 : FAILURE in 10m 19s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/2d661ee6a6114df8b97e8a9a04799c5e : SUCCESS in 14m 33s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/16b92fac3ea14445a5337692dfc5fd99 : SUCCESS in 18m 43s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69aaec8fdd7e4fa7b81b625d2179389f : SUCCESS in 14m 16s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/dc0735b2c1294392926b1e6f38b595b9 : FAILURE in 50m 23s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09c638cbeb584404b514e08288ccdd20 : FAILURE in 54m 15s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/8d5cfb45e84040ea9477676f9f0f2c6b : FAILURE in 9m 22s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/185717f1d68c41af8b8b9ede2dd6585c : FAILURE in 9m 32s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/19b94feea293449dbc7b0fee9dbb35c4 : FAILURE in 13m 09s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d00495a477d2482cb3be3f499a66908b : FAILURE in 19m 09s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c63d5139b994a128c9bd081991f2063 : FAILURE in 17m 26s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3138fe63244645aa9741a8b2320358d0 : FAILURE in 11m 31s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2bd7beabffd242a8bd66e9388d08cebb : FAILURE in 12m 08s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0849e669c74c4432ae8e2c03b96a11ee : FAILURE in 12m 19s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4156032a1cc843f2b40d05f947c33a9d : FAILURE in 9m 56s","accounts_in_message":[],"_revision_number":1},{"id":"a22e5bf94d7ccbe5acc717c16e06c741f7f3e29d","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-07 14:29:21.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"ff3f9f13bc2823eb062c91f9a8082b95855590c9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-09-12 08:48:50.000000000","message":"Abandoned\n\nSorry, this feature is too large to be backported.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"da7f38de1cd3a010696ea190219db984ba9f5c25","revisions":{"da7f38de1cd3a010696ea190219db984ba9f5c25":{"kind":"REWORK","_number":1,"created":"2023-09-07 12:11:32.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/78/894178/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/78/894178/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/894178/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/894178/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/894178/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/78/894178/1"}}},"commit":{"parents":[{"commit":"3da7b06711342f426d54538c1d699f054db3012a","subject":"Use merge_configs and merge_yaml to generate Kolla custom config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3da7b06711342f426d54538c1d699f054db3012a"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-07-23 11:18:41.000000000","tz":0},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-09-07 12:10:17.000000000","tz":180},"subject":"Kayobe environment dependencies","message":"Kayobe environment dependencies\n\nAllows you to combine multiple environments by declaring any dependencies.\n\nStory: 2002009\nTask: 42911\nChange-Id: I4d9f96ec4cf3c6cd0d28dfe5ddb239d863498a72\n(cherry picked from commit 0ae9e8d4892c5e7a0fd3477452b1a134f5dc11d0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/da7f38de1cd3a010696ea190219db984ba9f5c25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/da7f38de1cd3a010696ea190219db984ba9f5c25"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
