)]}'
{"id":"openstack%2Fkayobe~894180","triplet_id":"openstack%2Fkayobe~stable%2Fzed~I4d9f96ec4cf3c6cd0d28dfe5ddb239d863498a72","project":"openstack/kayobe","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-09-07 14:28:47.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I4d9f96ec4cf3c6cd0d28dfe5ddb239d863498a72","subject":"Kayobe environment dependencies","status":"ABANDONED","created":"2023-09-07 12:17:35.000000000","updated":"2023-09-12 09:29:18.000000000","total_comment_count":2,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"f2f86bd96dbf89691984903bf4d80a36d5adda58","_number":894180,"virtual_id_number":894180,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-09-07 12:40:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-07 12:17:35.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2023-09-07 12:40:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"90b5b8d8296689e3951c4143b0da44c4984cb079","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-07 12:17:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ecc226f473cc528c134e91738ea3bc2449dfcbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 12:40:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d789565c64f42b2930b694eda9a2d78\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c19399c6dbc460389de5fab85c9af9b : FAILURE in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c27456d565ae438f868e20b937d8f5d1 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7fbe83b2d74e4dbf8f9f1584441bcfea : FAILURE in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e64d79b5314c45819ce051f0c32ad024 : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1a2711dbdf554e539b4af07f9f5abf51 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70a62632db0e412d8a3c0381ef788b5d : SUCCESS in 7m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/020cbe45e2424d268a48e0abda0a8ec0 : SUCCESS in 4m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0dd3ca20265c42ae9273ff1d4d3988ed : SUCCESS in 4m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9690652540b94488ac690e18c33bbf59 : SUCCESS in 10m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0c7b504dbd0f4164a662d0ec3ac0eea1 : SUCCESS in 10m 05s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/84a0067a22cc41adb2309cf2c1c94e4a : FAILURE in 8m 48s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/9f90b6a685974acbb0fe66f1ce671b8d : FAILURE in 10m 59s (non-voting)\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3a580d5cd40943c3b9b233991202d376 : FAILURE in 9m 00s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1ac8bf1d62604707b84c8236603084aa : FAILURE in 14m 37s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/8fd5b4ec164342d3a39a03629dc5b296 : FAILURE in 14m 24s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09c7f6b60030432bb9163a1075b136b2 : FAILURE in 13m 34s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2029d46e130b48a6938909fbdca03511 : FAILURE in 9m 43s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0eb680406bd41bda97271e80969f614 : FAILURE in 6m 34s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/e119e8d9d31049b1bfafa670aaf86d03 : FAILURE in 10m 41s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8c79afe066e4c5e94a25fac1dd34152 : FAILURE in 17m 04s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/61e49e4b6f3a4927b2567f68cac9d3a1 : FAILURE in 13m 55s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/441db8e6a9d8491db5288832decac4a5 : FAILURE in 12m 09s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f08dacb088644e0888732157a45a54f1 : FAILURE in 11m 23s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65e14ec2564d4fb5b7952d7eef31e2db : FAILURE in 11m 24s","accounts_in_message":[],"_revision_number":1},{"id":"c7d3a843a371a3d1cb5f6463f4ad5bdf4062e9ef","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-07 14:28:47.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"b02114ba1e4b1e96042220b8e3c03f23be45c67d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-09-12 08:48:39.000000000","message":"Abandoned\n\nSorry, this feature is too large to be backported.","accounts_in_message":[],"_revision_number":1},{"id":"34d6e5634b9f989697aeeb059a0427c252748c23","tag":"autogenerated:gerrit:restore","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-12 09:26:14.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"540fca9c7ee3bd1454e8b410afeea0452af148b2","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-12 09:26:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2ea80c2ed25f5ce027f6ba30a1356430ea88c3c7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-12 09:28:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f2f86bd96dbf89691984903bf4d80a36d5adda58","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-12 09:29:18.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3bae2dae033d879dedf05c7819fa28cf7477d0fa","revisions":{"3bae2dae033d879dedf05c7819fa28cf7477d0fa":{"kind":"REWORK","_number":1,"created":"2023-09-07 12:17:35.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/80/894180/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/894180/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/894180/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/894180/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/894180/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/894180/1"}}},"commit":{"parents":[{"commit":"c090da7491f074c8e47f7b435256ac301aad6f33","subject":"Use merge_configs and merge_yaml to generate Kolla custom config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c090da7491f074c8e47f7b435256ac301aad6f33"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-07-23 11:18:41.000000000","tz":0},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-09-07 12:15:43.000000000","tz":180},"subject":"Kayobe environment dependencies","message":"Kayobe environment dependencies\n\nAllows you to combine multiple environments by declaring any dependencies.\n\nStory: 2002009\nTask: 42911\nChange-Id: I4d9f96ec4cf3c6cd0d28dfe5ddb239d863498a72\n(cherry picked from commit 0ae9e8d4892c5e7a0fd3477452b1a134f5dc11d0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3bae2dae033d879dedf05c7819fa28cf7477d0fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3bae2dae033d879dedf05c7819fa28cf7477d0fa"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
