)]}'
{"id":"openstack%2Fkayobe~894870","triplet_id":"openstack%2Fkayobe~master~I5b77c92a6e7fc1bc0685fc0f7a5e46bcabcdcfca","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-09-26 14:34:30.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I5b77c92a6e7fc1bc0685fc0f7a5e46bcabcdcfca","subject":"[DNM] Test editable installs","status":"ABANDONED","created":"2023-09-13 07:30:00.000000000","updated":"2023-09-26 14:34:30.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"50c74b59ce5daeb9587329c3fe3ee55105afb91e","_number":894870,"virtual_id_number":894870,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-09-25 12:43:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-13 08:26:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"343aaa6b63c73c4633b3e00c9dd83fa7ba1c919a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-09-13 07:30:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08fe9577d4dd2fe980c83be553db81c13fc1f1de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 08:26:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/099ec2502f824919a4fa35930ad99102\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2b2de60030b4fa8beed57f1e3fa3d51 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf2cc272028742bea13d784ce9667a78 : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c90756241154e11be16dbf4f9ae7a96 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/211d56fd65764ae5965a434c15d54739 : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/66fb44f6ad54419bb167a9ddf48f4fed : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/44a4546e07764e30bad7ded50f6b670c : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ad40185f5f949bca5fcd1ab01075663 : SUCCESS in 7m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/917860a73ed246af9aff364fd5807536 : SUCCESS in 4m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e87578f4c4664b4a936f67dcf1885dc1 : SUCCESS in 11m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2616c98a71f34d56b4b2c948fd61923d : SUCCESS in 10m 55s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/93996eeceff9403daddc1f277b9703c0 : RETRY_LIMIT in 6m 12s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/265366caab374bc98370beb755726c01 : RETRY_LIMIT in 5m 20s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e4cd6bdd07494d9ea758fb2b403b16fb : RETRY_LIMIT in 4m 55s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/dc1047bc64694d578480cad35df1f022 : RETRY_LIMIT in 6m 04s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e93664f59fe34b40bea19bf3dcad587f : RETRY_LIMIT in 6m 47s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/ff5e37bb3a7442718feee7f068ba98a8 : RETRY_LIMIT in 6m 03s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/119592eb0d1d451d9b0ff990d20e451b : RETRY_LIMIT in 5m 22s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e9224a5df44b4cacbe07f1ebd71a6f81 : FAILURE in 8m 45s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e119a717ade74d3aa15a7d82d0ba130e : FAILURE in 7m 42s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2dcf9179ad11468aa72ead74a1437476 : FAILURE in 6m 22s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa6f0f7b2410459894fc5341268d72b7 : FAILURE in 7m 25s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/5f285b8ef3374a178a9ef16cf44b4f23 : FAILURE in 8m 11s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b558367ebaec48f7bd4188958921dec0 : FAILURE in 7m 08s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f8cb680681b412e821cf2464a63d8f7 : FAILURE in 7m 05s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8fa2bdf443bc4bee8a913faf1d0db417 : FAILURE in 6m 15s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ee184085ea641ff8be78f30e496c576 : FAILURE in 5m 56s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8eef3a962fca45debef36eebd5a1d648 : FAILURE in 7m 29s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f1fb9ba9fed34dc18ef6ecabd7e1e8b2 : FAILURE in 6m 11s","accounts_in_message":[],"_revision_number":1},{"id":"4050e59927a5c4e245cb6445fa4b28f055fb2fd9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-09-13 08:49:26.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9f828180fb29fb37f8557cf5e6ccb346b1fba121","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 10:01:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/29a4252184464d9888d52d22e7545386\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f72b8cb0f0649e09648b4940e684762 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b96cc41779f4610a4307a9d18f344fa : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81558757017f448a995a198114d619e2 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe71f4f6eff7472d9ad0644373b0cc65 : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f067384442a341c88a5ed8f2418cb7fa : SUCCESS in 4m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3c8e2c9864fa4a739ba1a54649b40efa : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d66b5e188d24810a0b46b29a40391c7 : SUCCESS in 7m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9d1eeaeb415f42dfa7d90f5be174f6ad : SUCCESS in 5m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/20d7488f0ba8475ba8c59869736013d9 : SUCCESS in 10m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/07079b0766744169a4656c11796cbee5 : SUCCESS in 9m 38s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/feaa6ea171bf4b10963b8ab3b55f9241 : SUCCESS in 55m 49s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/38df6be3c1bb4f4eb859d5d272bb29a2 : SUCCESS in 52m 08s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/125113d794c943d09b1d2338b5dd500d : SUCCESS in 45m 24s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/1ba0967557d943f4aa4eb4d045e7ccfd : SUCCESS in 1h 07m 32s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3c9e4b57b52c45f0a76b84330e42bcbf : SUCCESS in 19m 34s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/a30c463747fe4afe98034eae71d20dd2 : SUCCESS in 14m 53s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa2f18743e6246b58b08baa417ba4cf6 : SUCCESS in 13m 00s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a3bb94e9296147b2b0a4828056033ddd : FAILURE in 5m 39s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57fab8994a024c3da588aec451a751c5 : FAILURE in 7m 12s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/33cc35cfd2ab490a940dda8afda02b7f : SUCCESS in 17m 13s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1f376588bfee47b398d3cb39ceb324c2 : SUCCESS in 12m 52s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/988d5de2e6564a9a8039bd42f4e1d8f3 : SUCCESS in 35m 28s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/254b9b3fbc5d4b608e32c3d56328b1e4 : FAILURE in 6m 51s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b2091e2d164144f2ac3a599a4dc27740 : FAILURE in 6m 22s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b0ad3a0b30024e648b7659a4ef47eb28 : SUCCESS in 13m 47s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d41132e90924301ab2f19f14faecbc4 : SUCCESS in 8m 35s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2e133fbaa2344bb4870e40be9d1dd1bc : SUCCESS in 36m 41s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3921bfed3cc34cd7bb82d6c1609504ea : SUCCESS in 20m 11s","accounts_in_message":[],"_revision_number":2},{"id":"d9f18dedb599fd5bb94d8947c280c45b8ce2e79c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-09-13 10:02:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3e4dad7f272d95f85366e33071095c0e3d6f87c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-09-25 11:41:03.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"0547226605540c4d874a460cbb2e15cf32f9e4a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 12:43:50.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f9e9344eb5214ca88a92854b5e348ec6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/949141bce651419f8bba538861c63cab : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c56b0b7d334c4d21bcaffdf2e25b4571 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e633d8ba1ca4b2db0782c1e17c42d46 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2c9aca4f19a4312a405b04fd2074abb : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/51cd9e18cb114f7ca185189c4d70df08 : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a683b30f6cc847ceb87751953706df80 : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f55e0064153546c981745d6c26ce984f : SUCCESS in 7m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/baf161a9be6b44bf9ec83ec8173cbe4e : SUCCESS in 4m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/caf8be65246a4d45b573ba3cbca74f8a : SUCCESS in 11m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/478af35fa5b740d295f3c18afc83a8bb : SUCCESS in 9m 44s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4e698b42875d4b3ea6f7882672b669c7 : SUCCESS in 55m 09s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/8f59e88bb88b4c1098b535532a3edfb2 : SUCCESS in 57m 49s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7f118998a7a423e9f70a9d13e0bdc74 : SUCCESS in 59m 11s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c883e735c5f541529cf4190aab870b84 : SUCCESS in 53m 52s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/323a9aba71104cd2847e2bfb153bf672 : SUCCESS in 17m 51s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/4a97edb1ff7646678ca003d2671d671e : SUCCESS in 17m 30s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c4e3e7458934130b2dae9fbe8744f52 : SUCCESS in 14m 26s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/81ab26012230483eb66dc41def4e3866 : FAILURE in 5m 00s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ed0bce6d14c54dc2b6597c70c05bce7e : FAILURE in 6m 21s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d8235a1058c04b7183ba8ee09b1ba73a : SUCCESS in 18m 49s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eedf449b868844508c407b329182a979 : SUCCESS in 16m 32s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/ecc34187018546588e38c27928f4e289 : SUCCESS in 30m 03s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/92a9d8fd1fd14b89ae4beb64418bfa62 : FAILURE in 6m 32s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f125d52e2e2e42f9b5fc63e5173ae9d9 : FAILURE in 8m 25s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f008b3c26db1407180bcafad966af73d : SUCCESS in 16m 29s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6972234b225b4e2d926d51185a067ce5 : SUCCESS in 10m 03s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9148047a4b3e4ed080f928341fd0546e : SUCCESS in 25m 29s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c420188620d545cdac6ff4aab187cdf2 : SUCCESS in 28m 34s","accounts_in_message":[],"_revision_number":3},{"id":"50c74b59ce5daeb9587329c3fe3ee55105afb91e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-09-26 14:34:30.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6451c767992ffb4bbad87e40137e53cf6781634f","revisions":{"5a3706b364ac609455b1041da5c554c4b097c3bd":{"kind":"REWORK","_number":1,"created":"2023-09-13 07:30:00.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/70/894870/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/70/894870/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/894870/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/894870/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/894870/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/70/894870/1"}}},"commit":{"parents":[{"commit":"cf5feb71e20b4fbf1dbd51ca6f08f2f2215dfc24","subject":"Merge \"Kayobe environment dependencies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cf5feb71e20b4fbf1dbd51ca6f08f2f2215dfc24"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-09-13 07:29:47.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-09-13 07:29:47.000000000","tz":120},"subject":"[DNM] Test editable installs","message":"[DNM] Test editable installs\n\nChange-Id: I5b77c92a6e7fc1bc0685fc0f7a5e46bcabcdcfca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5a3706b364ac609455b1041da5c554c4b097c3bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5a3706b364ac609455b1041da5c554c4b097c3bd"}]},"branch":"refs/heads/master"},"be48b28db471db68e6e0854a853fc5342f8b8029":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-09-13 08:49:26.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/70/894870/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/70/894870/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/894870/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/894870/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/894870/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/70/894870/2"}}},"commit":{"parents":[{"commit":"e1adffa75f1da545082fd66de801747c2478b5a4","subject":"Fix data file path detection with new setuptools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e1adffa75f1da545082fd66de801747c2478b5a4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-09-13 07:29:47.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-09-13 08:49:26.000000000","tz":0},"subject":"[DNM] Test editable installs","message":"[DNM] Test editable installs\n\nChange-Id: I5b77c92a6e7fc1bc0685fc0f7a5e46bcabcdcfca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/be48b28db471db68e6e0854a853fc5342f8b8029"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/be48b28db471db68e6e0854a853fc5342f8b8029"}]},"branch":"refs/heads/master","description":"Rebase"},"6451c767992ffb4bbad87e40137e53cf6781634f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-09-25 11:41:03.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/70/894870/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/70/894870/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/894870/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/894870/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/894870/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/70/894870/3"}}},"commit":{"parents":[{"commit":"1847ad3f17a039f29e22f465508991fcb6e1cca7","subject":"Fix data file path detection with new pip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1847ad3f17a039f29e22f465508991fcb6e1cca7"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-09-13 07:29:47.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-09-25 11:41:03.000000000","tz":0},"subject":"[DNM] Test editable installs","message":"[DNM] Test editable installs\n\nChange-Id: I5b77c92a6e7fc1bc0685fc0f7a5e46bcabcdcfca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6451c767992ffb4bbad87e40137e53cf6781634f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6451c767992ffb4bbad87e40137e53cf6781634f"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
