)]}'
{"id":"openstack%2Fkayobe~895025","triplet_id":"openstack%2Fkayobe~stable%2F2023.1~If62fe0db3c107c52f80a481aa267335224456064","project":"openstack/kayobe","branch":"stable/2023.1","topic":"skydive-removal-stable/2023.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-09-14 12:46:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If62fe0db3c107c52f80a481aa267335224456064","subject":"Complete Skydive removal and Skyline addition","status":"MERGED","created":"2023-09-14 09:23:40.000000000","updated":"2023-09-14 12:47:50.000000000","submitted":"2023-09-14 12:46:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"895025-skydive-removal-stable/2023.1","cherry_pick_of_change":894799,"cherry_pick_of_patch_set":1,"meta_rev_id":"2725fc05855a5af8382a4c070f9f11e0a6c300b0","_number":895025,"virtual_id_number":895025,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-14 12:46:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-09-14 09:48:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-09-14 10:08:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-09-14 09:27:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-09-14 09:48:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-14 09:27:42.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-09-14 09:48:20.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-09-14 10:08:31.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-09-14 11:07:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9b0fc6914f13f9517aaa150c2321d75a0f45b4a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-09-14 09:23:40.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"bf0677668c1e6338604a28ea0c208193da60172c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-09-14 09:27:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ae71e446f3a158b8b6594dd397939db43868343b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-14 09:48:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"45817aa0232129438ed137c37b001fb4bfa6ad31","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-14 10:08:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b732aa56dc9966387ed7b7f0c0833124fdc1e1be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 11:07:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/497af53388e944459044e3467f5ff1fc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7c331f356d644159c9b5d843bf75691 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82dbaa9f815546e29bacbc254d0dadf5 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bea3a0eb48a74deea8bd14399e54fce9 : SUCCESS in 4m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/52fac0c112014ba8ab81de77d852ce7a : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afd80eaf523848ea968883e6f9e15e62 : SUCCESS in 6m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/635d157a9d584cd1bbc86d5a8e28b615 : SUCCESS in 4m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/792eb5420acf4fa88c95f7f635b3a0da : SUCCESS in 9m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/50797a7f0c534981a911a31bed933848 : SUCCESS in 13m 43s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/fc13f0ac4d764628affbe0e85cae1be0 : SUCCESS in 55m 17s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/adf210cbfe1d4d35a84d44be9d3b3e13 : SUCCESS in 53m 34s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d4a8ba817734570934ee2439c269a95 : SUCCESS in 58m 13s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/874c05ffe0464ddca57085bd3888b9a7 : SUCCESS in 59m 46s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6b8ead71e012475499c4c45bfa62412c : SUCCESS in 15m 41s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/5e52d08686f440bc8b7f288092155761 : SUCCESS in 17m 01s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/90a1dd99a8c748cb85230471c6e0a10c : SUCCESS in 15m 29s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f1acbdc5a8474f7d806013f95736e19b : SUCCESS in 1h 33m 54s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6e5f20fa3dc49be83b28e1df7acc800 : SUCCESS in 1h 37m 11s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a45b326ad2114e458ad3e08b3c7a41a8 : SUCCESS in 18m 36s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/90eca4ff50d944e4ba2b9c5b0eec668a : SUCCESS in 14m 13s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/167a3bc29765439da7d9d0ca6473ce50 : SUCCESS in 42m 11s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/10d62346d0f2495b8b5df502e1e62d45 : SUCCESS in 24m 29s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb5bf24fe04f413b82b5f71a54106a79 : SUCCESS in 26m 43s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/eb0ee588fae84f6f9c3632b9f2ee76a4 : SUCCESS in 13m 33s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/356fc2d6053247339b06b8bab5400746 : SUCCESS in 11m 18s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2f40ae88f689450cbe2ee3122bd45535 : SUCCESS in 30m 47s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b943b4bb3e8b4dcbab8e8c7eb632218f : SUCCESS in 25m 07s","accounts_in_message":[],"_revision_number":1},{"id":"448e5774be189b8b0c69850e050f4e2e31a355ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 11:07:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"315d5add7edecb8875cb9adfdc351a5bfa59b3e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 12:46:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e807cb4141054b1c9bb00d158265061e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27af29833c354b979ac5724e9eb3880b : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18aeabddaead43ffb51dd1aabe3731dd : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b62ac3cd7eeb4d14be33b4460e6834d0 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58db6b15aea345b2b1300b8c9fb175c8 : SUCCESS in 6m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2ea2ca97256b46108f5d5d7872036bd5 : SUCCESS in 4m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/11fa3ccc479a4361a6fd5ebfb0f199b9 : SUCCESS in 9m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/04312528963a472ab013a8685e466713 : SUCCESS in 14m 08s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/c87c65fd66834ae6b7e51b6a60adab1b : SUCCESS in 1h 01m 41s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a49f61e0455644048f92107c9256762b : SUCCESS in 48m 27s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/7f3147c561984ad89bace0db6742d330 : SUCCESS in 53m 19s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/11f728a2386e494085653d4d644652fd : SUCCESS in 12m 49s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/36fd7e7e37464058b6b1484415ad0340 : SUCCESS in 14m 46s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ab051eee57c5440e987809fac429106a : SUCCESS in 1h 35m 05s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99a718aa7b204028ae19831c1ca7e604 : SUCCESS in 1h 22m 04s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/6a9763adb63f4ba989a2c9bccaa07f53 : SUCCESS in 19m 13s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e016d66b5de547b2a552159ac8d8da74 : SUCCESS in 14m 56s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6716a3e2874a428ab3603b1143f7ca4e : SUCCESS in 28m 04s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f1bc984fd9304b28ad0ae311d6fda897 : SUCCESS in 25m 27s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3d4134a19ae74658a7e75eeb710cbf27 : SUCCESS in 14m 36s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c568c0f2e8a4431b9732d558a829b0f : SUCCESS in 12m 19s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/39c814c3f11242d6b1aef5876f6cf770 : SUCCESS in 35m 45s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/776ab396167d45d3bbe62d0c0786beac : SUCCESS in 29m 57s","accounts_in_message":[],"_revision_number":1},{"id":"39e736160fc1c51decc71355ef21d999509ec736","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 12:46:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2725fc05855a5af8382a4c070f9f11e0a6c300b0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 12:47:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/197c0324786d4498afe1e086c2ff000a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b862e97399604a53b641b48e15887217 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6ec4e8a5c7ea0a9f6c6b8a9e40c429dd5efd3a81","revisions":{"6ec4e8a5c7ea0a9f6c6b8a9e40c429dd5efd3a81":{"kind":"REWORK","_number":1,"created":"2023-09-14 09:23:40.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/25/895025/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/895025/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/895025/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/895025/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/895025/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/895025/1"}}},"commit":{"parents":[{"commit":"109d40629c484bb7f8c5cc984f2c52aec5016264","subject":"Merge \"Pass through kolla inventories as is\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/109d40629c484bb7f8c5cc984f2c52aec5016264"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-09-12 15:51:48.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-09-14 09:23:40.000000000","tz":0},"subject":"Complete Skydive removal and Skyline addition","message":"Complete Skydive removal and Skyline addition\n\nSome mentions of skydive were forgotten in the original change [1].\n\n[1] https://review.opendev.org/c/openstack/kayobe/+/884471.\n\nChange-Id: If62fe0db3c107c52f80a481aa267335224456064\n(cherry picked from commit 93ab685241c4bf6d38f7c6eb37ea9189aafdcb21)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6ec4e8a5c7ea0a9f6c6b8a9e40c429dd5efd3a81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6ec4e8a5c7ea0a9f6c6b8a9e40c429dd5efd3a81"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
