)]}'
{"id":"openstack%2Fkayobe~896380","triplet_id":"openstack%2Fkayobe~stable%2Fzed~I9dd5b97dec93c0e5393a1e7d9640f85003651b56","project":"openstack/kayobe","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-09-25 11:53:25.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-10-04 08:22:42.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-10-04 23:23:10.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-10-04 21:19:30.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-10-04 23:23:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9dd5b97dec93c0e5393a1e7d9640f85003651b56","subject":"Fix data file path detection with new pip","status":"MERGED","created":"2023-09-25 11:42:48.000000000","updated":"2023-10-04 23:24:14.000000000","submitted":"2023-10-04 23:23:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896380","meta_rev_id":"3a4ecdaa40ea3f82fb1df05de3ff6550e1ceb847","_number":896380,"virtual_id_number":896380,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-04 23:23:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-10-04 08:23:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-10-04 08:22:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-10-04 21:19:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-10-04 21:19:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-25 11:43:46.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-09-25 11:43:46.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-09-25 11:43:46.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-09-25 11:43:46.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-09-25 12:36:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"67b2384988d3983aeead7df1853fb1e6a1b66b99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-09-25 11:42:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50ed76c14049614d1e1b68956f0714c7e89c0590","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-25 11:43:46.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"645e143bf0edae722ab62fda2f63b3d557fe6121","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-25 11:53:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bffb88c618fea96d19375367827ab49ce9b250bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 12:36:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed6654290bf547188ad141e932ca4c34\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e09917246c64835a7fcf493e701f3d1 : FAILURE in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27d52113226d4005a17f9bbf4546850b : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8cbc86dc7d04c5a8845226e7af91a39 : FAILURE in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/472f3c2d1d2f475aa8e2737893ef1fbf : FAILURE in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1422f111c7f14338b7b0325913dc2b93 : FAILURE in 4m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1e8b71c817b4d579f42eaa38cda9177 : SUCCESS in 9m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19b26fbe98c84ef7baa052ecca54d8c9 : SUCCESS in 4m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f4b034ea24c249b2b158f829b129cc82 : SUCCESS in 4m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6f0e4129484d4bae97d02ff94bdafdf7 : SUCCESS in 10m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/475b689f0a6e4437a577442780f620dd : SUCCESS in 14m 42s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/673f71f437164758874832f6a14f96ad : SUCCESS in 39m 51s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/53568fd3ebc14c65aa31d232cf2763a0 : SUCCESS in 49m 35s (non-voting)\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/31ed445104fc4d6bad3e16a9837b1a48 : SUCCESS in 43m 06s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f22d33115a80403299240aa993651f1d : SUCCESS in 19m 15s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/3045741aa21c4cde9a26939d7a29dc90 : SUCCESS in 18m 27s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/49ac57546a6a430ea5b7bafac5ecbdc5 : SUCCESS in 15m 54s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/171af2d337634e06a299945db4efa941 : SUCCESS in 17m 03s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a65a230e197243189ba01b6e0816c4a5 : SUCCESS in 11m 10s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/b3bf5a8b34034648a068e7c9b2858e0e : SUCCESS in 34m 32s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc63c615ac0545fdacd8ad425ced527f : SUCCESS in 29m 41s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ab4c3c55a6774576969e632ef165ec3c : SUCCESS in 11m 59s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/331ecaebdff24242ab3224bc62673c8d : SUCCESS in 10m 49s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/04ddce43bdfe4a7c9ae64cf38aec87cf : SUCCESS in 17m 01s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee47845816d54749b84794b36841956f : SUCCESS in 14m 10s","accounts_in_message":[],"_revision_number":1},{"id":"db829a9853d6b88ebd2db4e2ffe841ed011c7b28","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-25 13:13:37.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"6c0b3b344842477ded31afadc74c0d6e73210ef8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-04 07:59:00.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f125841b209133b7a8259acb16cae03f902d7e86","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-10-04 08:22:42.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9e3e40e3ff42a2b015b4812eda482e373ab54bf6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-04 08:23:33.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7f9ced80a3eee1fe9d013e3d096f50c0715947f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 08:58:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c577bc4b61d4d1897bd459e33b58e67\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6e4fbf0112b41ac95d2e41be1e1756a : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1d3e08a0b484707b33dbfab3c98ac2f : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48d117f3e8eb494fbe6e913771bad3f8 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc0ca26d453e423b859b0874baa3d0e7 : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3408def805d8405d98e69365ee950aea : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82b1840cb0044c90b943506267dd8149 : SUCCESS in 7m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cfb77ccf478145939f2f8621002c29af : SUCCESS in 3m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/52d395fbdd374bdd8b45f94470ad0a1b : SUCCESS in 4m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0d4b50d07c224390971b213bdaa174e4 : SUCCESS in 11m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3633e71c6dc84987a84cb1f188b6c3d9 : SUCCESS in 12m 22s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a0f9a3ec7d6d469cbb39acf8bf8831ab : SUCCESS in 51m 25s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/8f2673baa8c54cf6a70f30ec326ebfd4 : SUCCESS in 50m 51s (non-voting)\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/0b25bd6ac3d24234bb8471512d7248a7 : SUCCESS in 58m 07s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/bb64e4c08eb5450ea12a5d93d77e4624 : SUCCESS in 19m 19s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/106df18f52614c2c8f1c39ed0f1c121b : SUCCESS in 17m 08s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e4fac6f031d2403586a1d3e9697108b2 : SUCCESS in 15m 49s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/89863f39e86a4a95a13711e4c4d772b6 : SUCCESS in 16m 51s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e8ee56f8c2d48c1a9c707c3cfebaba2 : SUCCESS in 16m 18s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/955a8bbad86d4e7988d4f182b2c1f955 : SUCCESS in 33m 58s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64616793719e4376821a956400317f0f : SUCCESS in 23m 08s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8e76e7de981b4bda8341d9e5af55386c : SUCCESS in 12m 38s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/77c22821319643a9ad8954bea83a17a5 : SUCCESS in 9m 53s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/921310dcef06495abd46f9b5af1487b0 : SUCCESS in 16m 34s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/635df09ce92146f28014a00849339ba8 : SUCCESS in 13m 53s","accounts_in_message":[],"_revision_number":2},{"id":"afe041a735567d458fe9fd513ae2cab4a7d8bf2b","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-10-04 21:19:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"da1fd261583573c00490b42f38938f1bd6eec465","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 21:19:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"035b592ef2a9c049730f1812516787403d0c1343","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 23:23:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/017a48c0cdd74bc799af6eedebf36494\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb0a95aef58444869c2e2bc0d2f90187 : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d91d6ffdc8f4762a11c9c425eb820ad : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29afdfda6b364eed87932f31ffcae5ca : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfd494e6f01a43828451e6648a2ef6e1 : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31b9ccf9eab0436ba1dae0d2af62bf32 : SUCCESS in 4m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7f72f23434b84f7696670a9c70bac066 : SUCCESS in 4m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c10b833ce49c4a1db83e094fc0522888 : SUCCESS in 10m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8a758b45a07a41a6b356098628c23adf : SUCCESS in 10m 19s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e29d9e7d52da40be874609f57d4f500a : SUCCESS in 48m 43s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/80ee27134d35419da4b820acefddf791 : SUCCESS in 56m 05s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/12d179e30e134a538241dda8ef8f660f : SUCCESS in 13m 03s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee266a67ddb749abb2884ee9c00edb89 : SUCCESS in 11m 16s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2c5ef3b90a5b4631a2bb90cbf54878cf : SUCCESS in 19m 37s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0623139b0c004aee9aaee71a997e1784 : SUCCESS in 11m 40s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae94893e4a45420a8a96c5d8a9f387f2 : SUCCESS in 27m 54s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c4a18511558744bf9017a9fcc9d51293 : SUCCESS in 10m 08s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/79af9411e98c4850a54f722979a9d2b9 : SUCCESS in 8m 50s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/023150a34ee3492ea5eb401fb77a07db : SUCCESS in 17m 24s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87a480a010dc40529e3509146ee261d3 : SUCCESS in 10m 57s","accounts_in_message":[],"_revision_number":2},{"id":"44d350a3d244d741627b825e28dd105ccbb0b6d0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 23:23:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3a4ecdaa40ea3f82fb1df05de3ff6550e1ceb847","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 23:24:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75d8cc9a081e499193a399fbb4b09835\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d50cf8d8b583498baab487cd78adccd8 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01ad3091bb4b4aeeab97d3735942db66 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"77a0e36fbf430706951138fbc4d187417ece20d3","revisions":{"2fdf78f163510240e91c1025c75da6fdfd877bbd":{"kind":"REWORK","_number":1,"created":"2023-09-25 11:42:48.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/80/896380/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/896380/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/896380/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/896380/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/896380/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/896380/1"}}},"commit":{"parents":[{"commit":"b172d87b31adbcc8a81d13f2563a8d2572a57489","subject":"Merge \"Limit interface names to 15 characters\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b172d87b31adbcc8a81d13f2563a8d2572a57489"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-09-21 16:29:23.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-09-25 11:42:40.000000000","tz":120},"subject":"Fix data file path detection with new pip","message":"Fix data file path detection with new pip\n\nUsing an editable installation of Kayobe fails on Rocky Linux 9 or\nUbuntu with an error such as:\n\n    ERROR! The requirements file \u0027/home/rocky/kayobe/kayobe/requirements.yml\u0027 does not exist.\n    Failed to install Ansible roles from /home/rocky/kayobe/kayobe/utils.py/../requirements.yml via Ansible Galaxy: returncode 1\n    Control host bootstrap failed - likely Ansible Galaxy flakiness. Sleeping 5 seconds before retrying\n\nThis is caused by recent changes to how pip manages editable\ninstallations. The egg-link file that Kayobe was using to find the\nsource path does not exist anymore. Instead, there is a direct_url.json\nfile under the kayobe dist-info directory that can be parsed.\n\nChange-Id: I9dd5b97dec93c0e5393a1e7d9640f85003651b56\nCloses-Bug: #2020135\n(cherry picked from commit 1847ad3f17a039f29e22f465508991fcb6e1cca7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2fdf78f163510240e91c1025c75da6fdfd877bbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2fdf78f163510240e91c1025c75da6fdfd877bbd"}]},"branch":"refs/heads/stable/zed"},"77a0e36fbf430706951138fbc4d187417ece20d3":{"kind":"REWORK","_number":2,"created":"2023-10-04 07:59:00.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/80/896380/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/896380/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/896380/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/896380/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/896380/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/896380/2"}}},"commit":{"parents":[{"commit":"b172d87b31adbcc8a81d13f2563a8d2572a57489","subject":"Merge \"Limit interface names to 15 characters\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b172d87b31adbcc8a81d13f2563a8d2572a57489"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-09-21 16:29:23.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-09-29 05:11:43.000000000","tz":120},"subject":"Fix data file path detection with new pip","message":"Fix data file path detection with new pip\n\nUsing an editable installation of Kayobe fails on Rocky Linux 9 or\nUbuntu with an error such as:\n\n    ERROR! The requirements file \u0027/home/rocky/kayobe/kayobe/requirements.yml\u0027 does not exist.\n    Failed to install Ansible roles from /home/rocky/kayobe/kayobe/utils.py/../requirements.yml via Ansible Galaxy: returncode 1\n    Control host bootstrap failed - likely Ansible Galaxy flakiness. Sleeping 5 seconds before retrying\n\nThis is caused by recent changes to how pip manages editable\ninstallations. The egg-link file that Kayobe was using to find the\nsource path does not exist anymore. Instead, there is a direct_url.json\nfile under the kayobe dist-info directory that can be parsed.\n\nThis also includes Ic53efd03cecbd53ad3e3b64b664e084f4e25be0e to work\naround mocking issues in unit tests.\n\nChange-Id: I9dd5b97dec93c0e5393a1e7d9640f85003651b56\nCloses-Bug: #2020135\n(cherry picked from commit 1847ad3f17a039f29e22f465508991fcb6e1cca7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/77a0e36fbf430706951138fbc4d187417ece20d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/77a0e36fbf430706951138fbc4d187417ece20d3"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
