)]}'
{"id":"openstack%2Fkayobe~896394","triplet_id":"openstack%2Fkayobe~master~I19f188cc002f8578618003e90c0a4a154b806e49","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2024-02-06 12:33:37.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"36216":{"account":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"last_update":"2024-02-16 19:36:21.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-02-06 17:24:35.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"36201":{"account":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"last_update":"2024-02-16 19:36:21.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2024-02-16 19:36:21.000000000","reason":"Change was submitted"},"36238":{"account":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"last_update":"2024-02-07 14:24:17.000000000","reason":"\u003cGERRIT_ACCOUNT_36238\u003e replied on the change","reason_account":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"}},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2024-02-16 19:36:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I19f188cc002f8578618003e90c0a4a154b806e49","subject":"Generate KA config before ``kolla ansible run``","status":"MERGED","created":"2023-09-25 13:04:27.000000000","updated":"2024-02-16 20:07:27.000000000","submitted":"2024-02-16 19:36:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"896394","meta_rev_id":"9552517e069281c0152e9aef0bcbc5a8d067a34f","_number":896394,"virtual_id_number":896394,"owner":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-16 19:36:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2024-02-06 12:33:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2024-02-06 17:24:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-02-08 08:40:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2024-02-07 14:24:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-02-08 08:40:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-25 15:04:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-20 16:16:19.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-10-25 15:39:44.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-12-22 16:00:11.000000000","updated_by":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"reviewer":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"state":"REVIEWER"},{"updated":"2023-12-29 03:50:03.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"},{"updated":"2023-12-29 08:29:15.000000000","updated_by":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"reviewer":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"state":"REVIEWER"},{"updated":"2024-02-02 13:13:30.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2024-02-06 12:33:37.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2024-02-08 08:40:17.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"4fb638061bdd6578b41edfc6fb783b0d6201267a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-09-25 13:04:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4071c2beaf91439f1329ca55c53f0af0c284b3b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 15:04:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fedeb65c8e4c452da883dd20b02478d7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3350841e8ac94dfcb4cd572d7d25e4d2 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c466643ecbf64b4a90a8dfdd84980062 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eaaf7ae84aef429c836d73c4c07962b9 : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea3afbcc74ad48d58d8e604a68b870fc : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b497533ed7ec4df38e32d5bf463aed0e : SUCCESS in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0f6fd35572414f3f89e696332b4ae5d4 : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb50dbc1fc794b3ca768991397d26cbf : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/defa3b4fbe244c28919f6c27ec81287c : SUCCESS in 3m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6a676a192ce5402bb1e82721a306b814 : SUCCESS in 4m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/010f28c32d614b32a7c2f536b33e1305 : SUCCESS in 12m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/34d28bc29aad47c898103a23a0c7477a : SUCCESS in 12m 38s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/93b551d6b67f470b9a445b73871f8af1 : SUCCESS in 56m 57s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/d405c80b1b914d20a9b89de1e9f9c998 : SUCCESS in 59m 21s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/330fd662130f40fc9c6fdd71e01e5e79 : SUCCESS in 1h 00m 25s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8ba6036dcd8a4c36907a9f049ff417ad : FAILURE in 9m 34s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/35e9b2bb27864f0c815f2f955730ad74 : SUCCESS in 17m 54s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/1c11439488e2478fbe31406f0ea9cec0 : SUCCESS in 15m 44s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f86b029e306149f4a15ea15c6fdddd03 : SUCCESS in 15m 38s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6fc97f613a82492aa22317829df6e5fa : SUCCESS in 1h 56m 16s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fd262c0495f84db195d1a3cb365aa8a3 : SUCCESS in 1h 58m 57s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/938f43ed7cb44c0691496b62a1b8049d : SUCCESS in 16m 14s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5578f31838164e5fa4df558c439b06f7 : SUCCESS in 13m 18s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/b24db726efe341cdbf15b6a86a03d54e : SUCCESS in 30m 28s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/757ce8986f4f478db8d11aabb981e785 : SUCCESS in 25m 33s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/514ddaeccf6a4bd8bc7e276996dec65e : SUCCESS in 24m 52s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/64652763942648fea8f4052f27524b73 : SUCCESS in 13m 10s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ed0ee9c6a58469dafe72d4633440976 : SUCCESS in 12m 47s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9c11fd93235e435c96e621af392136d9 : SUCCESS in 36m 13s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76b7353813c34f3191eb398b0cfae2cc : SUCCESS in 20m 01s","accounts_in_message":[],"_revision_number":1},{"id":"a9d8c5955aef70ec1e1cfa04a8fc0a357df626ac","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-09-27 09:01:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"37c1eb6bd7397ef685c3ae7a0f627525aa638a3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 11:24:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/80152cb012cb49bcb553bbf0fa9e0742\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0ffb49a2ca74bf59afa0e2af7784878 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53c7f5d9c5df488fb9bf409ac270c95d : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed55347e1a9545ddbb01f54e6261d60d : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6db1865b3b0a40708e28a56a22ecb06d : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d57dcef6f9b343899739b93f12ef4546 : SUCCESS in 5m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7ba3f1d98d94401b95404755d81a4112 : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8915a14655b0451892777c0c04c8fc27 : SUCCESS in 9m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd931624a89f44f69ea55bbd989cea5f : SUCCESS in 3m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6b57e8c0b3f2485cb3062e4d545267dd : SUCCESS in 4m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/591482fcb18649c4852d2e7ae1e99c48 : SUCCESS in 9m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b72e7e4ff8004fdb87453dae6b4e6c9d : SUCCESS in 12m 04s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/79d2f523cb6e407488f2530483918eb2 : RETRY_LIMIT in 4m 40s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/7de9cc7845f44362b02a6e9da75dab3a : SUCCESS in 54m 07s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/116a297885ca43ea81236bb417df48f0 : SUCCESS in 1h 01m 43s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/f5222f130faa4a578d955effb267acae : RETRY_LIMIT in 4m 43s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/dd0782c90cd74fb8aa79e7e30b3db80c : RETRY_LIMIT in 4m 41s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/c39c9e1b34b242a6b9988e61e353204a : SUCCESS in 15m 23s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b093133ce5574da3ac1669ad2ffdb64e : SUCCESS in 13m 49s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/33c3532a181848c2b167f7e15d2c7da4 : FAILURE in 10m 00s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/35935f4a109c48e0912e802a66b83fb5 : SUCCESS in 2h 16m 36s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/abaf1a932dc945439c02461709557927 : FAILURE in 18m 47s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/548e7f6900d8434494373080b704490b : SUCCESS in 15m 40s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/dc1511cf2b5c41b6adba3b9935d10583 : RETRY_LIMIT in 11m 37s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c168744cf2834b0b819a132a99074eaa : FAILURE in 8m 54s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33946bf6e62241a2aa04f761aa11c7fb : SUCCESS in 23m 05s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/60329c864a064d499514c8ae7536f70f : RETRY_LIMIT in 6m 17s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/abc1efa899ea4211b869e89560856c56 : SUCCESS in 11m 10s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bfc76aaae19e41218cd459d9979ff97e : RETRY_LIMIT in 4m 25s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a096a1b8142945f8a247cd4bf064ed08 : SUCCESS in 21m 33s","accounts_in_message":[],"_revision_number":1},{"id":"4d0a6b7d221fcaaba2d2a0788af137860113ab01","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-09-27 15:28:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2b0b92199b00867ae9cc28ba3c00288de25b899f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 18:11:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/12f55b9fd66848d7ac349f4ea1cfa2ae\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b727162a1a342caa4a987c186ab9782 : SUCCESS in 7m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16ff477f34e340f58e9ab025f0ad3902 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c99c24a0a4cc4d549fe2a02450c94de0 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/429515b7a1e14fd8a1ec73a49d7852cb : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6584f124a654411c8b055b1418ed5182 : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/392f003a50734a60a963f799bd72051a : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edc232218a8e405f9dd16788f9ef710d : SUCCESS in 8m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7da39a7d78f84fc8a639d6b6d1dd6902 : SUCCESS in 3m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/66efa8e9f5e341dd8b18212e8c905e58 : SUCCESS in 4m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/438885d1af63408fba6baeab0aaf5c23 : SUCCESS in 11m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/25b06bc66ece45ce83407815c0ec0bdc : SUCCESS in 10m 43s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/082dff216d0244b590478494f07a6479 : SUCCESS in 1h 00m 22s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/9cf9bbdf97994cb28d765c47aad873d7 : SUCCESS in 1h 01m 05s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d881ea8aa36d4a2e9676113e5de75073 : SUCCESS in 56m 59s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/875425e701a8438f804296487daec011 : FAILURE in 10m 39s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/255930e6bfe64b94bd85eda9be0dd5e5 : SUCCESS in 21m 33s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/06cba8fd49b3468f830f928a53716a0d : SUCCESS in 18m 36s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/afda21f7e3ce485f85e203a2ea111a2c : SUCCESS in 17m 26s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/676477e288a9406f90873b80b1b818a1 : SUCCESS in 2h 22m 34s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a67fefc8971f4ca28095a02b67f0c6a1 : TIMED_OUT in 2h 33m 56s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/0e27324338924b83afb8cb69783288ca : SUCCESS in 18m 17s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ae7aa6ff16c40888aa54d8336ebf6d7 : SUCCESS in 16m 36s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f7deec24e4e444aba425a1d8f9e343ac : SUCCESS in 32m 57s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7ec73597e08e46ba9c02ea7221197d1a : SUCCESS in 32m 30s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a44039a1d4642f6b110bc6b856d3372 : SUCCESS in 25m 36s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/360609bdbdb1429db3c49fc93e070cda : SUCCESS in 12m 17s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1dffe408205547c1825837c2fe0e26c3 : SUCCESS in 9m 46s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0a54562d0d014b239830ec526e0bafa7 : SUCCESS in 30m 57s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7679701b757432e8753eaa757668089 : SUCCESS in 22m 17s","accounts_in_message":[],"_revision_number":1},{"id":"81551d144e064b3c9165e1a59e0994752220e370","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-10-09 10:53:40.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"fb362ca3ecae8bb68d4b5b7dc5b4d866380057c7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-10-25 15:39:44.000000000","message":"Patch Set 1: Code-Review+2\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"7369e008badde936b563e6d08a0d25713c0b3d21","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-10-25 15:39:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8b913c3f4020565692ca24a79a580748bf42b1a2","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-10-25 16:23:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"66f714d25a2866463908f59e8aab0020afcd7779","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-10-25 16:26:14.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"6cfefcb6d0bb98874d52d7fd2327962ee8518f6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 18:25:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/70837b8bbd594f5e8fa7b29675b91d82\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a189abb4368143daae10f96a11f6153e : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a0e8e41983c484990e56bd596ba0190 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94b34b60cfd145dc8b8530e75a5176d1 : SUCCESS in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/795cbdbbfda14243a6af865cd9e56117 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1905dc8d691f43c6b4b62a10f96f7757 : SUCCESS in 8m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a32a49f399d5424fa77a61e41e00b437 : SUCCESS in 3m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f713cdfaf4b64f7eadca37cd34b3a956 : SUCCESS in 4m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8f071ae0001449c38ad4e93c942b741f : SUCCESS in 10m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f332bfb34d0d475b99a9a7c05bc71182 : SUCCESS in 9m 27s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ff32f5db79674f308fd7c04c70e2c47d : SUCCESS in 52m 28s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/5c281127ab31443b972fb3850e7e738e : SUCCESS in 1h 06m 38s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/46853e998091484ab8cccebcc9c894c0 : SUCCESS in 58m 14s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/6e6c5dd515f34fd2a48a4fb729281baf : FAILURE in 9m 10s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/71980e4d8864422e8c5d5f11aed12b0b : SUCCESS in 16m 44s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/e3a899f70eeb48f69c2eeb1144258f4b : SUCCESS in 16m 52s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/960d2f00fb4e46afa5c124bf133c9935 : SUCCESS in 13m 07s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d5031d721d2341c0957af3123501c7e8 : SUCCESS in 2h 38m 47s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02737c32beda4757825fbfd7a24729a5 : SUCCESS in 2h 15m 26s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ad18a2e872ab4e1cbe3758c555062442 : SUCCESS in 17m 26s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e540cb1fb744444ac4957a52931c867 : SUCCESS in 15m 08s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/a9a4838f4de647fe9498ded98c9c6119 : SUCCESS in 36m 01s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/66b5802796fd4dbdb5010b8e6bdc9255 : SUCCESS in 31m 26s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/309da513713e43e08e08a49d68969089 : SUCCESS in 26m 11s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4c93a07174ef4920b80f03bda2455f0f : SUCCESS in 16m 22s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6434fad830a348f99453b404fc906a05 : SUCCESS in 11m 01s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c023ca02502d4b77afc79b53aacafef1 : SUCCESS in 17m 36s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/852fba5d971d4382939a390f3af65c0f : SUCCESS in 15m 30s","accounts_in_message":[],"_revision_number":1},{"id":"ab4a353dcea55482661e0ebf046cac12e357ba86","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-10-25 18:35:54.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"130fd20aba447de7246c1f7ad8d28d8c4c714122","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-10-25 21:48:10.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"06c7e66a0f9a3b4ee00add93236bd71198816cfd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-10-31 16:09:27.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"eb4f2de438636cdc2b42dcb2c09943a23b4e55a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 18:20:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e5ec3bfdcf0d4c68b6f2fcfb8d3c9f9a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7570d38d7e164a098ceea797bf9e5638 : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b347ec2a7b14899b8393f9316951da3 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7561832e5ae34d029425da5b9f35dc77 : SUCCESS in 4m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ddb048a1e5944ce78fb5b70589ed2bfe : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84c79d96c26c4fa2886e013f8d7b37d3 : SUCCESS in 8m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe076f8187bf45f6b61bc91e17bd44e7 : SUCCESS in 4m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c305fe1ceb9941ba8f51661617fb35c1 : SUCCESS in 5m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e96437b1bdc7404286fa923d36e66b4b : SUCCESS in 10m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/763751c3af5240a388356bb8e8c012d2 : SUCCESS in 12m 58s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4d2d9b88f6eb403abcb4a924c7ec0a9e : SUCCESS in 59m 01s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/0ecc1a9575854bed8d933d69cee16e9d : SUCCESS in 1h 00m 04s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd56c077164e46168120a2df13ce8117 : SUCCESS in 52m 18s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e0545c5cb3064202a439100165480800 : FAILURE in 8m 45s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/421c672a9329401fab7e19229471322c : SUCCESS in 14m 14s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/211130db135f40dd92d47504d33f489f : SUCCESS in 16m 08s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/583c18f5023c4e8180e4de60262b5d9f : SUCCESS in 12m 14s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b0ae9d2a54b0420581fb76fe17fdd7d4 : SUCCESS in 2h 04m 05s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d7dec1a1335a4eab8ba1787b45c8c9de : SUCCESS in 2h 00m 16s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e48791e8a5b148038ee231ff58a04b7f : SUCCESS in 16m 02s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c5dd1d0f83c442db3c963a1591102bb : SUCCESS in 13m 13s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/7dd6ef4de10c4560a032ff198197e903 : SUCCESS in 33m 23s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3fb040bf1d4d43fdb8d0835ec383e411 : SUCCESS in 30m 17s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f120fd10a42442e90bf4ae077270566 : SUCCESS in 24m 10s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e81f67cfa44e4526a1458851a1e91cf9 : SUCCESS in 12m 50s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/802b1692206143b3b5780e92db4b9de3 : SUCCESS in 10m 27s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9cb46158bff8442bbede1859b3905b15 : SUCCESS in 16m 07s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c2935b628c974e64a61f376f22b4b532 : SUCCESS in 14m 14s","accounts_in_message":[],"_revision_number":2},{"id":"718501c0d0fab9800f6efa4b8c1891acdf5505c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-11-02 13:35:23.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"6ab0153852acd177977bfef3277a85616341caab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 16:06:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b10063266b19454f9fcd2a4c8fcc2a05\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db61a6c73c9843bca3139f82f0650c34 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d163efb44c2a4c1193f59a08e0838365 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a126edfc718444bb8b99ac0273ec51eb : SUCCESS in 3m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/04f450dcf79648cd848eeb605eff3551 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f919fa75ffb04471a5e1cbca82cd3f7d : SUCCESS in 8m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ce1539df1854d598ed31f3c5ecaf2c1 : SUCCESS in 2m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/06a0c9c6819941fb867cd9a97d77158a : SUCCESS in 4m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d053ab727f4940f584f7b1be6c36f2db : SUCCESS in 10m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6eff7caf6e7442d8b70e04758d48b3b1 : SUCCESS in 10m 42s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ab735a017bf34130ab424f17240d5029 : SUCCESS in 1h 12m 38s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/2108a29f29c3494492f19a7e5359dade : SUCCESS in 1h 08m 50s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/41355ad83b624dcd980d7928176c2720 : SUCCESS in 45m 44s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/2ee7fd63ca2f4e2bb9ddba435fd9ace2 : FAILURE in 8m 46s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/164fc7fed4fd4d978a7cb1f238c08f38 : SUCCESS in 17m 58s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/81089fdf067246f385d1942bb4963852 : SUCCESS in 16m 14s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4eda0dce18cf47e2aa47f49ec420e19a : SUCCESS in 12m 15s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3962c693651d4735b6726ef0806f4751 : SUCCESS in 2h 25m 30s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb60aed38bde47949c685bf1dbc1c913 : SUCCESS in 1h 54m 52s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4d25d61a3746413d9a5d8d12975a3d59 : SUCCESS in 17m 35s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/113429bde74740828a57aaafbdf7323f : SUCCESS in 14m 37s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/34e0e966a28c4a1bae32eedd38abf55c : SUCCESS in 30m 54s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3557aefa8d5f44edbf3571d3d603a02a : SUCCESS in 28m 52s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ffad35b7e1db471faad48e6016e5aa86 : SUCCESS in 20m 48s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a071efdcfe9846928fd12ad9c0540750 : SUCCESS in 12m 40s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47e2761a70924e028099b8c8cd8274dd : SUCCESS in 12m 00s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f7d3640966da47e5a2ef001546fc7a17 : SUCCESS in 18m 08s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9db5533557c44e5da561e8e93168f8cb : SUCCESS in 14m 35s","accounts_in_message":[],"_revision_number":3},{"id":"7a248fc30696422cf086c6f0d8f0b1d99c80c680","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-12-14 10:25:06.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"5f88fbdbbcb485bd0b6fe6910fad0d405c86ed91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-12-14 10:26:56.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"204d9cfe7ece97b765b4eb32a0718ebcc94acf62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 13:11:16.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d93d7aa904e842d7959c32f86da50047\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32ddf6f2be174ab882b2f35e98b9c232 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be9dcff4e1fa4c86800d70756d24361f : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55939950f91743e8a7be45c7762f572a : SUCCESS in 4m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/18f54c20df224980a1999e63c748634a : SUCCESS in 4m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8982a902b02548ccb9272d9f22e528f3 : SUCCESS in 16m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/91213924fefb48928f31cce067e7c1b6 : SUCCESS in 9m 16s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/6dfd8ff013fc4c7d972027e0e24f7eb2 : SUCCESS in 1h 03m 30s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/01c714190f3045c8804cd7f35b7b204b : SUCCESS in 1h 03m 45s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/afff4d2aa285484ca309e1c689bcbc22 : SUCCESS in 58m 31s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a445f816c1b94939b361b9bbf04a7589 : SUCCESS in 1h 08m 17s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/08d3a8496ced42d69749e67c0714c72a : SUCCESS in 16m 14s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/99410440521b47118ecf521412879dd0 : SUCCESS in 18m 35s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/418f6b8d359440d899cceee6d741a769 : SUCCESS in 12m 35s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c3060cd157704231915bd6b679fbe936 : SUCCESS in 2h 38m 45s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b96d2bea2464449aa78de0fec1b4489 : SUCCESS in 2h 16m 37s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f5a7fa2203b646a2a478296117cc33b5 : SUCCESS in 17m 12s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/91bf8db8d8df4c5c97889112a2aad144 : SUCCESS in 12m 31s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/1a454f20b83d49f1b330d3a9f63e8fe6 : SUCCESS in 34m 09s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e55bde4f5f734a06853a92e3e6ec4e71 : SUCCESS in 30m 23s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e5e2af0878c42e1a03dd4ef2c4ac986 : SUCCESS in 22m 28s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/adcffefea23642fa977831aca593730f : SUCCESS in 13m 10s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/457933f35b294abea421aa22df95f247 : SUCCESS in 10m 12s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/32380d7b614846b78e0ea6fbab998deb : SUCCESS in 17m 54s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3addcdf27d0c438799a2d687859124fa : SUCCESS in 15m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5487091996c404e85af97d72998a521 : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2ed69e4dbcce4e22999e3ac4c464fca9 : SUCCESS in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7f54a056ad7a4e3e8b7e1d990277ed55 : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":5},{"id":"28839c7436300808f1cce2e4a16cb36a1fb3b165","author":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"date":"2023-12-22 16:00:11.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"0336f082bdff61be1585a7577ef2156debfa8239","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-12-22 17:06:06.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"fb434853febed954a93112eb9d802d1c3ec42dc4","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2023-12-29 03:50:03.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"6a3b24cecd96522d12003296abec90982a19118f","author":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"date":"2023-12-29 08:29:15.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"a9a5a34d11cb58873b9467c1245dbadc69220e89","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-01-03 10:36:57.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"dd121fe378430d6deb5c120fee4c1d0b78cc825e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-02 13:13:30.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"de3424f5c9628e45278be0af814d274aafbbd3f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-02-02 13:26:07.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"c2d8f17b7bdb87214f950baa9c44a3dcbc535709","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-02-02 13:26:31.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"cab15375ccf630ac73ee2a67437079c44b74fa12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-02 15:55:08.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/038e3867653646db9edbbe47eb085c94\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f7232fa9a8446f4bc5125717442b10b : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bda99e5ba8ca42ee88a9f6cffc107b70 : SUCCESS in 8m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9633f4721561431a8cd01c95f99158fc : SUCCESS in 3m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f50ba10e5f1745f899b30f1604143372 : SUCCESS in 4m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4caf37c1bf874dbeac040d750af32240 : SUCCESS in 11m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/68b5d3862db941d8af61700ddd4a18c5 : SUCCESS in 11m 32s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d5cad5e0745a473696db03fb7aee0c8c : SUCCESS in 1h 08m 05s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/22d396342a454e2f957d97d6955139de : SUCCESS in 1h 07m 22s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c6a286a42702454cb26ed769bfedbfe4 : SUCCESS in 52m 21s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/40d2a22af0e24c8ba7bd34cb01a12e3e : SUCCESS in 1h 09m 54s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7e0e2159785a47a8b1a2858eda406e34 : SUCCESS in 17m 04s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/d169411bf3384024907de6ab8752d8ec : SUCCESS in 18m 34s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59e177c6f2fa4f2cb1f852e47ed1ac8d : SUCCESS in 15m 51s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/324328dd8a724dca934510d50ead8555 : SUCCESS in 2h 25m 05s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e4b622e5f634b249ba4ffea618593a6 : SUCCESS in 2h 01m 09s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/92a8ea70f95d46e4ada26ad86c4bdfa7 : SUCCESS in 17m 06s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9971821dd42b4fbab5d92ac01a5590f1 : SUCCESS in 16m 20s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/15f2dffd16384dfd9a775c95a1745e7f : FAILURE in 35m 07s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d3ae5b16dcb447f5953d70ed1efabb6c : SUCCESS in 39m 02s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/50c572012ab449aea9d4da807b328c76 : SUCCESS in 30m 56s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0cb7b898042b4421bfbbff3857fe2f85 : SUCCESS in 13m 51s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93256965bf014c0c8ac1fa52a35955a2 : SUCCESS in 13m 20s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3672449c8792466a9a0c9353130a95ba : SUCCESS in 21m 23s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/134a254a90674baf93f1a1f7339db658 : SUCCESS in 15m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fef8dab6d6fe419bb6a7a44baf5171ba : SUCCESS in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ea269da4b7e64f819bc95fce5537e67a : SUCCESS in 5m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0e98abe3967940389536a2b2fb7269a7 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":6},{"id":"09c865201f34e6a5b2b67f836a871c14162ad771","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-06 12:33:37.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"5a5f771dff2bef4b5ad01c250b1264af593a62ab","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-06 17:24:35.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"500aaef2e5f5a52dd4e48885d00aaff38e824fba","author":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"date":"2024-02-07 14:24:17.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"02c09f4061c74c5a2bab0ffe2bb0dc92c1f2125e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-02-08 08:40:17.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"03a481b3e883a7dcd90b99e8954546cb12e1a95e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 08:40:44.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"8b8064c9f80c7981b0a68545bb7243aabad41614","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 11:27:19.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/771c62fcea1f4d9c8079f7ebec2a8697\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/525d6bcc12ff46b8995051c1109d7253 : SUCCESS in 7m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7efa80b85fb9432d978281489473b463 : SUCCESS in 2m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/eee1176d32b64fcabb8e5ce90382e3c9 : SUCCESS in 4m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ccd6f49cc8a44f9993414d1e2ec22c3a : FAILURE in 9m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/57162022c5064fbf93a3f44b5c8cc9d8 : SUCCESS in 12m 55s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2e1e95ca8ecf4476b8296267dcd83869 : SUCCESS in 1h 02m 51s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e77d4d429384733816733e289dda2cf : SUCCESS in 46m 39s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d82ffb2596ff41d886889e57918c47e4 : SUCCESS in 1h 13m 19s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c56c7375cc73444c8c5cca38be4583e9 : SUCCESS in 16m 24s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0723458958d04324b3ea52a009dda738 : SUCCESS in 13m 44s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f107306b655445e6881ca2d1c1853a3b : SUCCESS in 2h 37m 33s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/800596061c9448a1bcd9cc4dc4c5b31f : SUCCESS in 1h 45m 26s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/0774806ddfe04ea28d97144b5fcf854c : SUCCESS in 18m 52s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d89d6b69804447658a17321c7816044b : SUCCESS in 13m 57s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f2b6d5174ddc4e5ab04bde0327583f3f : SUCCESS in 29m 46s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fbb474aae0a5442aa23868ad6da14bde : SUCCESS in 21m 41s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fe5b1547b9cf4a02bd5d2e4cb84392fc : SUCCESS in 15m 33s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a1f3a6810a45433d93a8225dac21d91c : SUCCESS in 12m 36s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4a6af9e0d0174869839b078abbc4c039 : SUCCESS in 19m 10s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a549647d5cc4b1ea29e0f014d161266 : SUCCESS in 13m 36s","accounts_in_message":[],"_revision_number":6},{"id":"36b5218b00b9d95aced06398b7cb8724e2f72749","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-16 14:34:15.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f5739831bb569e9fda35ac94a66de6d3a317a47f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 17:03:14.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6abc999269b348149f38026c5bca88d3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74092d4b40fb432f99f0bbf4c7977100 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edf9155a50c745b89201c497c39e2866 : SUCCESS in 7m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/801ef64e7615475fb85c91480d4eb885 : SUCCESS in 4m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/67f0e8d5b2d8491c8a14e02af057eac6 : SUCCESS in 4m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e862c25d2fb7483c82dc561bae072ebe : FAILURE in 10m 43s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7eb73fc5146e41e3a7c63440afe47421 : SUCCESS in 13m 19s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/3390f4915e7a47d098ff30de198f44cc : SUCCESS in 1h 05m 42s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/f1fd549ae11946839918c62d74d22df8 : SUCCESS in 59m 29s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c625df791c34fd0be47dc97c3feef2f : SUCCESS in 56m 46s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ed3813f2a30d4ba090dbc5cfeef7e708 : SUCCESS in 1h 13m 19s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/16f4de755030489aaecc4e416edc6793 : SUCCESS in 17m 04s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/1fe21270d5f34c508e4aabb3f983f462 : SUCCESS in 16m 36s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4de46ac4ba2f4e98858f66270ac3286e : SUCCESS in 14m 36s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/21e4db4d60af4217a2527b03c579fd65 : SUCCESS in 2h 27m 15s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f7cfa49fb0b4b3287ccab477faebc7b : SUCCESS in 2h 00m 16s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ca450065878846ed92390b5cccf5140c : SUCCESS in 15m 52s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ff5f2fe9f7c4e69bea0d5f3fd9e3969 : SUCCESS in 14m 26s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/e3023f7f57cb4fa0868318f4f62754c1 : FAILURE in 36m 53s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a6ee5f4e09374dfe8e334741201d2308 : SUCCESS in 29m 00s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a50f964517af4b5dbcde41032876ccad : SUCCESS in 27m 07s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2fbfe9f82f4d46cab67bfb4b872e38f5 : SUCCESS in 15m 32s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fd202a7eefca4f649c86d1e744dc1a8e : SUCCESS in 12m 12s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/dae8e434d856490abb443bd34f045e77 : SUCCESS in 22m 12s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f92edf39362d4efaad8fc728f55905bf : SUCCESS in 16m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f97f5908769475f9f5e6e7d1bb203c5 : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d8b8e47fb0c3425292bd2a295eb73f06 : SUCCESS in 4m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8c1a67c37fb3446ea31729ecbdc06c08 : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":6},{"id":"ead1d6c708d713a5ecac33840344cc266df3d550","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 17:03:42.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"60607b9e221baa3074b8a8641fe5ebceb42ed398","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 19:36:21.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9daff2f04be44e1bcc67110e1d497d2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6b07826e9d848d987e0e517a2f6ca42 : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15a69881a4214641a42078e012b6bbf7 : SUCCESS in 2m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d416c07e19d74883a2cd4ce2eb5b02f8 : SUCCESS in 4m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/97fc92b0999e4490b6d8ee9167e0d6c0 : SUCCESS in 10m 22s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f98f31a714a44d1ca6533bed84353851 : SUCCESS in 1h 10m 19s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8181ebd259474c0fab17f41a0a1a362f : SUCCESS in 55m 37s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/03cbe895986a4344a890588bc1fab443 : SUCCESS in 1h 19m 07s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5fb2f8d816614831b28eb7a560126f2d : SUCCESS in 18m 31s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/92bb8411d6a74f53947531ba52ec04b2 : SUCCESS in 14m 05s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/328ec308c44c4e02a0662ca6d7f8b716 : SUCCESS in 2h 32m 29s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8478fce697a54d26acc325bd98db832b : SUCCESS in 2h 07m 59s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e570ea2195e945558e4118546eb48c61 : SUCCESS in 18m 32s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f4d8d0c19a004a2889c61ac48dbc38a5 : SUCCESS in 14m 50s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c0f19f6b821041cca1258669a8902b07 : SUCCESS in 30m 14s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1135923d6e53487bb7905612a7ec6cc6 : SUCCESS in 23m 33s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/42f9c1a34d9743d2bdbd91af2a1ad774 : SUCCESS in 13m 33s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/50b39cc7f8564c3bafe45a281ed42a3a : SUCCESS in 13m 24s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ee01f4ba6f2248d894a5a2d8340bd925 : SUCCESS in 17m 19s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/81c18ef4b77643da8eed23601a508104 : SUCCESS in 14m 44s","accounts_in_message":[],"_revision_number":6},{"id":"e3ddc97f9344d84c15c856ef579dd25117ea3661","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 19:36:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"b0dbbc5359c79ae9f6a5a82606bf5250e2900f39","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 19:37:23.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8386ea23a6f46839b6fab7ff1bb0b57\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f0752de247b419cb0656bb904f099fc : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3da06936c174efabbfa1c15e5c6abd9 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":6},{"id":"9552517e069281c0152e9aef0bcbc5a8d067a34f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-16 20:07:27.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"bc98c494aecf84ddae070502b00de0b9db97c5dd","revisions":{"959a44bca5e4509ba10d902dc382dd3d7d95a529":{"kind":"REWORK","_number":1,"created":"2023-09-25 13:04:27.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/94/896394/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/94/896394/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/896394/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/896394/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/896394/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/94/896394/1"}}},"commit":{"parents":[{"commit":"1847ad3f17a039f29e22f465508991fcb6e1cca7","subject":"Fix data file path detection with new pip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1847ad3f17a039f29e22f465508991fcb6e1cca7"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-09-25 12:55:07.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-09-25 12:55:07.000000000","tz":60},"subject":"Generate KA config before ``kolla ansible run``","message":"Generate KA config before ``kolla ansible run``\n\nSupports calling custom Kolla Ansible commands directly after a\n``kayobe control host bootstrap``.\n\nChange-Id: I19f188cc002f8578618003e90c0a4a154b806e49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/959a44bca5e4509ba10d902dc382dd3d7d95a529"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/959a44bca5e4509ba10d902dc382dd3d7d95a529"}]},"branch":"refs/heads/master"},"d30ec5254abe13f1ed8dbb9062d82e1991fdf32f":{"kind":"REWORK","_number":2,"created":"2023-10-31 16:09:27.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/94/896394/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/94/896394/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/896394/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/896394/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/896394/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/94/896394/2"}}},"commit":{"parents":[{"commit":"1847ad3f17a039f29e22f465508991fcb6e1cca7","subject":"Fix data file path detection with new pip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1847ad3f17a039f29e22f465508991fcb6e1cca7"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-09-25 12:55:07.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-10-31 16:09:10.000000000","tz":0},"subject":"Generate KA config before ``kolla ansible run``","message":"Generate KA config before ``kolla ansible run``\n\nSupports calling custom Kolla Ansible commands directly after a\n``kayobe control host bootstrap``.\n\nChange-Id: I19f188cc002f8578618003e90c0a4a154b806e49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d30ec5254abe13f1ed8dbb9062d82e1991fdf32f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d30ec5254abe13f1ed8dbb9062d82e1991fdf32f"}]},"branch":"refs/heads/master"},"510a4fdbe219f1ef0772e867171d91d65b75f193":{"kind":"REWORK","_number":3,"created":"2023-11-02 13:35:23.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/94/896394/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/94/896394/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/896394/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/896394/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/896394/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/94/896394/3"}}},"commit":{"parents":[{"commit":"1847ad3f17a039f29e22f465508991fcb6e1cca7","subject":"Fix data file path detection with new pip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1847ad3f17a039f29e22f465508991fcb6e1cca7"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-09-25 12:55:07.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-11-02 13:35:11.000000000","tz":0},"subject":"Generate KA config before ``kolla ansible run``","message":"Generate KA config before ``kolla ansible run``\n\nSupports calling custom Kolla Ansible commands directly after a\n``kayobe control host bootstrap``.\n\nChange-Id: I19f188cc002f8578618003e90c0a4a154b806e49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/510a4fdbe219f1ef0772e867171d91d65b75f193"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/510a4fdbe219f1ef0772e867171d91d65b75f193"}]},"branch":"refs/heads/master"},"71674e25333c4aebad89f8441552844dd1f0e03e":{"kind":"REWORK","_number":4,"created":"2023-12-14 10:25:06.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/94/896394/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/94/896394/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/896394/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/896394/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/896394/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/94/896394/4"}}},"commit":{"parents":[{"commit":"bb2af07567152832700a2c2971cd72ff4e3b77d3","subject":"Merge \"Revert \"Use OpenStack 2023.2 release\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bb2af07567152832700a2c2971cd72ff4e3b77d3"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-09-25 12:55:07.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-12-14 10:24:50.000000000","tz":0},"subject":"Generate KA config before ``kolla ansible run``","message":"Generate KA config before ``kolla ansible run``\n\nSupports calling custom Kolla Ansible commands directly after a\n``kayobe control host bootstrap``.\n\nChange-Id: I19f188cc002f8578618003e90c0a4a154b806e49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/71674e25333c4aebad89f8441552844dd1f0e03e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/71674e25333c4aebad89f8441552844dd1f0e03e"}]},"branch":"refs/heads/master"},"41de338993773ae78a46f49f5ebfedb5446434fd":{"kind":"REWORK","_number":5,"created":"2023-12-14 10:26:56.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/94/896394/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/94/896394/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/896394/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/896394/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/896394/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/94/896394/5"}}},"commit":{"parents":[{"commit":"bb2af07567152832700a2c2971cd72ff4e3b77d3","subject":"Merge \"Revert \"Use OpenStack 2023.2 release\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bb2af07567152832700a2c2971cd72ff4e3b77d3"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-09-25 12:55:07.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-12-14 10:26:47.000000000","tz":0},"subject":"Generate KA config before ``kolla ansible run``","message":"Generate KA config before ``kolla ansible run``\n\nSupports calling custom Kolla Ansible commands directly after a\n``kayobe control host bootstrap``.\n\nChange-Id: I19f188cc002f8578618003e90c0a4a154b806e49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/41de338993773ae78a46f49f5ebfedb5446434fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/41de338993773ae78a46f49f5ebfedb5446434fd"}]},"branch":"refs/heads/master"},"bc98c494aecf84ddae070502b00de0b9db97c5dd":{"kind":"REWORK","_number":6,"created":"2024-02-02 13:26:07.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/94/896394/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/94/896394/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/896394/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/896394/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/896394/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/94/896394/6"}}},"commit":{"parents":[{"commit":"691454965fc12108db65479f4dd29ef25ac75fa8","subject":"Merge \"Use dummy1 as bridge port instead of eth1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/691454965fc12108db65479f4dd29ef25ac75fa8"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-09-25 12:55:07.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-02-02 13:25:58.000000000","tz":0},"subject":"Generate KA config before ``kolla ansible run``","message":"Generate KA config before ``kolla ansible run``\n\nSupports calling custom Kolla Ansible commands directly after a\n``kayobe control host bootstrap``.\n\nChange-Id: I19f188cc002f8578618003e90c0a4a154b806e49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bc98c494aecf84ddae070502b00de0b9db97c5dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bc98c494aecf84ddae070502b00de0b9db97c5dd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
