)]}'
{"id":"openstack%2Fkayobe~896414","triplet_id":"openstack%2Fkayobe~master~I9d7d71d42026f3c12da6de9c5ca55dc647554fd0","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-10-17 13:28:52.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-10-17 13:28:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9d7d71d42026f3c12da6de9c5ca55dc647554fd0","subject":"Fix last return value in _get_base_path","status":"MERGED","created":"2023-09-25 16:42:36.000000000","updated":"2023-10-17 13:29:58.000000000","submitted":"2023-10-17 13:28:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896414","meta_rev_id":"f8997e994407cf08bbd032dfde76d5bf6d8c898b","_number":896414,"virtual_id_number":896414,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-17 13:28:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2023-09-26 08:10:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-26 08:12:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2023-10-17 11:17:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2023-10-17 11:17:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"reviewers":{"REVIEWER":[{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-25 19:09:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-26 08:10:10.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-09-26 08:12:57.000000000","updated_by":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"reviewer":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"state":"REVIEWER"},{"updated":"2023-10-05 17:29:07.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-10-17 11:17:33.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"0d47a6d9f01126e63d7d98498d6e926ac0068d69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-09-25 16:42:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4f52de11ccdc95c3bdf4750d0fcb7226e7f3557d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-09-25 16:43:21.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5e8711ff6b30d1a688c27f0a51aa297d507adabb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 19:09:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d947fa7b041f4aab8b25dbe83e312551\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33d61cad4f6c42cb8f79bd2e064f30d7 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adb79de1f6b44aa5a560ca3544f77cc6 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/902760d75085493aaf661d02720c8247 : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ca8608e0cf3485a97ca896e380511bb : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a8bd71e8895b4bf088382666c5a05471 : SUCCESS in 3m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5ab9a9ce307e4b4c9620bf9838930207 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe7497d0b3e24a0abdacb568b854d92b : SUCCESS in 7m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e9caabb114364c1a9fd9f9a5f971b0f2 : SUCCESS in 4m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/49bab273ff9c4a7e90321ac6473a9941 : SUCCESS in 9m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/85e3a22bc0494a05b6ce4633152b1492 : SUCCESS in 10m 45s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/dd45bb8182544e569bda5fd9e104b6ac : SUCCESS in 55m 13s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/2e0631c65cff4becb3d9a5a0c0b4f0af : SUCCESS in 53m 25s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dba4e3077e5042c59ee14d3b2bbda683 : SUCCESS in 41m 46s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3458fa8b7b744b2eafbca00ff5cfbb7e : SUCCESS in 1h 03m 00s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4f66924999654878a8f580e66e01bb8c : SUCCESS in 18m 37s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/bc7e8d309f9a4c759b09ad94bfa05fef : SUCCESS in 15m 57s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fb36fad245e04def89f6924abcadb5f2 : SUCCESS in 14m 44s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/953b15f4a8694bb0a405ecf6be484a58 : SUCCESS in 2h 21m 08s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/86600b0eaed6416e89efc83acd68f37a : SUCCESS in 2h 06m 20s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a9d1b39fca3d49dbabbd4dc344191794 : SUCCESS in 19m 21s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/af5b837d08c94986b914aac3c419f003 : SUCCESS in 16m 39s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/c062aae390f54625bae433ba099cd252 : SUCCESS in 28m 30s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/17a5b6e0bcb64bd0ae375533afa83bb7 : SUCCESS in 29m 38s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c11662a92035458cb861f448908de798 : SUCCESS in 23m 51s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c486f62bf41742b2934f72a11096e8e6 : SUCCESS in 14m 00s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cbbc26269ca34706b499198d0b0f6816 : SUCCESS in 9m 43s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/60682401168041699410e7df5ed68b2f : SUCCESS in 25m 47s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ccec9cc682b4b3baed78de92c95bb84 : SUCCESS in 23m 16s","accounts_in_message":[],"_revision_number":2},{"id":"59f9a6ce424f5f8393d09810f374753fb0be9efe","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-26 08:10:10.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1bc441f604fec8daf7aefcab183e18e362f97bed","author":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"date":"2023-09-26 08:12:57.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2803892d60983d9694b04eab3354e8cba7e13592","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-10-17 11:17:33.000000000","message":"Patch Set 2: Workflow+1 Review-Priority+2","accounts_in_message":[],"_revision_number":2},{"id":"7fcd171e9cb769aeec7912a30ec65b4534a20059","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 11:17:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fd2d3b9a613d3254e2eccb921af06b1fd517cc02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 13:28:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3af469e0c5614c2da70c5e9e114256a9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0b2d38b5d2241ad822aa291888cd9a4 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a804dc4c8dec462c932c08e3236cccff : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e51bbc00a5d34f59bbc3022d52f7529b : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/59bd482d1946489388cd39d3662bd2b6 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3200782f28ba4786950ee3d53f72d0c3 : SUCCESS in 8m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/660f164ac7af46dc83a755289b1c1c8e : SUCCESS in 4m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7fc880f427ac4d22b8e6a40156930b5f : SUCCESS in 10m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e32137a5e17f4f2cabba7886a26933e9 : SUCCESS in 9m 23s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0ff2ebd00e37405fbcf624a4cf5b4f1d : SUCCESS in 55m 12s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e728340cf0b84ec5a7fa32e6f93d9bf1 : SUCCESS in 44m 58s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c0a4926da8a94d38a9d31c12092c81d1 : SUCCESS in 58m 31s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e7151e22652e47199c342491a78b8880 : SUCCESS in 13m 08s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db33bc75e3e74405baa9a40392146d23 : SUCCESS in 10m 49s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/37fa32a791f845db96826e2c1b4995d8 : SUCCESS in 2h 09m 43s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/64c708fa3dec4dbca3ee3c79512ddf56 : SUCCESS in 2h 02m 53s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d40670be3bbb4094aa23e2b564c2c256 : SUCCESS in 17m 12s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/25c49625aff54dd8bf0f899b497b2b53 : SUCCESS in 15m 45s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9d22acd1d2fc4437a373834609a344bd : SUCCESS in 25m 32s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8fdb694dc7da444c9ffaac8fc9b2c221 : SUCCESS in 22m 34s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e36d5d6147f6429fa715a341b7aed84c : SUCCESS in 13m 40s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b14165dba13c474d896d2bc7b2feff99 : SUCCESS in 11m 19s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c894141245cc442f8e547d8746c035c1 : SUCCESS in 20m 23s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84e4f358f508452b8bd2ba161681ac92 : SUCCESS in 15m 36s","accounts_in_message":[],"_revision_number":2},{"id":"687e4ce3a1263e735b36bd5307d2c00bcfdfa955","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 13:28:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f8997e994407cf08bbd032dfde76d5bf6d8c898b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 13:29:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b70de7f2df346218257e126b980496a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/938be7b859994d23b29410e54ef35cae : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"810e021d8bdca17b1e217569f9bf24f917f6810d","revisions":{"64e781285e81f482d5beb55f59e91f4563c339e1":{"kind":"REWORK","_number":1,"created":"2023-09-25 16:42:36.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/14/896414/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/14/896414/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/14/896414/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/14/896414/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/14/896414/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/14/896414/1"}}},"commit":{"parents":[{"commit":"1847ad3f17a039f29e22f465508991fcb6e1cca7","subject":"Fix data file path detection with new pip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1847ad3f17a039f29e22f465508991fcb6e1cca7"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-09-25 16:40:57.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-09-25 16:40:57.000000000","tz":120},"subject":"Fix last return value in _get_base_path","message":"Fix last return value in _get_base_path\n\nThe return value was set to the following path which is not valid:\n\n    kayobe/kayobe/utils.py/..\n\nThe intent must have been to use dirname.\n\nChange-Id: I9d7d71d42026f3c12da6de9c5ca55dc647554fd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/64e781285e81f482d5beb55f59e91f4563c339e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/64e781285e81f482d5beb55f59e91f4563c339e1"}]},"branch":"refs/heads/master"},"810e021d8bdca17b1e217569f9bf24f917f6810d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-09-25 16:43:21.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/14/896414/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/14/896414/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/14/896414/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/14/896414/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/14/896414/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/14/896414/2"}}},"commit":{"parents":[{"commit":"1847ad3f17a039f29e22f465508991fcb6e1cca7","subject":"Fix data file path detection with new pip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1847ad3f17a039f29e22f465508991fcb6e1cca7"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-09-25 16:43:07.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-09-25 16:43:07.000000000","tz":120},"subject":"Fix last return value in _get_base_path","message":"Fix last return value in _get_base_path\n\nThe return value was set to the following path which is not valid:\n\n    kayobe/kayobe/utils.py/..\n\nThe intent must have been to use dirname to produce:\n\n    kayobe/kayobe/..\n\nChange-Id: I9d7d71d42026f3c12da6de9c5ca55dc647554fd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/810e021d8bdca17b1e217569f9bf24f917f6810d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/810e021d8bdca17b1e217569f9bf24f917f6810d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
