)]}'
{"id":"openstack%2Fkayobe~896877","triplet_id":"openstack%2Fkayobe~master~Ic53efd03cecbd53ad3e3b64b664e084f4e25be0e","project":"openstack/kayobe","branch":"master","topic":"bug/2020135","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-10-03 11:04:18.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"Ic53efd03cecbd53ad3e3b64b664e084f4e25be0e","subject":"Work around mocking issues on stable branches","status":"MERGED","created":"2023-09-29 04:48:03.000000000","updated":"2023-10-03 17:55:07.000000000","submitted":"2023-10-03 16:58:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896877-bug/2020135","meta_rev_id":"0bb30a3ad82bfaa272dfdfeffea7d87ae693b499","_number":896877,"virtual_id_number":896877,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-03 16:58:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2023-10-03 08:15:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2023-10-03 07:51:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-10-03 17:55:07.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2023-10-03 08:15:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-29 07:49:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-03 07:51:52.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-10-03 08:15:58.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2023-10-03 17:55:07.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"6125b45dfca5982cd66c6ed0394544a3c3834042","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-09-29 04:48:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7939bd9c4a58bf05695e053900e52d6828e964bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-09-29 05:12:05.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"96603be346a3a2a9e71673cd220bb8c46d257156","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 07:49:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44ab94c41aba433fad0c1dc625ace88a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7010b4e3edfa4fd2ae1fade72f97240b : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01e3681b01f244bab1b2c50744f061cf : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48b84078fa2b4563a0c93f0bf71066c0 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c2969b4ee9a4a41a49ef1c8d10b968d : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b309f35f884e4e078ce229452e92c31c : SUCCESS in 4m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/50ceb8f2562a409fb5541bfb135150e2 : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfd75443c041407d9ad878f77a4689af : SUCCESS in 6m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/971d4e5ffb364f48a1f9b19a31b08f17 : SUCCESS in 5m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/41ade73213d74ef3b6ff2d81b6076a90 : SUCCESS in 11m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2f8e041f341144a494906c6f2f7e3d94 : SUCCESS in 11m 36s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a2b1417936af4e709a7c958163141e4f : SUCCESS in 46m 32s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/ab890bf444a64612b3bfffe69295ed03 : SUCCESS in 51m 51s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76fef410af954b7f9e3d6481c5e3799b : SUCCESS in 48m 44s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/63d2b699e4944839a249f56f1cf90b91 : SUCCESS in 53m 38s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/075f35805ccd4283ad5175b8da5b750e : SUCCESS in 17m 04s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/0476b4935003484a90d0343c7d05d1ab : SUCCESS in 17m 49s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ef88e051cb143969b70ebe0c79c4f8f : SUCCESS in 12m 04s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/651b82419d344757a2ba1fcfb3d75666 : SUCCESS in 2h 30m 33s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2af36be8353d4c8bba16a12ea0660e32 : SUCCESS in 2h 16m 40s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/39b8753f3d2a4f28a131dbff3539dfa1 : SUCCESS in 16m 42s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/355d03fbf5fe4120929ccb6f4745df9b : SUCCESS in 13m 11s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/e94a88b7436a4485be65515502727f17 : SUCCESS in 25m 42s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/295b3eb47a874d47b36558c04787c2de : SUCCESS in 23m 36s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c51ee794a044dae90aa97313642a101 : SUCCESS in 21m 26s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/524c320b691f4ee29da9023521195779 : SUCCESS in 13m 30s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc067bd36523410e99684feea2f4ebf6 : SUCCESS in 10m 50s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/59b8dc613e194dd18659f5b2df20d4aa : SUCCESS in 32m 02s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/46ecaf4be5964ee386c036e1ab7c4584 : SUCCESS in 19m 54s","accounts_in_message":[],"_revision_number":2},{"id":"80e8bd37e5cc38c225e233119ee607f6420ad637","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-03 07:51:52.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e41323baf6b8c98498a2b487fb5f93298df231b4","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-10-03 08:15:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"021794510ae8862d88796b684e1029d7dc4a57f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 08:16:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ab500592ca914b356277690d3b0d4872fea1a481","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 10:50:58.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/da352ba93f8845dbb2f18abba85913ad\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ea54d25ab424d5494e75d60a2675be3 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4512445a08774da695c52dbd24a2d47e : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf4b6f2741004615a54f39a2e8c80f72 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a1e665b1e3624cafa22e72590f6f9398 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d5c67620ca3493d95e0927b1775af40 : SUCCESS in 7m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ba43247a73f247b1b076c58663e816fb : SUCCESS in 4m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/92848802cb5e4bed906687fbc475256b : SUCCESS in 11m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4a9478fec6ae493bb28eea30085adde3 : SUCCESS in 10m 29s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0b16e00158254ec786534994ce0a29b7 : FAILURE in 1h 02m 11s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c2eefc266b043faaddd25fddd81aefc : SUCCESS in 57m 59s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/4b45c6f2cc3d456cb0c1902c4327d011 : SUCCESS in 1h 00m 47s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/38afee8345a340f2883759f29fb51d7e : SUCCESS in 17m 15s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e6c3d9d03d3f4a02bdc82a3aded95b37 : SUCCESS in 14m 05s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d58a51d1361744b5813738e7e0e4dad0 : SUCCESS in 2h 11m 38s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/46452092c6f34cbf92705461b1cbbd69 : TIMED_OUT in 2h 33m 15s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/8256b970b7ff492c8a9a262c13b3c087 : SUCCESS in 17m 15s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/915bd8bfabee42b58635207d0291b50e : SUCCESS in 14m 23s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8fe9f2d5284e46ceb59f0f070cf8ed4e : SUCCESS in 31m 21s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3584cff668ef476fb534977c3c609a48 : SUCCESS in 25m 18s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1234f39b8d2e461aa6604e24c15ad602 : SUCCESS in 13m 10s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5369ea79ffed4ea1bc430a0a778a5232 : SUCCESS in 10m 00s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/45d24364f587405b985a4f53ab28fef2 : SUCCESS in 30m 17s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fabcf3c871924757b0540ce381fce932 : SUCCESS in 22m 04s","accounts_in_message":[],"_revision_number":2},{"id":"da05814f5ea4a420f5143f67229f96ae257d37d4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-03 11:04:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c7c033da3070014666e6ef98560edc77bc67bfab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 13:54:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03472d548e044c51b35260616702c6ae\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d1a255f5589479e8c413b68549c7dd6 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36061a82795c4ff085f4ef9aa6322388 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aaec42bfab9242a5b3c27d18dd09b9a2 : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c6eb3b43b2d4f21b09515e9300b5e72 : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d9198da06ab54982ba6e83d61c958c54 : SUCCESS in 4m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6afb586448c847c49079a8bbbffaae35 : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0d4049ba0f34fb78fc95782b84020b5 : SUCCESS in 7m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/15a8701e5ef84945a0bb0aae7650e33e : SUCCESS in 4m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2d7ea6408c134446a1e1ef7638a1ca57 : SUCCESS in 11m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/caf2a10ea60d4fc1a75d190262d0e32f : SUCCESS in 10m 22s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/096cfeb17394472f871f5df33e8c9cc7 : SUCCESS in 52m 56s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/37ca30d967714d5ca2d76722caca8dd9 : SUCCESS in 47m 23s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/70f2204549524ad8afe6efab4a20f5e3 : SUCCESS in 55m 54s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/bb247edbcd7f49ada7326849114622d9 : SUCCESS in 1h 03m 03s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/bfe559f67cbd418b8ee21cf2249486cb : SUCCESS in 19m 14s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/d151c932293e4993b46e36fd50665404 : SUCCESS in 18m 17s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09fda99cef73488cabb15d5350a00ce5 : SUCCESS in 13m 45s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8b42e5c539f444a4856895f582e14919 : SUCCESS in 2h 19m 01s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d70116bd57984f33a12c24618a065b88 : SUCCESS in 2h 14m 12s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/57e2c76d92b8466ba5c017e06c41ad85 : SUCCESS in 17m 53s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd5e6aaa11e842089e5401d583366173 : SUCCESS in 16m 19s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/1981895067cb408f82536eb1b6576c79 : SUCCESS in 56m 02s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/06facc0b4aed4d9eb717286e36600389 : SUCCESS in 32m 17s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/069702dc1c1b400192b8272cdb15adad : SUCCESS in 24m 07s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5ca7e7d390c345f8b541789b95406839 : SUCCESS in 13m 44s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52bdfd72d0a946b9a3ce5eb2851886bc : SUCCESS in 11m 32s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/520c23cae148468eb6702ce2a6a65f39 : SUCCESS in 33m 33s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d28da2dc9d874b198332c80a2b8aeb38 : SUCCESS in 31m 32s","accounts_in_message":[],"_revision_number":2},{"id":"6fb04ede76a2bcbede9270a1300d61c6d9e49f93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 13:54:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"060735a3969c0db83fa7aa67a3899a42b94b3998","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 16:58:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d15e863f9841402a9aed077296a7ff63\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8f4d183617b4798be236d4eacadd30b : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0653e6fc95ea4fd79cb7e789936f2bfb : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b955052e26f4494bee08ded31534936 : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/102fa56e58b04707816c7041f44e2e60 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/242e39f4d81f4ce090aadb719170706e : SUCCESS in 8m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fa67b23309bd4ed7883f43723bd4f2b3 : SUCCESS in 4m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3d57f1ff2a01407d98b62d66a6f8a5ba : SUCCESS in 10m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f5aedcd2e6b64bb2bf625811ca57744c : SUCCESS in 9m 58s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2b8a1cb4d2aa4c31b82b9e26d48ebd61 : SUCCESS in 57m 13s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/018cfc62b8224a68bddf4d32a98e0175 : SUCCESS in 55m 27s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/110d52bbce8645b8b818a09d04c719ff : SUCCESS in 1h 08m 59s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9104713e42bf4019a0ad74d36026f00c : SUCCESS in 15m 46s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/95dfe8e6d3b34d7097b9fb4500c45f77 : SUCCESS in 15m 18s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/38d3a22a982c4953aa08877f775ada2f : SUCCESS in 2h 16m 29s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6d58e56fdaa43f58d77113c4c9ff31d : SUCCESS in 2h 20m 59s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7c6d3965bd3d4d7784000b4f52d04ca8 : SUCCESS in 20m 14s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76e129b335c94e3b9ef236e9bfcca587 : SUCCESS in 16m 37s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a0c80c0b67d54bb293f75ce62c77a035 : SUCCESS in 29m 57s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/98a0e7c1401e4f83b443f264a654cfbc : SUCCESS in 24m 20s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a2c63df6f78844a583a9d57c16a88e9d : SUCCESS in 15m 27s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/31f6d33f723941f28fa04012602d7488 : SUCCESS in 13m 54s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8cb56e80e7c74e0085b1423f5c920102 : SUCCESS in 41m 19s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0cd995e9624441589db5f59c2ba792a0 : SUCCESS in 22m 16s","accounts_in_message":[],"_revision_number":2},{"id":"2973a3c655c1f3cb6cb19e2e913c017b59ee9129","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 16:58:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"51c4a896bbc6364d86572d8d41c5f189bddd3081","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 17:00:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8a64c87744a48629d3e6831b7f910cf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2a76ac1bd11404da5f3d6a0b0c9d908 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2},{"id":"0bb30a3ad82bfaa272dfdfeffea7d87ae693b499","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-10-03 17:55:07.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cfaf571794296d6e0cfed2771f8046a8280ffb22","revisions":{"af5fc6a0872bbfbfcbf9d85436ff6c4bf46483b9":{"kind":"REWORK","_number":1,"created":"2023-09-29 04:48:03.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/77/896877/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/896877/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/896877/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/896877/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/896877/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/896877/1"}}},"commit":{"parents":[{"commit":"1847ad3f17a039f29e22f465508991fcb6e1cca7","subject":"Fix data file path detection with new pip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1847ad3f17a039f29e22f465508991fcb6e1cca7"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-09-29 04:47:23.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-09-29 04:47:23.000000000","tz":120},"subject":"Work around mocking issues on stable branches","message":"Work around mocking issues on stable branches\n\nBackports of I9dd5b97dec93c0e5393a1e7d9640f85003651b56 to stable\nbranches are failing CI checks because several unit tests mock\nos.path.exists and would need to be updated. This does not affect master\nbecause test code changed in I4d9f96ec4cf3c6cd0d28dfe5ddb239d863498a72.\n\nWork around this issue by replacing os.path.exists by os.path.isfile. It\nis a better check anyway since we are reading the file.\n\nChange-Id: Ic53efd03cecbd53ad3e3b64b664e084f4e25be0e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/af5fc6a0872bbfbfcbf9d85436ff6c4bf46483b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/af5fc6a0872bbfbfcbf9d85436ff6c4bf46483b9"}]},"branch":"refs/heads/master"},"cfaf571794296d6e0cfed2771f8046a8280ffb22":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-09-29 05:12:05.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/77/896877/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/896877/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/896877/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/896877/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/896877/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/896877/2"}}},"commit":{"parents":[{"commit":"1847ad3f17a039f29e22f465508991fcb6e1cca7","subject":"Fix data file path detection with new pip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1847ad3f17a039f29e22f465508991fcb6e1cca7"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-09-29 05:11:43.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-09-29 05:11:43.000000000","tz":120},"subject":"Work around mocking issues on stable branches","message":"Work around mocking issues on stable branches\n\nBackports of I9dd5b97dec93c0e5393a1e7d9640f85003651b56 to stable\nbranches are failing CI checks because several unit tests mock\nos.path.exists and would need to be updated. This does not affect master\nbecause test code changed in I4d9f96ec4cf3c6cd0d28dfe5ddb239d863498a72.\n\nWork around this issue by replacing os.path.exists by os.path.isfile. It\nis a better check anyway since we are reading the file.\n\nChange-Id: Ic53efd03cecbd53ad3e3b64b664e084f4e25be0e\nRelated-Bug: #2020135\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cfaf571794296d6e0cfed2771f8046a8280ffb22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cfaf571794296d6e0cfed2771f8046a8280ffb22"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
