)]}'
{"id":"openstack%2Fkayobe~896878","triplet_id":"openstack%2Fkayobe~stable%2F2023.1~Ic53efd03cecbd53ad3e3b64b664e084f4e25be0e","project":"openstack/kayobe","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-10-04 07:57:19.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ic53efd03cecbd53ad3e3b64b664e084f4e25be0e","subject":"Work around mocking issues on stable branches","status":"ABANDONED","created":"2023-09-29 04:48:57.000000000","updated":"2023-10-04 07:57:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"58e1e48b92b9a105817e46ef2d99e80d04d406f9","_number":896878,"virtual_id_number":896878,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-09-29 06:36:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2023-10-03 17:55:13.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-29 06:36:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-03 17:55:13.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"d247c722e623e95103a15da91549fa6566258f69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-09-29 04:48:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd91e58837f2ea4af489f2f2a92e655739b89e7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 06:36:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ba0f999627e49ae8779b36b396add82\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c493bdcae6e642df9dbb05022f2c890e : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c47201ac7bb4476b8e490dd2b68d8b88 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9347c5561644969aa6e868587895892 : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e62112b0df774df1b181e41cc6beef50 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b109541fe244c408be0ad13098370c4 : SUCCESS in 7m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/43549b6091b04fd297455f44ae903137 : SUCCESS in 3m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/51a65ee49e744d748ab6be5a884467f4 : SUCCESS in 12m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a4acacbfbeba446ea45753a60bd7ec47 : SUCCESS in 14m 21s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/47db41cd8a6f4ca38bac6be9c1b58673 : SUCCESS in 1h 03m 01s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/6b9e1747472b44b39cdf760358acb0cc : SUCCESS in 58m 42s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2cfff319b5524356baa21bda2723a8d4 : SUCCESS in 54m 26s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/2237764bc82e463e96c9388bb6b7cb9a : SUCCESS in 58m 39s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/2c20cb38a5f14d4c985d236a444ac3ed : SUCCESS in 14m 59s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/ca3303b7aa114e68897ecf9949e77dfc : SUCCESS in 14m 48s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6c2d3d1658b41f88f3435ea84c3d240 : SUCCESS in 12m 51s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0a14bf7108e74d569388203624de84a4 : SUCCESS in 1h 43m 01s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/95f576eef4b844c8b0d0a6562f2bcb96 : SUCCESS in 1h 39m 33s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a4bf234fbe3e4142a5c90beef0500496 : SUCCESS in 14m 58s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a68f98597660446496d3a8fcfd71e753 : SUCCESS in 16m 53s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f19eaf1c540140e292a07737904277ef : SUCCESS in 32m 11s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/00616e66cdcc4dc0bbd7a98b5f805235 : SUCCESS in 30m 07s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f2260b165c74caabfefbb7c7a340192 : SUCCESS in 26m 14s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f4e62c7f77ca40f9b0dab591d5a06366 : SUCCESS in 14m 44s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a2f6ebc367044f4bb8a2858581f6115c : SUCCESS in 11m 09s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d683d7e363e241f3bc06552c3d7e3f29 : SUCCESS in 35m 30s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9fa49e9a74fb4a6ba97a0eb8e89f3e11 : SUCCESS in 22m 21s","accounts_in_message":[],"_revision_number":1},{"id":"af7d746fa80e566754d89cf89dcb694ec1329950","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-10-03 17:55:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"58e1e48b92b9a105817e46ef2d99e80d04d406f9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-04 07:57:19.000000000","message":"Abandoned\n\nSquashed into https://review.opendev.org/c/openstack/kayobe/+/896379","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4f652bed120ccd452849db85865fb24805db285b","revisions":{"4f652bed120ccd452849db85865fb24805db285b":{"kind":"REWORK","_number":1,"created":"2023-09-29 04:48:57.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/78/896878/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/78/896878/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/896878/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/896878/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/896878/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/78/896878/1"}}},"commit":{"parents":[{"commit":"1eae7b5a62e5dc08a73fd3131638691504a5570d","subject":"Fix data file path detection with new pip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1eae7b5a62e5dc08a73fd3131638691504a5570d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-09-29 04:47:23.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-09-29 04:48:50.000000000","tz":120},"subject":"Work around mocking issues on stable branches","message":"Work around mocking issues on stable branches\n\nBackports of I9dd5b97dec93c0e5393a1e7d9640f85003651b56 to stable\nbranches are failing CI checks because several unit tests mock\nos.path.exists and would need to be updated. This does not affect master\nbecause test code changed in I4d9f96ec4cf3c6cd0d28dfe5ddb239d863498a72.\n\nWork around this issue by replacing os.path.exists by os.path.isfile. It\nis a better check anyway since we are reading the file.\n\nChange-Id: Ic53efd03cecbd53ad3e3b64b664e084f4e25be0e\n(cherry picked from commit af5fc6a0872bbfbfcbf9d85436ff6c4bf46483b9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4f652bed120ccd452849db85865fb24805db285b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4f652bed120ccd452849db85865fb24805db285b"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
