)]}'
{"id":"openstack%2Fkayobe~897306","triplet_id":"openstack%2Fkayobe~master~Iecf1d469caaa777c5b253eb0e44dc00692197ef7","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-10-06 19:23:54.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"Iecf1d469caaa777c5b253eb0e44dc00692197ef7","subject":"Use importlib.metadata instead of importlib_metadata","status":"MERGED","created":"2023-10-04 09:13:43.000000000","updated":"2023-10-07 00:00:37.000000000","submitted":"2023-10-06 23:59:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897306","meta_rev_id":"a43cb99246fc0e91a305c94ffcfc3190d37356d0","_number":897306,"virtual_id_number":897306,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-06 23:59:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2023-10-05 14:19:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2023-10-05 09:34:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2023-10-05 14:19:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-04 11:41:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-05 09:34:28.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-10-05 14:19:01.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"949d8f84c31b901cbf682b568e7a56782d3bca03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-04 09:13:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b442eac5012b1a8834a4be8b7dd8402b1607e29f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 11:41:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27bdc3be65cd404294b78eea6fc90ff5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b9ea6b983bb434a95b2ec2c73ef57ef : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e3d3fd7e9184dd385a1782fb254fe5b : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f14be5d58ff64a4097aa6c45b706c3e8 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92912cd317a248e38fcff5f6ea06a5c8 : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/36c87b243b364fe288cedcac5895d73e : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f26d7c8a595941fca1a917a150a93fe2 : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac98a72a0faf4eb6a373a9e922e5e23e : SUCCESS in 10m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2e16d436dc754827abbe977d892ede10 : SUCCESS in 4m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/72b863e9ff184e4f8a23ad6e09ac29db : SUCCESS in 13m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1a616cc3e4464616b5db80e0683f1d67 : SUCCESS in 12m 07s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4d48729837a846efad99573cedbcd5ac : SUCCESS in 57m 01s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/a011777a8b2e48c2acf478602b6905e0 : SUCCESS in 1h 00m 13s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7fdd5964e2048089018eb84669821e5 : SUCCESS in 54m 38s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/57bb9f3c754848aabd1efb9a8a94bfa9 : SUCCESS in 1h 03m 56s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a07ebbec656348ed94e2dce0e14bc4a5 : SUCCESS in 16m 28s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/7a127d17bb3543838db5ad5bcc0ca609 : SUCCESS in 18m 23s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/68bd44345c9645db811996841ac210a2 : SUCCESS in 14m 29s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4a5e7de8af56414b8f859a7b122180f4 : SUCCESS in 2h 22m 54s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/903c41018c7e46569ef42dc2c522ac1e : SUCCESS in 2h 01m 19s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f4e4bfae9f6f4d7f977734198c275b75 : SUCCESS in 17m 16s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9ff95bd514a43fd8299f20cffcf43e2 : SUCCESS in 13m 27s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/630d909a953e4943b5de66494699ff0a : SUCCESS in 30m 36s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2a425b1972754342aa13b9a4fd21e51a : SUCCESS in 31m 06s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/392de30b02ce4718947bac7029497648 : SUCCESS in 21m 11s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/13343bcececf4357921ff9f06084a92c : SUCCESS in 13m 57s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d9f47916a7c649518341b1336308f9b6 : SUCCESS in 11m 30s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a00f6f5aa2c94f06a84773eee98de84c : SUCCESS in 26m 43s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7692124891e466c910bceefbf6dd0e1 : SUCCESS in 23m 43s","accounts_in_message":[],"_revision_number":1},{"id":"4a63397c2cc2d8aebc2ca99b0584a8e9c7f58200","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-10-05 09:34:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b85a6025ea13a41f7257c0e787e18bfe9b3fb44b","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-10-05 14:19:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8279288f7b7102a600164a8dc8e3dc99c1151d99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 14:19:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1316041dabe17509478288f8e18205301111a1d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 16:41:49.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d959f24157a64c5c85afab348e7876ec\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af84a337745040e59117d10ad873f9cc : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cbd02efee5e34639af057927d1ae2220 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/76e59aebe9b44e85bc85844ce02ad232 : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f6267db192b745cdb853bdf6ddb59248 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75d78556e5db4536983068168939c983 : SUCCESS in 7m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/177bd1600a8f429393cfe641ff91f56a : SUCCESS in 4m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/940b743a2452499cb659d1b2cf354e6c : SUCCESS in 13m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/db2d17b2b12641d0aa4e68deefad2190 : SUCCESS in 13m 31s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/6706a11948e845a580fe641264d7fbe6 : SUCCESS in 1h 03m 38s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f80cad866ea64b4cbfd32aa518857b95 : SUCCESS in 56m 40s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/db75ee1362e4465f8bfd42c5cc7549da : SUCCESS in 1h 01m 20s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/cb4b826e276746ecb6029e0f814d5256 : SUCCESS in 17m 07s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a25129f08c44efdb4539f55732dab2b : SUCCESS in 14m 24s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/dc2685595b0345f1ac34781f10451a09 : SUCCESS in 2h 04m 25s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30519c108e7948bd9cd70b25136e5040 : SUCCESS in 2h 16m 59s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/eae213f9eb2d4771bc6c050d1033a579 : SUCCESS in 16m 21s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a54068740d64894a9e449ff83d66e6b : SUCCESS in 13m 26s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/cf2d8545b6b644ee8d93e2a0c8bf8863 : SUCCESS in 30m 49s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f1a6b958e5dd481eb269470748f6a27b : SUCCESS in 27m 45s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/188e1ac4ee2d401d9543421e192daf25 : SUCCESS in 11m 50s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29aecbaabb6c4408b14e9ff188c2b57b : SUCCESS in 10m 31s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b65b41493e7f44e1a0e99b4eec631924 : FAILURE in 10m 55s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/58638720bae44e7db196522120ca0ae3 : SUCCESS in 19m 07s","accounts_in_message":[],"_revision_number":1},{"id":"c9e74777f0565f4fc4d4d3ca21de082018c49dfa","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-05 17:26:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1a26d68716b4c68bc8a878ea98ab586354284ee5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 20:09:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e06a0dc586a54ee287bbe2d90f2ca847\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce5c7458a0814aa7af4cd6dac449fb3f : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f8e273cd7284b6db8a0c0bbe7f45574 : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2750c226ba01425fa8a668c4e3fcae04 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60808d8564c64e5c9fa419cb8fc90ceb : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/24afdc2f33a546ca9a13068553516a36 : SUCCESS in 4m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7fee048467ff4e04814f789b1da447fe : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1889114b9efd4aa0b1b591c47a012871 : SUCCESS in 6m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/49769b5b2207449499cda8d4353a22cc : SUCCESS in 4m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a1d27006550a42c7bf680a02fd26c0a0 : SUCCESS in 9m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cf155dd3c3fd4fdbb9237ded4b1e9bf7 : SUCCESS in 11m 57s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/6f8eb340e3984b3ea4be82a606b0334a : SUCCESS in 54m 21s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/9497fc5526fc4b4997d66e446334eb73 : SUCCESS in 49m 45s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/169561316f844f4884a39df7e2db13f0 : SUCCESS in 58m 45s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/5e38b256cda542158d07121131c6ce60 : SUCCESS in 44m 30s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ca6fec19b98e4c768d76597590d60781 : SUCCESS in 14m 29s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/ce5246e3497d4b058a5f3603199a5db2 : SUCCESS in 14m 45s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/50d6bd527d9d498c97904712d961bf19 : SUCCESS in 12m 16s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4f84bfb31d074a9fb41c53d3c6f40856 : SUCCESS in 2h 27m 02s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/537268fc36ed40a4a91ad210961c6f50 : SUCCESS in 2h 34m 38s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/9e47b143866846b2b9febf30925eaf9f : SUCCESS in 15m 09s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e4d8b540d7224fc7a0f0730195851b16 : SUCCESS in 12m 23s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f26129b4da3e4473b9ff4e22968e3763 : SUCCESS in 31m 44s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5e0b6dc52fb44c3a84d3be3b41e68ed8 : SUCCESS in 41m 07s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f59896ec48384fdf9dcbbb55ab1614b6 : SUCCESS in 21m 30s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/71543e3ad4a242c4ae4c82de703abcf7 : SUCCESS in 11m 11s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e4a3f90b404484e858908000eabae0d : SUCCESS in 10m 05s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7a9b284b7986426798316aae6be11929 : SUCCESS in 28m 26s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e5a8ec09e79e4077a2ef3329d051d53f : SUCCESS in 30m 37s","accounts_in_message":[],"_revision_number":1},{"id":"1d39bbd00f1b2a761303c563553e1c801bec79bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 20:09:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fca13ee993dc3ba891a4bae0fd2837ff6cc5c819","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 22:03:37.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/69ad6e69b3d045dfaedfa1f8fa9ea893\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c266c321f88048209444d8c3e072854a : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4909b121c8346ceaf8f5d6dafd04da7 : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f4d4fda1cea46748528d17364818118 : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c66bb8c26926405597725f2cef8a0632 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25c78ee78aaf493ca3f79bfb26fc421b : SUCCESS in 8m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/71188218cd444ed886cb256cffcdbdf5 : SUCCESS in 4m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1da8874b71764ef8b906e93c7bac423a : SUCCESS in 13m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e9563cba0682446a98618611bc9e60eb : SUCCESS in 12m 59s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/62b73a2017c5418fba6d6755541497dc : SUCCESS in 1h 00m 11s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7d6202ebacd4fb7af132c6d6353a569 : SUCCESS in 56m 51s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/5463a1b8b3134b839af01452c3390889 : FAILURE in 46m 19s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1dfeae08313c42fe8c2d043b3f86b14b : SUCCESS in 20m 10s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f16387d8b4e4d2c99ba307eeeff427b : SUCCESS in 14m 05s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e5793d372b9944d2a9ba6172ea76e5b5 : FAILURE in 1h 01m 54s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/61aaf104d4a846088dee15bf27a83b47 : SUCCESS in 1h 53m 08s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b34a1663346543c4b21ee54124c62e0c : SUCCESS in 21m 07s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a6a864767814a148bdb7a1593e5c967 : SUCCESS in 15m 35s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c5d7e61604ef498588ea2f7c9e7767d6 : SUCCESS in 32m 49s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0422c674179e40b189f1ce2a48855019 : SUCCESS in 29m 44s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b15e01fabbe24daba02fb83abe18d6e3 : SUCCESS in 14m 56s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/80750169de994a92be7666037e189aa7 : SUCCESS in 10m 29s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ebf498209e1d48f8878fa604d01df414 : SUCCESS in 28m 48s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b0e39c2a5e6341968e5e2709bdb48cec : SUCCESS in 22m 33s","accounts_in_message":[],"_revision_number":1},{"id":"1f4d3f352208fb1371b4d2685932fa79f564f413","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-06 19:23:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"313b30d4c31b7c56e3b8148b04d52faadcf8a471","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 21:47:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c6640180bb14e819f5f5392d8728626\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7dc94c275690479caadf071684025d05 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46b82490f5964141a35731dbf8ecac1f : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0971a633fde24b3ab9a052a4e598e9d3 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f20e2c78d31346e7878e5ccb72e84afa : SUCCESS in 5m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6c17a9a16755467791a8451de952408e : SUCCESS in 3m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2a5b6cfc2db7498691cb39ebba3d050f : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b445bdfd82674531af5d3ad435c567ad : SUCCESS in 9m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7fa4a7c468fa43259c5127fd9814ec22 : SUCCESS in 4m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ddc75222640d46a4b2831df9939f3106 : SUCCESS in 11m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7203e0ac664246bb9776564b0603b87e : SUCCESS in 11m 01s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/9d32ae5082424dddbde5c62f71cdf514 : SUCCESS in 56m 53s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/482f6a328faa4502a2f12c04000cf85c : SUCCESS in 56m 43s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/167addc62eeb4e5f9db59192678f5463 : SUCCESS in 51m 40s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/fe01c3ea02914693bc6f95990b171356 : SUCCESS in 1h 04m 57s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/dea6917da840436cb7b62a331993414d : SUCCESS in 14m 15s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/b84817b0bb894d77b992844f8b534dad : SUCCESS in 18m 50s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/40f8f351dddc46aa96f1e51c4fc6d49d : SUCCESS in 12m 54s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4e552c85668c43099ede4f5e99d3345b : SUCCESS in 2h 17m 04s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a93feaeea4854eba9e5c4dcb8445f492 : SUCCESS in 2h 13m 03s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d64b49bf56a8431b9cdd3a522ea4cd39 : SUCCESS in 18m 03s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2617700416b4486b92a87563a402a304 : SUCCESS in 18m 07s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/c9aceef951ab484a8c5832776ad65234 : SUCCESS in 33m 05s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/70c68ebf1e3b427b8c1ad0394f516fcc : SUCCESS in 31m 05s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7f58bb4192143609ee1cb43ea4091a6 : SUCCESS in 25m 13s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6e7cf1a1e65941bcb105acf85a79ef5e : SUCCESS in 16m 22s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/944313b6bcba4d8ca0946ddea9bb2601 : SUCCESS in 13m 30s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c3c78fdd21824aae953bab8c9721bee0 : SUCCESS in 31m 27s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e86c8a135c79406ab42f07985fa5063f : SUCCESS in 26m 47s","accounts_in_message":[],"_revision_number":1},{"id":"94e6fb2aef2b163d24033559d412b503cfbbd3b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 21:48:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"965578b2cf73f7c404b415964000ed8203983a3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 23:59:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06d9fae42ac243f3b05e0ff4727940a5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/759c365fc3a644a4bf4d32c4b32b4df7 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3ac921f76e34025b35574f8776616cd : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b8c50540b344914b7e8d33b6a4c5c3c : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94c14f8ecaec4900996c8d20e2219d99 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ada101c8818745a2a782ba1c7abec94f : SUCCESS in 6m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7d92260761e849e98ce9d5d0e948a921 : SUCCESS in 3m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5872a53631be4ee2afa8320a4727e33e : SUCCESS in 8m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/476fc55f0caf4724a6a138b099f711ff : SUCCESS in 8m 28s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/43ea42aaef3c46869d821657e26c6c16 : SUCCESS in 49m 19s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cde967930fd5467c88b8cb734b995be6 : SUCCESS in 46m 41s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ddf87e12afe041608afba43ec4beb016 : SUCCESS in 55m 44s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/65ed04ceb4184d48a299579bdc80a9a2 : SUCCESS in 12m 59s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/679c81aaae574e1f9d9b3e132cbc7309 : SUCCESS in 12m 10s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/cf4efd3a2c764ee6a0eaf4487c15875f : SUCCESS in 2h 03m 20s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e0b72b98fe2c4e50afca03c64849f52e : SUCCESS in 1h 58m 10s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/0aba3fe2be164b13ba8c05b42cc2aa48 : SUCCESS in 17m 23s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66b168a6ac4f4169bc097876d33635fc : SUCCESS in 14m 12s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/cbac82aa915d48bdaff75021fec5d971 : SUCCESS in 26m 34s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c7f9e6898df84400be55997ab8ec07bc : SUCCESS in 25m 00s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8a523584356a4cef9f7b391398045b5a : SUCCESS in 12m 44s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2d420cfc9755489b84dac9c1fd5c9267 : SUCCESS in 10m 13s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7efc8a26aef44b5e9c439bf5c022732c : SUCCESS in 33m 17s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ad676422f7f4ccea4ba50e91481e271 : SUCCESS in 24m 55s","accounts_in_message":[],"_revision_number":1},{"id":"04ca54fe35c12b1d44fdb8fdba9e2c2cb54539e5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 23:59:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a43cb99246fc0e91a305c94ffcfc3190d37356d0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-07 00:00:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbd9ac395f4b4092af42c58574cf4f0d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5297c0d05b404799b58dccc1a2694ee9 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8b465eae49b42e9d8c497ee8afb966cf5b983f7e","revisions":{"8b465eae49b42e9d8c497ee8afb966cf5b983f7e":{"kind":"REWORK","_number":1,"created":"2023-10-04 09:13:43.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/06/897306/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/897306/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/897306/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/897306/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/897306/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/897306/1"}}},"commit":{"parents":[{"commit":"cfaf571794296d6e0cfed2771f8046a8280ffb22","subject":"Work around mocking issues on stable branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cfaf571794296d6e0cfed2771f8046a8280ffb22"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-04 09:13:23.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-04 09:13:23.000000000","tz":120},"subject":"Use importlib.metadata instead of importlib_metadata","message":"Use importlib.metadata instead of importlib_metadata\n\nThe importlib.metadata library is part of the Python standard library\nsince Python 3.8.\n\nChange-Id: Iecf1d469caaa777c5b253eb0e44dc00692197ef7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b465eae49b42e9d8c497ee8afb966cf5b983f7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b465eae49b42e9d8c497ee8afb966cf5b983f7e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
