)]}'
{"id":"openstack%2Fkayobe~897314","triplet_id":"openstack%2Fkayobe~stable%2Fzed~Iecf1d469caaa777c5b253eb0e44dc00692197ef7","project":"openstack/kayobe","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-10-05 16:43:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iecf1d469caaa777c5b253eb0e44dc00692197ef7","subject":"Use importlib.metadata instead of importlib_metadata","status":"MERGED","created":"2023-10-04 10:20:36.000000000","updated":"2023-10-05 16:44:38.000000000","submitted":"2023-10-05 16:43:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897314","meta_rev_id":"fd7d43b4b7cda700aa783ef631e80c95bcc2052f","_number":897314,"virtual_id_number":897314,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-05 16:43:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2023-10-05 14:19:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2023-10-05 09:35:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2023-10-05 14:19:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-04 11:13:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-05 09:35:09.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-10-05 14:19:38.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"2b2665148a278c00e621fa880939db6e5536867c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-04 10:20:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9af75494efbb416fd161d4c56e407e47d29b9d70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 11:13:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffac432383ed4d37984dd74244596a21\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba6773df80ff40debc79920fe8eb4d41 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee3a2d2390e8485fa76f9dc1dfb9a192 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2d82aa060a94049a6de73f0d2203397 : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d5b72335ce1473ba7e44eb2a3a508c1 : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b9057809b74e49498f4d9b86e58148e6 : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea8596d8a732412085da5312fd7c34d2 : SUCCESS in 7m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bb8f7717ea344c3a8d6edc6adb3a288f : SUCCESS in 3m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d656e3acf3a24fb2b98f84acb711b046 : SUCCESS in 11m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3484e02ec0c34c7589a7d987280d6273 : SUCCESS in 13m 28s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/b4e5dfe44cb344248947fbf724a9bdfe : SUCCESS in 44m 50s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/0b549287f3984799aa8d40f611b378a2 : SUCCESS in 44m 32s (non-voting)\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a5492980440d4f61a0596466f73d4958 : SUCCESS in 50m 41s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/8adc23d6c84f4c4e82fec48f65f71f08 : SUCCESS in 17m 16s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/9c4f8de5bd3b4cffb8547642e5ed61f5 : SUCCESS in 17m 50s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2c7b26bc2624882b7f6dac05c7be4af : SUCCESS in 14m 51s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/1b43de5846944fdd8dec82ed765b81d9 : SUCCESS in 18m 53s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2dcad5715c83497392bae8ca0a5a78fc : SUCCESS in 16m 00s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/8e6be4dcb2cf4af4996639d65cbba51c : SUCCESS in 34m 35s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95778ad0e8004516b9d45b65dd39f34f : SUCCESS in 27m 17s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a0c06aa6ed9f47dc86b05f644a048ee3 : SUCCESS in 12m 05s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04970d86100c440a8c4717ea71d73f45 : SUCCESS in 11m 43s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7ecffc3a497d47e0b3c0caf951a66a0a : SUCCESS in 16m 32s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad648db830704a9eb9a8a1e319fd3693 : SUCCESS in 14m 24s","accounts_in_message":[],"_revision_number":1},{"id":"cb24333793d16a92d8d94dc134058ccf0add2a6a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-10-05 09:35:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8b902de8983069f1cc5773b3e8a24ad2d6f62b87","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-10-05 14:19:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b96155cfa2cd5a99aa52511f587a00d13fb6eda2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 14:20:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6fd970e9205c358c145e7b6870b92ecc27f20565","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 16:43:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/371bb13ec96c49568b4d68dfc7207335\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b3562b15aa14b8f8bd409e0bb27fdb9 : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ad6db27cd4c45bd815eb205329e53b6 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ef32a1ddc39433b85522236ade746f8 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/994bcb7ec2c94b4fa58ed32466568ea2 : SUCCESS in 7m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9163510020c94b4c976c368570bafa48 : SUCCESS in 4m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/aa31e060c31e453aa18e596c616431fd : SUCCESS in 11m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b2412f4c20924759b068736cbbeb0cb4 : SUCCESS in 14m 48s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0e781262703f4b4a9f863f87a3dc5660 : SUCCESS in 48m 28s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b2fb773f71a44953bd20203c572b4f15 : SUCCESS in 54m 02s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/10e86746f9e44397a97329d224b56ef5 : SUCCESS in 54m 40s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3e2b99e76dc043ac8dfb0fc3f7d2c832 : SUCCESS in 17m 48s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c229490574634ea9bf08489fbcdc3f02 : SUCCESS in 17m 32s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5be225b259914bb7959b9fb698e1529e : SUCCESS in 1h 30m 54s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/dd824f65ce2c417aa04b4a4993d0c8d5 : SUCCESS in 18m 56s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b7406baef3143e8b2d461db059c141a : SUCCESS in 10m 46s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f31f48d7bac4dabbc8968e8f4dd7806 : SUCCESS in 27m 01s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/da683c0d6f8745d3af8f960728af3099 : SUCCESS in 12m 57s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/098cb3432042492985ac3ebc33ea4bcb : SUCCESS in 11m 05s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a986ac6649604382b479abed7f1750bf : SUCCESS in 15m 27s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b094ee831c9d477dbb2555292f175c0a : SUCCESS in 11m 49s","accounts_in_message":[],"_revision_number":1},{"id":"fcb0676a5a90e86f8c71b153ff9ba3b980211e1b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 16:43:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fd7d43b4b7cda700aa783ef631e80c95bcc2052f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 16:44:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db52c17e3944440e971eb0af08ac7ef9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5635d4ec272542739d5d8d670e8d7cd8 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d773ed5682bc43531810b59d809231750e9c8205","revisions":{"d773ed5682bc43531810b59d809231750e9c8205":{"kind":"REWORK","_number":1,"created":"2023-10-04 10:20:36.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/14/897314/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/14/897314/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/14/897314/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/14/897314/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/14/897314/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/14/897314/1"}}},"commit":{"parents":[{"commit":"77a0e36fbf430706951138fbc4d187417ece20d3","subject":"Fix data file path detection with new pip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/77a0e36fbf430706951138fbc4d187417ece20d3"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-04 09:13:23.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-04 10:20:30.000000000","tz":120},"subject":"Use importlib.metadata instead of importlib_metadata","message":"Use importlib.metadata instead of importlib_metadata\n\nThe importlib.metadata library is part of the Python standard library\nsince Python 3.8.\n\nChange-Id: Iecf1d469caaa777c5b253eb0e44dc00692197ef7\n(cherry picked from commit 8b465eae49b42e9d8c497ee8afb966cf5b983f7e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d773ed5682bc43531810b59d809231750e9c8205"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d773ed5682bc43531810b59d809231750e9c8205"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
