)]}'
{"id":"openstack%2Fkayobe~897517","triplet_id":"openstack%2Fkayobe~master~I42c4a44afe8b0b19a2d7d14d9b748bf1e2ca5c10","project":"openstack/kayobe","branch":"master","topic":"ansible-timing","attention_set":{},"removed_from_attention_set":{"35264":{"account":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"last_update":"2023-10-20 15:31:30.000000000","reason":"\u003cGERRIT_ACCOUNT_35264\u003e replied on the change","reason_account":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-10-25 18:08:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I42c4a44afe8b0b19a2d7d14d9b748bf1e2ca5c10","subject":"CI: Add timing information to Ansible output","status":"MERGED","created":"2023-10-06 07:04:42.000000000","updated":"2023-10-25 18:09:03.000000000","submitted":"2023-10-25 18:08:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897517-ansible-timing","meta_rev_id":"ba7c4dfba91abe2a4672b48ab3f730e9c755be4f","_number":897517,"virtual_id_number":897517,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-25 18:08:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2023-10-20 15:34:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2023-10-20 15:34:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"value":1,"date":"2023-10-20 15:44:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-10-25 15:36:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-10-25 15:36:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-06 09:30:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-17 11:17:59.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2023-10-20 15:21:58.000000000","updated_by":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"reviewer":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"state":"REVIEWER"},{"updated":"2023-10-20 15:44:38.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-10-25 15:36:57.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"1b8454fe64271856f8c42dca6568f863b94d08b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-06 07:04:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea54d0a719fd356f070212b5d101f0be450118e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-06 07:05:23.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"faf122fdf170548a420f587c2eedc92ead1b8af5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 09:30:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb0e9515fa264ac8864f48d083e6a3f2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84ebedae3ac0470ead7a84493728db2a : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d0dfab5364243d89505033b4ec87f97 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1adc676f7fc24c11810fe46225afe89c : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/783ae775f29a4839bbbd9eb73cb0bc4e : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84b2cc2a750846ed8a23d4de7215184f : SUCCESS in 4m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/36afbc02312744f1946b029cdb054280 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03b7199248794c348b0212974394caf8 : SUCCESS in 7m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f8685ae8b42543e8b347e74f65019524 : SUCCESS in 4m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7d06a43312f1480bbb8a9e187b19972d : SUCCESS in 11m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c3d09cadead8450385df929997ea6d89 : SUCCESS in 13m 40s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2c1734b988214edb9bcdf0e0472452fa : SUCCESS in 1h 01m 27s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/f1b3d226220741a98ec6c7167396cd43 : SUCCESS in 48m 42s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/048d331c640442daad505f23d7ccfb63 : SUCCESS in 55m 01s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/58fa1add4a8740b7b025091c9ae569fa : SUCCESS in 1h 02m 16s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/765688b4f8484f22922419173c1ebbe1 : SUCCESS in 17m 27s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/9d29aa2218b64d359b87b7e0bd164cab : SUCCESS in 15m 58s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/80a59b0c87434d23a882df926a650f99 : SUCCESS in 13m 47s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/891273ea86034083a98f3d4474690024 : SUCCESS in 2h 20m 21s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c454453729f54259a7d43eb512082a6d : SUCCESS in 1h 57m 45s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2a67e71dd381488facb32eafad457463 : FAILURE in 10m 16s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c9c3b5f67074e8fa947d8a9720d156b : SUCCESS in 15m 16s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/77b53a87e80941378845ee774d14b4d0 : SUCCESS in 26m 44s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e3898f47f912450cbb98d5f611b4e69d : SUCCESS in 28m 10s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/17c403e8d8624b0ab7754294de0f94dd : SUCCESS in 27m 22s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c205e2e0944945fc89057275461e2bf9 : SUCCESS in 13m 27s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8cbe7f2247b41aaad09e72132e36093 : SUCCESS in 9m 47s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4340adb84fac40f390c33942bbd4ded8 : SUCCESS in 33m 33s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d7d48d4af6c048cbbd0acb9f6faeac66 : SUCCESS in 20m 33s","accounts_in_message":[],"_revision_number":2},{"id":"52e48f20379fb335e7a3b889e53804418c3fcf69","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-06 09:59:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"57867e21b795dfb15efd055ba26502d98b285978","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 12:16:22.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1ef22a71425b41668afce9bf425f9f7d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e8c2b94543243bdb602460bd389116b : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6767acb2b2694eb3a8fba1a11c0b467a : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ea5334f5c544a45af15983d98147712 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05f8b253180b426bb5969ec05b21fd08 : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b1c2c78c107e4755abc64cc614da2680 : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4d4b5eae23b04162b7e864d37232f156 : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f1f0a43a7dc40e89405dd2effa6ef8e : SUCCESS in 7m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b1aeb9968bab4d49b9a8eefca51abbd3 : SUCCESS in 3m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5cc0051ca01844d99fd006e0aeec4a6b : SUCCESS in 11m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/08b495af356543f2a0d7cf779805d160 : SUCCESS in 11m 42s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0f3f7e7efb864ade81b66df0d9d2505b : SUCCESS in 59m 14s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/1d8d36dff12048908994accb9f488bee : SUCCESS in 57m 26s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b97df3c699a04f939206462bb31c6d29 : SUCCESS in 48m 13s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8cf4ccb2425e496685072ba3da9a9909 : SUCCESS in 1h 02m 30s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e1882a73e18a49588934fca2c2ae8c64 : FAILURE in 10m 24s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/01e9b62bcd874f68b587da8934682d02 : SUCCESS in 18m 43s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3545078ebb514adab31f4b87c1ae0bfe : SUCCESS in 13m 06s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e78d8ad3c1b243a6b77e7fcff0631ffb : SUCCESS in 2h 15m 17s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ae383034b7d493f9fddb65e182084a6 : SUCCESS in 2h 04m 55s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/18edb528207b4e568f46c15acab44946 : SUCCESS in 17m 14s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc09019366ab4fde9429141582e297c7 : SUCCESS in 14m 49s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/fbeb1409f91b429b8561fc0f0bf3e41c : SUCCESS in 32m 28s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3727824345c14542b4557bfd352a5f49 : SUCCESS in 29m 41s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f4600fac3b784d968662999336b63f97 : SUCCESS in 25m 31s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/80814da472dc4f158fe770e813b0aba2 : SUCCESS in 14m 33s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/045991327fc840749af4188e0098dfb3 : SUCCESS in 12m 17s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ea65e871334a4fb79043b675d0b8a2f4 : SUCCESS in 25m 30s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d92580067d7458191edf01c377ee1ba : SUCCESS in 31m 09s","accounts_in_message":[],"_revision_number":2},{"id":"9f3673c0e1dbaabf643a93a90e03cfde06cd86b6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-06 14:17:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e4827c3bbcc6080eb35dfdc09313bb6e5912d509","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 16:43:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2876419b07384375bbbcef2a7eefcd21\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef6ee406d50c4293804e62ebbe1a54b7 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8236d6f05a244944810319bc2bba2599 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a86e11c00aa741b29d6aa45b4a5d3065 : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/427441ae1fd74062852ed0023e75013a : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94ebc2f66fa143fba07a5bec2fbf8fb5 : SUCCESS in 4m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b7c27524f2b24f778570ff5be1c47c41 : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7ce7a58cccc4cd59d86a96901523901 : SUCCESS in 8m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9ef250b019ef4552bf88df4ac0ea2299 : SUCCESS in 3m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7ec6b365142545e090b20f8a2e613f5b : SUCCESS in 11m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/221239c4ceb4479d9171bda7fdfc7788 : SUCCESS in 10m 24s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f1fefb95183c42f5a09b3aa8274bce91 : SUCCESS in 1h 03m 40s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/24076b388a1d4b189e61b6ddc154bbca : SUCCESS in 55m 26s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ebe8e8026fd480aa97ad9bed5e85ce4 : SUCCESS in 55m 23s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/45a8138792224a8eb7ed172fbedae0c8 : SUCCESS in 1h 07m 43s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5ea49ae0cea3472e8561b44afec5b881 : SUCCESS in 19m 11s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/fad48ec46fcf4964bb288d822ae59b25 : SUCCESS in 19m 03s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/340d1498c8054f538ce03693413b1f79 : SUCCESS in 14m 47s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/55457ddeb3844d41a9750e64b4460986 : SUCCESS in 2h 19m 49s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8846ade11eaa480e95e62e5096abd0ac : SUCCESS in 2h 16m 52s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a7d936f01abd427abef34e01e55bacf5 : SUCCESS in 18m 37s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c30d4a1afd8e4831b0fd116e887e03c2 : SUCCESS in 16m 11s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/5e2df881cade4d168f7963b3d19d38b3 : SUCCESS in 29m 16s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/aee2a4d91b8440f3944b56e849859ef6 : SUCCESS in 31m 26s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db4472f5777a49c5846c255dd1594ebe : SUCCESS in 28m 13s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/df1c802e1c4c478e84b2cf8ccf7ad820 : SUCCESS in 11m 59s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d57112f6a8f4983bf65ada3d9fe646e : SUCCESS in 11m 44s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bc85f269a8a44d54aca4af500014d4ba : SUCCESS in 35m 36s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bad0266b4c2c46c19574f2dff1c808f0 : SUCCESS in 21m 24s","accounts_in_message":[],"_revision_number":2},{"id":"8acacb283f6d28b79f797d5e5eb6ad2865d217e2","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-10-17 11:17:59.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d6ead6240e4cfdf5ed8614366c21024357f3cf6e","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2023-10-20 15:21:58.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"071dccf24efeb89fdee83503cc2b78b481c34d20","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-20 15:30:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3abcaed7fcc1718dc3eac25dd8a193b7950adff8","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2023-10-20 15:31:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"af513abc1cc0034510f4540ce9c7b260cf98db49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-20 15:34:23.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"41a4d43f24670887ef19441bfd3b0cfd19ff8dac","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-10-20 15:44:38.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"15179c08130b1a22f02a780128f251f5ca8ca42d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 17:57:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9993795d60d4ab082c5fd93f87f9578\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8d8ea933ee0409cb5c6edf4745eb516 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4e097ebf2454eb386a275bfcd653219 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f630c197f11d420aacfadb080b2d8f3b : SUCCESS in 4m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/37c1fb84398547da9372f24780af520a : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77cb9d8d5ab34ada8e594086b5a3ec6a : SUCCESS in 7m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bd6b5156d017431896310d351f5d21e4 : SUCCESS in 4m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5cc45f33cfc74caabda5b69fd5bfa748 : SUCCESS in 11m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/29ccead47139473e85f2ccbc3d90a0e6 : SUCCESS in 10m 42s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/554c771685d449af8e99ee34abf33f67 : SUCCESS in 58m 55s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/b3ecb26129a94ca38a8c7ff099b76519 : SUCCESS in 1h 03m 41s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e04b6f2cf3b0466cb003bfc356b2e579 : SUCCESS in 53m 25s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d4ae94170369440ea0adb6a84142f0ce : SUCCESS in 54m 04s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/bd92454195b84d1cbd77366d74adbbde : SUCCESS in 15m 18s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/194adb60118a4799b87b63efb4311706 : SUCCESS in 17m 50s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c0301c963e6479a8c7f094c95493a95 : SUCCESS in 14m 16s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/39aa10ac9a9e4d898287d469ed6cf2ff : SUCCESS in 2h 15m 48s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/132dbdda8ede4c5aba79a0f6772e8c6c : SUCCESS in 2h 19m 23s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/68bca5a91b8248c095b78f304bdfd763 : SUCCESS in 17m 04s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/49ef3fc87ce64c9db6da6555fcdf0932 : SUCCESS in 11m 27s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/e32a4462340a4d659b79de325910a067 : SUCCESS in 33m 28s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0c2ef8d6e4c540d9908ae23cd88d897b : SUCCESS in 27m 26s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/322f966d4f9b4904a8b8e4c5960cf61a : SUCCESS in 24m 23s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/cd32cd0a64cc4821b6347dc1f7f0f04e : SUCCESS in 12m 38s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97a638dc351d4fdb873d8977b3bef8b1 : SUCCESS in 10m 54s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fbd3d5011cc949d4a6edddf7d20612a3 : SUCCESS in 17m 56s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0413f7176338498f89d0c25e0ba365be : SUCCESS in 15m 54s","accounts_in_message":[],"_revision_number":3},{"id":"174749fea7fe2fbfbcfa9046d5e0c0d2e8c6d90d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-10-25 15:36:57.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8b30a5a4a2b3c8402171f4b469116c6f095a0d24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 15:37:27.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7ca672ecbeccf5595833b26e82426042be8696f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 18:08:02.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/544397b12ffa4cafba81bf0764819834\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f5e06dd24854fb99948df8b83ecc18a : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b54133eb7fc0461eb4ab562d00413c76 : SUCCESS in 3m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a89edaae4f7c44fabfa7de07854c11f2 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70ce5352b1d2411a9f4eaab3d8ea3b1b : SUCCESS in 6m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8f70e8becff84e269a510e44a8a5102d : SUCCESS in 4m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d50511ababf441c6803413137ea067bf : SUCCESS in 9m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2b24f505f2f94eb19af01563571923b3 : SUCCESS in 10m 33s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e52ba8b737304662ba4a8ecc12d6ffa6 : SUCCESS in 1h 08m 25s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d400b99b6d740c8b2ec11efbe1f5931 : SUCCESS in 53m 49s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a85e615c6f23457f993059cddd856160 : SUCCESS in 1h 09m 19s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/efa8b9f561814af9aed9b1c1a632325c : SUCCESS in 16m 12s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/19e9c44f86bf4b26ae8b4c18d24a4deb : SUCCESS in 13m 30s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e991f8025c4b48b2a2483333cafbe2ad : SUCCESS in 2h 25m 06s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2e2ccf731c164e798d5413ec3049f095 : SUCCESS in 2h 06m 11s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3ffe259fae6043caad7557efad471d01 : SUCCESS in 16m 09s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/19333dbab2db446eb545cc78191ca3a2 : SUCCESS in 14m 47s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/345c079793174ee2809f88165ecb5a8d : SUCCESS in 30m 14s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df35334dc9bd4400bb6b20a97e0894ac : SUCCESS in 23m 22s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/438c8e50c67e4765b998c86f47566eee : SUCCESS in 13m 24s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99c9e3e4732e45a69b4aa3d624feb990 : SUCCESS in 10m 51s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d2611b4c721344aea48ce1496833dbaf : SUCCESS in 18m 13s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39528a891a4949d7ac8b8a3cece46f12 : SUCCESS in 12m 23s","accounts_in_message":[],"_revision_number":3},{"id":"e6461439075b06c37fb2fa8f076146a9378b0e2d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 18:08:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"ba7c4dfba91abe2a4672b48ab3f730e9c755be4f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 18:09:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a00a187c19f449ab81934bf0faa344df\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4bd9f484e31438b8a907cd44827c702 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8163f9ba79f0c6d162cfcd9cc0000615dab1cfe5","revisions":{"790dc596db71a5d8803c4029d57c5557c3c2d477":{"kind":"REWORK","_number":1,"created":"2023-10-06 07:04:42.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/17/897517/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/17/897517/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/17/897517/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/17/897517/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/17/897517/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/17/897517/1"}}},"commit":{"parents":[{"commit":"cfaf571794296d6e0cfed2771f8046a8280ffb22","subject":"Work around mocking issues on stable branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cfaf571794296d6e0cfed2771f8046a8280ffb22"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-06 07:04:22.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-06 07:04:22.000000000","tz":120},"subject":"Add timing information to Ansible output","message":"Add timing information to Ansible output\n\nChange-Id: I42c4a44afe8b0b19a2d7d14d9b748bf1e2ca5c10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/790dc596db71a5d8803c4029d57c5557c3c2d477"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/790dc596db71a5d8803c4029d57c5557c3c2d477"}]},"branch":"refs/heads/master"},"3c1968d9067f525f8ea84d780fe6009dacceb6fc":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-10-06 07:05:23.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/17/897517/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/17/897517/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/17/897517/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/17/897517/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/17/897517/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/17/897517/2"}}},"commit":{"parents":[{"commit":"cfaf571794296d6e0cfed2771f8046a8280ffb22","subject":"Work around mocking issues on stable branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cfaf571794296d6e0cfed2771f8046a8280ffb22"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-06 07:05:12.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-06 07:05:12.000000000","tz":120},"subject":"CI: Add timing information to Ansible output","message":"CI: Add timing information to Ansible output\n\nChange-Id: I42c4a44afe8b0b19a2d7d14d9b748bf1e2ca5c10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c1968d9067f525f8ea84d780fe6009dacceb6fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c1968d9067f525f8ea84d780fe6009dacceb6fc"}]},"branch":"refs/heads/master"},"8163f9ba79f0c6d162cfcd9cc0000615dab1cfe5":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-10-20 15:34:23.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/17/897517/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/17/897517/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/17/897517/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/17/897517/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/17/897517/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/17/897517/3"}}},"commit":{"parents":[{"commit":"4e2a0d2e8d3a92946e1b0eb0266670c190721a06","subject":"Merge \"Fix last return value in _get_base_path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4e2a0d2e8d3a92946e1b0eb0266670c190721a06"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-06 07:05:12.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-20 15:34:23.000000000","tz":0},"subject":"CI: Add timing information to Ansible output","message":"CI: Add timing information to Ansible output\n\nChange-Id: I42c4a44afe8b0b19a2d7d14d9b748bf1e2ca5c10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8163f9ba79f0c6d162cfcd9cc0000615dab1cfe5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8163f9ba79f0c6d162cfcd9cc0000615dab1cfe5"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
