)]}'
{"id":"openstack%2Fkayobe~897820","triplet_id":"openstack%2Fkayobe~stable%2Fzed~I2cfd2667abb1ae8988b7a7fd9761b75c20a0eaa4","project":"openstack/kayobe","branch":"stable/zed","topic":"libvirt9-tenks-stable/zed","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-10-23 16:31:22.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I2cfd2667abb1ae8988b7a7fd9761b75c20a0eaa4","subject":"Revert \"CI: Disable bare metal testing on RL9/c9s\"","status":"MERGED","created":"2023-10-11 12:01:02.000000000","updated":"2023-10-23 19:45:39.000000000","submitted":"2023-10-23 19:44:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897820-libvirt9-tenks-stable/zed","cherry_pick_of_change":897809,"cherry_pick_of_patch_set":2,"meta_rev_id":"df13902a4e193d7e29f2a4b2febf0f7b752c3f11","_number":897820,"virtual_id_number":897820,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-23 19:44:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2023-10-16 08:14:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2023-10-20 16:10:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-10-23 11:03:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-10-23 11:03:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-11 13:35:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-16 08:14:26.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2023-10-20 16:10:18.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-10-23 11:03:46.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"6671c0a9726791dc0e3c49031f91fef47867ce75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-11 12:01:02.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.1.","accounts_in_message":[],"_revision_number":1},{"id":"cadcf522601a382aa1c9a56b6b3226bc7275953c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-11 12:03:03.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7df46a622494a83c25962ad472b194d51e874a9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 13:35:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7ff07679adf4c44a7d213c539e1412e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01b134d988c94548bdd7b6d0a03f2763 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86d19bbe61ef4ecbb134d5eff8c23562 : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12d4d44457604e5da60cec0ad476483c : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2cd651424d4746ee9cc1b25178c046d6 : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/385cec07c25c485f826e709dd8be4921 : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8b0c3d769a74c0f9997da4a4698bdcd : SUCCESS in 7m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1ea9f838d783427b9552fa0822e2a9df : SUCCESS in 4m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1fbef30e8e164ae38ee57b27dc0e71db : SUCCESS in 10m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c4a111ab2e6c489182f11c75459b075c : SUCCESS in 11m 39s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/45c8bd806c6a40708e7f0f413526cffd : SUCCESS in 1h 01m 10s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/715ce00ee7dd47d6a09b6cbe1430ac97 : SUCCESS in 50m 56s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a5b7029a9184fd897fbb962cca27b62 : SUCCESS in 50m 03s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/90f903cd55db4a89b2dfe89a27dddd45 : SUCCESS in 55m 44s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3db20d95f4b04a4a935afc1731f61654 : SUCCESS in 18m 26s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/9d953cd3846b4ec181e97b10cf9e0436 : SUCCESS in 17m 06s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/335186e8b8de41dd803223ccc1910758 : SUCCESS in 14m 42s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b72137d565074941a2da1b9929706fad : SUCCESS in 1h 27m 02s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/cae8d46e55bd4c72871b2cbb874d10de : SUCCESS in 17m 23s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3455aeb960f24fa884dd791861b0fd03 : SUCCESS in 14m 22s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/3cc25ccfebc748b08f7f9f3386c422c7 : SUCCESS in 28m 05s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4ccb738e10c40ceabdc020b49d77b27 : SUCCESS in 24m 17s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a52f2231670344358b4b19dca9b065c2 : SUCCESS in 10m 22s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/773ec9d607a4494e851bca6801acc87f : SUCCESS in 9m 46s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8953e792fb674f728d4146fb4e58540b : SUCCESS in 14m 33s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ab3067032574091a88e08fbd6105fea : SUCCESS in 11m 55s","accounts_in_message":[],"_revision_number":2},{"id":"2667cc4857da03b63763650199bb89a605696bc8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-12 09:49:01.000000000","message":"Topic libvirt9-tenks-stable/2023.1-stable/zed removed","accounts_in_message":[],"_revision_number":2},{"id":"ca76e93c1f5e375a7d216974f83a81557781d5a6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-12 09:49:06.000000000","message":"Topic set to libvirt9-tenks-stable/zed","accounts_in_message":[],"_revision_number":2},{"id":"e9bb05e16245ad2c325c7416db2473e8cf442638","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-10-16 08:14:26.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"786f39ab33a9ba83264623de3c99a62ea9380f45","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-10-20 16:10:18.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"73613c892cef19dd67981b54af05a4c9d7a81795","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-10-23 11:03:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0e0d6b95f2211327bc6dd0b631382eced8395a96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 11:04:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1e1a68609580786b2365cf3138f69269f494cc0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 15:05:59.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4de542c86448497d9e8d05d94d500e9f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f14d3f233804b26bd27627c34bd2bcf : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67b5e572e528464a8a27f7c24bb01f2a : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a253350c7a244436a6377097177ae031 : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f015a80e677c4e7288ccc29aa77c5c5d : SUCCESS in 8m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ec496460a7ca4a37b8aceaff1fd9f84a : SUCCESS in 3m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ced7870892c8425fbb9a61e47285d87d : SUCCESS in 10m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d0206dcae8984b95bc41594999baf655 : SUCCESS in 11m 37s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/3f8eeadfdc7e4eccb0deb485b0ba275f : SUCCESS in 59m 08s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/506555e1ff0a421d8812f4974698657d : FAILURE in 38m 29s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e1a3b443d3504f358ad7cc34b612d57f : SUCCESS in 57m 47s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/fb2093b4024a4671adacf4eefa842eda : SUCCESS in 17m 08s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c22b4c73409f4fa3a29cfd45830efb40 : SUCCESS in 14m 28s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f95cc0dca1914945a96a0b41d1e9effb : SUCCESS in 1h 30m 26s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/03586a73fb98454faf42702818d72a57 : SUCCESS in 17m 31s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2806686a089a49719fa19837b3262c2c : SUCCESS in 15m 51s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45ead7607ec546b8b3711f63876d2246 : SUCCESS in 23m 32s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9be8ceac6428434f8689e9bd9bd1ef60 : SUCCESS in 13m 06s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6521642c5a3e49f4bfd0acee3058d178 : SUCCESS in 11m 16s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3d2c156dee38429aa42a7ba9791f9c38 : SUCCESS in 13m 29s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/856a7684f225422fbc2d17652651a125 : SUCCESS in 12m 01s","accounts_in_message":[],"_revision_number":2},{"id":"386cf07c1e97dd86ef94145e9d7d953df88cd2d7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-23 16:31:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5616a9585ebe06f78ad669be6ad2e9b6873395ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 18:08:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92c995e2f0e94df485b4fd56f2d1f85a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d318d029c7134fcdb9e1a3e18e09a444 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4ddba753c274d248488aa8b6c6a1695 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d61752606ff42d6b0dfe84c6399c96a : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f74bbdb578a2457eb6178e8cf8b9d8b5 : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/72d9c1507fc345209e2c969d8d730e09 : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80d1ac6887e847c48d3d2c8624b53b12 : SUCCESS in 7m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7b58f794fd0f429d99c889dbc169e21d : SUCCESS in 4m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9abee041649646018769e9445a25edd4 : SUCCESS in 10m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2764e141ee7c4f339da53f7acabb3daf : SUCCESS in 11m 43s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/c7b278002add4332b2db649ed7e231cc : SUCCESS in 53m 38s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/a24a3f4a25614921beb50e815bd9820d : SUCCESS in 51m 29s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e1b15ae9557c446d91f3f3e3ce22e316 : SUCCESS in 49m 09s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/06a9c04987cc4d35bf24cff97bfbc48e : SUCCESS in 1h 04m 05s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0f5286cdc4c74bf28ba0180d9a8a8f42 : SUCCESS in 14m 07s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/12beec6815e24687bc5bead250bee59c : SUCCESS in 16m 39s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bd598d931fda42eda44504f81186ec10 : SUCCESS in 15m 32s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9420c92c5ad24dcfb9bb53c83368f782 : SUCCESS in 1h 30m 10s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/6c8fe21a26e14f809a2d8454ffc657dc : SUCCESS in 17m 49s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e19f83475fe40c0aa6fe4aa7b073ea1 : SUCCESS in 15m 01s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/4cada7fd57014a46a06410063a03200d : SUCCESS in 29m 41s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6dfe18ac7cc745d18f63210032cb6283 : SUCCESS in 21m 57s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/24508e310edb4d71b0cdbd1396ca1984 : SUCCESS in 13m 44s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/56f0426809c64eb5a1b9021a075b4aa2 : SUCCESS in 11m 34s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/adb288bd83564bbeb5c5851f44953583 : SUCCESS in 16m 31s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9787d5b0971a4e7a87550a37030c753b : SUCCESS in 11m 36s","accounts_in_message":[],"_revision_number":2},{"id":"006324ab8ffd251365b02bd0ed7c5a23a001d79a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 18:08:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fe4d0f5eb8fa8ae7054610aa70bc678b9f5b47a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 19:44:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc03745871144b4f8ce9c267fa3080b5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c11477cc7bb34424a0ff9ba3807e922e : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb11f1b907594d75be196e1faf0aca2f : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/edc7359f54fc4c8e85b6ed9d8ef778a9 : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83675b6cea3c4115bf430234fe9d8ecd : SUCCESS in 7m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/34d8d582a78d4d66bcca83a075133d34 : SUCCESS in 5m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7189958d38a840c2af065a5356cd8545 : SUCCESS in 10m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4b878a9c07c844c0b4887f3a4ae62b26 : SUCCESS in 15m 24s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/bc49664d5f9441658dcc94f34b80c741 : SUCCESS in 51m 01s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ec150bc97562407499f32a9f5a365941 : SUCCESS in 46m 29s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/40d896a7fcf44ee8b13dfb2f60b00e86 : SUCCESS in 57m 04s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/79d50a535f234d41a307a57dce8d859a : SUCCESS in 16m 28s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b2c772e165b4ecd8aeb190d9a3e39f7 : SUCCESS in 15m 28s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b28def4047d54c37b6c5d54b2d8ad213 : SUCCESS in 1h 35m 46s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ab3244ac7657454aafc06bcfbba6d11c : SUCCESS in 16m 44s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d9ea13458c03465f812da6c7890e64cd : SUCCESS in 15m 25s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e1dfcde08d14483ad3df2d2ca897f97 : SUCCESS in 20m 46s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d022b46870fc4da79801e6595ff9a493 : SUCCESS in 12m 49s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/973f56d97ca8455091f6ca443fa9843b : SUCCESS in 10m 25s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6f4d10bc1f524213923e5109e7d91f41 : SUCCESS in 14m 03s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76f2017b552b4b6ebfd628761b2660c8 : SUCCESS in 13m 42s","accounts_in_message":[],"_revision_number":2},{"id":"a27f371b58f099de96e02a78d2fb2d4aed6265d9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 19:44:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"df13902a4e193d7e29f2a4b2febf0f7b752c3f11","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 19:45:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22e034ee109446939cf17164d492d0c3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4fae48bfb034907b5d4a74e0df4836d : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8b7b8e373982b9f2e315afe1e16c58a270ed185b","revisions":{"59cc545434d4b1e4fe83222f4ddb3de594832f3f":{"kind":"REWORK","_number":1,"created":"2023-10-11 12:01:02.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/20/897820/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/20/897820/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/897820/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/897820/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/897820/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/20/897820/1"}}},"commit":{"parents":[{"commit":"c4a6883f1b2040c3553e0f29555264b44f15220f","subject":"Merge \"Use importlib.metadata instead of importlib_metadata\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c4a6883f1b2040c3553e0f29555264b44f15220f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-04 06:57:52.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-11 12:01:02.000000000","tz":0},"subject":"Revert \"CI: Disable bare metal testing on RL9/c9s\"","message":"Revert \"CI: Disable bare metal testing on RL9/c9s\"\n\nThis requires disabling libvirt_vm_trust_guest_rx_filters, which when\nenabled triggers the following errors when booting baremetal instances\nwith Tenks on Libvirt 9 (and most likely since 8.9.0):\n\n    Cannot set interface flags on \u0027macvtap1\u0027: Value too large for defined data type\n\nThis is apparently triggered by a Libvirt commit refreshing rx-filters\nmore often [1].\n\nAs explained in I71a2051d8acd63379bd70bc1287a059d4a7f6387, this setting\nwas added to allow traffic destined for other MAC addresses to reach VMs\nwhen using a macvtap interface.\n\nThis will prevent multicast from working, but we don\u0027t need it for\nbaremetal tests in CI.\n\nThis setting will be enabled again once the issue is resolved in either\nLibvirt or Tenks.\n\nThis reverts commit 21c68bbfafe529e1c337ba242c2e501c75bfedaa.\n\n[1] https://gitlab.com/libvirt/libvirt/-/commit/060d4c83ef436cf56abfad51a4d64c39448e199d\n\nChange-Id: I2cfd2667abb1ae8988b7a7fd9761b75c20a0eaa4\n(cherry picked from commit 990370a3673b8bdf4882816926868dd7b422db60)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/59cc545434d4b1e4fe83222f4ddb3de594832f3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/59cc545434d4b1e4fe83222f4ddb3de594832f3f"}]},"branch":"refs/heads/stable/zed"},"8b7b8e373982b9f2e315afe1e16c58a270ed185b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-10-11 12:03:03.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/20/897820/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/20/897820/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/897820/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/897820/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/897820/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/20/897820/2"}}},"commit":{"parents":[{"commit":"c4a6883f1b2040c3553e0f29555264b44f15220f","subject":"Merge \"Use importlib.metadata instead of importlib_metadata\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c4a6883f1b2040c3553e0f29555264b44f15220f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-04 06:57:52.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-11 12:02:54.000000000","tz":120},"subject":"Revert \"CI: Disable bare metal testing on RL9/c9s\"","message":"Revert \"CI: Disable bare metal testing on RL9/c9s\"\n\nThis requires disabling libvirt_vm_trust_guest_rx_filters, which when\nenabled triggers the following errors when booting baremetal instances\nwith Tenks on Libvirt 9 (and most likely since 8.9.0):\n\n    Cannot set interface flags on \u0027macvtap1\u0027: Value too large for defined data type\n\nThis is apparently triggered by a Libvirt commit refreshing rx-filters\nmore often [1].\n\nAs explained in I71a2051d8acd63379bd70bc1287a059d4a7f6387, this setting\nwas added to allow traffic destined for other MAC addresses to reach VMs\nwhen using a macvtap interface.\n\nThis will prevent multicast from working, but we don\u0027t need it for\nbaremetal tests in CI.\n\nThis setting will be enabled again once the issue is resolved in either\nLibvirt or Tenks.\n\nThis reverts commit 21c68bbfafe529e1c337ba242c2e501c75bfedaa.\n\n[1] https://gitlab.com/libvirt/libvirt/-/commit/060d4c83ef436cf56abfad51a4d64c39448e199d\n\nChange-Id: I2cfd2667abb1ae8988b7a7fd9761b75c20a0eaa4\n(cherry picked from commit 990370a3673b8bdf4882816926868dd7b422db60)\n(cherry picked from commit 46d15a76304f51b02836e37c5bebf1ebab875db5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b7b8e373982b9f2e315afe1e16c58a270ed185b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b7b8e373982b9f2e315afe1e16c58a270ed185b"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
