)]}'
{"id":"openstack%2Fkayobe~897866","triplet_id":"openstack%2Fkayobe~master~I7ef0c5240585bfdc27fbdf6b32edd5b51ce2d147","project":"openstack/kayobe","branch":"master","topic":"dib-remove-disable-selinux","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-10-26 15:38:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7ef0c5240585bfdc27fbdf6b32edd5b51ce2d147","subject":"Remove disable-selinux from default DIB elements","status":"MERGED","created":"2023-10-10 19:04:30.000000000","updated":"2023-10-26 15:40:04.000000000","submitted":"2023-10-26 15:38:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897866-dib-remove-disable-selinux","meta_rev_id":"107d4ae77e764c8f13d28613abeb6e44a98d936f","_number":897866,"virtual_id_number":897866,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-26 15:38:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-10-26 12:45:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-10-20 16:15:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-10-25 15:41:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-10-26 12:45:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-10 21:37:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-20 16:15:29.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-10-25 15:41:57.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-10-26 12:45:13.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"169739ecefa417b72a2bf99959706f30d474fd17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-10 19:04:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7006622704ebdc7e3cfd03419755a56acd65f23b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 21:37:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f0987acd120427ca2a4dc4d5affad0b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5303c955fc84d88b31bf4ea91209f4a : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e73de846791644b7a9b8a4a6851086d5 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/617a6543b5484bbb91164a612d68999a : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8bcf23ed6f0049aa9c20eeab877d92ec : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/254ffc8d1e2c468fb65970ed31a0eae8 : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6154f73073e84db3a2f8a7189901f562 : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6588469f741244ef94bc366b6e88398a : SUCCESS in 6m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/491a1199d4f144908efef6c31ca2431f : SUCCESS in 3m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/117ab06230df474b895083ec1de12ca2 : SUCCESS in 6m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/54264d2c8634497b93f83593bda6ed42 : SUCCESS in 10m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8f852204551b43e5b40a7cb5f336602a : SUCCESS in 12m 53s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/255df51e718f400fa7bba6b81fcdd9a5 : SUCCESS in 1h 01m 27s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/2b8b3de1a0764e47b9c049b559e1331e : SUCCESS in 1h 03m 20s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c602fd787dc346b9ad7f8481a3626a15 : SUCCESS in 50m 55s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/25a2a596be58425e91a453449fa7facb : SUCCESS in 1h 08m 11s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1f935aa9aa0645fd9d5182c4440e2b21 : SUCCESS in 16m 48s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/018137ad1a7e4af6a79f68add9a52ad0 : SUCCESS in 16m 32s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df84b0ea775a44a7857090296f8000f2 : SUCCESS in 15m 40s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/feebe459835c4e4e9cdeff3b34adcd0a : SUCCESS in 2h 28m 22s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b6961bdd50c74fcc829198db7265c70e : SUCCESS in 2h 08m 10s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/99f9f92b89cf42aaaa089929569b0e2f : SUCCESS in 16m 30s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/73dc484e838f4be4a9653a30bb96cea3 : SUCCESS in 16m 02s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/ac813c50353945b88810a964e229bc70 : SUCCESS in 24m 21s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5e92b99822684f6ea774477310ad0a72 : SUCCESS in 23m 35s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0daf4d0e7a0b42d696e93c4b1731ca8d : SUCCESS in 22m 50s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0bbfc8434e814fe2857b9fc7a808d6c6 : SUCCESS in 13m 41s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e9222d451f0460094e327fc7c7a57ce : SUCCESS in 10m 08s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/df9fe27f15bd40d3a890341d4e45636f : SUCCESS in 18m 46s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1825a46f4afa49efbad27aa113b3df34 : SUCCESS in 13m 17s","accounts_in_message":[],"_revision_number":1},{"id":"d522e2ab76092362d465079aa15b0934c10e3c66","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-10-20 16:15:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a778df14473b9765e97eca1c6bb5e06f8ae0077d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-10-25 15:41:57.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"29d9c413cf3b172efdbfd90b0a08f00c63bec89a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-26 12:45:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"797d5819250ebac0fec5cf3a0ac1077a1da55baa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 12:45:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3fa33882b9efde2dbe92b4838aba6b8274662949","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 15:38:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b737b4ac7bb64d93806d8905e5b703e0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25b310e0f8754c9fba6f3ada126e469f : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48bda77b87904c6d94335912ff821a00 : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5bd79b09ed234d1288b7d01a1c2a8872 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11e009103833480e9341d40840ac06fc : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c7f9368ff8e475a9d57a480326a6925 : SUCCESS in 3m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a62020a662d3405c89c40e9692f04aac : SUCCESS in 4m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5151e75ea9f744feb3eacaf7e85d7373 : SUCCESS in 10m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e44caeca949a4e6ba3470b0a20e0fd3e : SUCCESS in 11m 26s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/939df202d9d54c759f22d7c757b30f0c : SUCCESS in 1h 00m 06s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0086ac85cbb04caf9f8e4a7ce70d6450 : SUCCESS in 52m 32s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/fc0d106f83b64dc0b153e1c1d54d692a : SUCCESS in 1h 04m 11s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/23be557d4ab947b3a29372a97f71670d : SUCCESS in 13m 33s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1362ceb358df41b28467ddd80c254b1c : SUCCESS in 13m 29s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/addfd052d4c94c719fb6062c8aed936c : SUCCESS in 2h 15m 43s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9d2dc4397de24105bf7b96d5ee658314 : SUCCESS in 2h 45m 45s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/66db60c8914742c0bf062e77c82314af : SUCCESS in 18m 13s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/36c6cd5a7f404e42b687313a484ed990 : SUCCESS in 14m 17s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a73928ef0cd447dc840df0ab7ee2628a : SUCCESS in 29m 22s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b6d5a5b688a444d1a3bfc5378deb7be4 : SUCCESS in 22m 19s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f23797226b4a4f86b142bce070fae4c8 : SUCCESS in 12m 38s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e534e0effe54299a9868a38c69e4c75 : SUCCESS in 10m 56s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5d30a679dbf44fa9ac59d1bfc801078d : SUCCESS in 17m 05s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/151fb460c03c45f1bf1d3c16cbc21d6d : SUCCESS in 14m 09s","accounts_in_message":[],"_revision_number":1},{"id":"1ca800ef3c8899f7d9094fc6a18953425a7dd490","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 15:38:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"107d4ae77e764c8f13d28613abeb6e44a98d936f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 15:40:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0634f29093645918095397334fdb94f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6991cbd3abf346ff95b3ac402da190d8 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d21a14442a474841b6dabca10f2e5d3a : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d149118449140577bd6105dc21803ae0d559aebe","revisions":{"d149118449140577bd6105dc21803ae0d559aebe":{"kind":"REWORK","_number":1,"created":"2023-10-10 19:04:30.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/66/897866/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/66/897866/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/897866/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/897866/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/897866/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/66/897866/1"}}},"commit":{"parents":[{"commit":"990370a3673b8bdf4882816926868dd7b422db60","subject":"Revert \"CI: Disable bare metal testing on RL9/c9s\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/990370a3673b8bdf4882816926868dd7b422db60"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-10 19:04:08.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-10 19:04:08.000000000","tz":120},"subject":"Remove disable-selinux from default DIB elements","message":"Remove disable-selinux from default DIB elements\n\nChange-Id: I7ef0c5240585bfdc27fbdf6b32edd5b51ce2d147\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d149118449140577bd6105dc21803ae0d559aebe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d149118449140577bd6105dc21803ae0d559aebe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
