)]}'
{"id":"openstack%2Fkayobe~898066","triplet_id":"openstack%2Fkayobe~stable%2Fzed~I6ae44dcae78424b4638762d015cf24336444707f","project":"openstack/kayobe","branch":"stable/zed","topic":"bifrost-deploy-kernel-stable/zed","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-10-23 13:45:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6ae44dcae78424b4638762d015cf24336444707f","subject":"Add introspection rule to update deploy kernel location","status":"MERGED","created":"2023-10-12 07:09:43.000000000","updated":"2023-10-23 13:46:34.000000000","submitted":"2023-10-23 13:45:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898066-bifrost-deploy-kernel-stable/zed","cherry_pick_of_change":808111,"cherry_pick_of_patch_set":3,"meta_rev_id":"7a34bc2ca3e898e770a5000bf1528fa692af79b9","_number":898066,"virtual_id_number":898066,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-23 13:45:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2023-10-16 08:15:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2023-10-20 16:07:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-10-23 11:04:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-10-23 11:04:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-12 08:32:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-16 08:15:39.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2023-10-20 16:07:32.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-10-23 11:04:21.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"8fb0262e1e28430340c67fa5681d6064a332ca2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-12 07:09:43.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b267cf9ce8a948dc61a1dcd5811fddef2202fe25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 08:32:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3849d8b0b3be4f9fa04480ddf7f1ed6a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/473fba2a7d284606b331ca67af82c31b : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45f5d9e65b7d4a70905cc2d0620663dd : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19980dec308b42cb8aeaa2620948780e : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca2da3ff3d9b436db878a7363b06e161 : SUCCESS in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6e8895ad39ea49948962c35588d22d95 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3c5d7bf81f34c42ab13b59ec1fbdca4 : SUCCESS in 7m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e55e92d225b6431bb86fa53613e939ee : SUCCESS in 3m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/83519cc0ce2b47a1923263dd06b29da3 : SUCCESS in 4m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/da979041dbb94eaab4aee004394a2a4c : SUCCESS in 10m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a1bd921f005242f4a2c15b591b5d79d2 : SUCCESS in 14m 23s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/21fb1132337042e9ae10b267a4d101f6 : SUCCESS in 46m 56s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/640acb97601540d3888983ea0e8d0d96 : SUCCESS in 47m 12s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d245dd81bec8424c8c93d9b04b2056fb : SUCCESS in 48m 52s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ca4c537d190d4b2d9badc5242846cddf : SUCCESS in 53m 51s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/85a1c0f8deee4fc9b72935713d7ecfff : SUCCESS in 19m 10s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/d254bde15b8c48b8a6d546999eeeefb1 : SUCCESS in 16m 39s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b6ec402b00714b57a26b4d91be3867e0 : SUCCESS in 16m 36s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0dfbfbc99e614cfc8b233f4e48356efb : SUCCESS in 1h 21m 25s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e9f88c8ecb904c5cb9dcaf9970cbb56d : SUCCESS in 16m 21s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/785b10eee9b84b8e8b602bdff9ae5ad5 : SUCCESS in 15m 43s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/434042ef9b3b48579308d9e606f441f2 : SUCCESS in 34m 24s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf3b81711b544cb7b305badbb6dc2999 : SUCCESS in 22m 37s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4c7c431bf5d548f78ef153af7d348c84 : SUCCESS in 12m 59s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e38987c397fb4f2fa29f5a5630e043f3 : SUCCESS in 11m 07s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8ea3a8cf456e4a33860a0cf40404a681 : SUCCESS in 12m 07s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/036cce4e5a374a4ba911c10aa6852fd2 : SUCCESS in 13m 37s","accounts_in_message":[],"_revision_number":1},{"id":"34bd386a76f01488384349f88ffa6e49bae6608b","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-10-16 08:15:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dfbfbf6a4c9ffa982a36ac114694c18ce863f83c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-10-20 16:07:32.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c9f7bf4407be1b4f073eef3bf8a32ac6d6cf3a8b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-10-23 11:04:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ecfbb09d9b2db04b79392c9d3f3520e89f93a813","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 11:04:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"80f22ae10fe4ddbc2152de701225fc7d4d848be9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 13:45:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25b841a4ea8441968aa2d30c80d0cc6e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3207f34a47394f5588c0d4e728dc83ce : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67472658bb6d4a5a9c916f0caedb561f : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9479e0d9db09402a8fb9b8e53f7e949a : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8c298eefb924e368bc91dff0e1f2aa7 : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c21a8db0484e48eca9b06d2020917614 : SUCCESS in 4m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bc1f28c3e7ab4a22bb5c56b2ad4eb4c5 : SUCCESS in 3m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2b107b36cfa54d2f8874c0642085336f : SUCCESS in 11m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bc600119b395419fa796e07a810950a2 : SUCCESS in 12m 56s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/fef7d700c68f477caa03498ab8041ca0 : SUCCESS in 53m 21s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d6bebe5408fc483c96f7f75730b1bc87 : SUCCESS in 53m 08s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c3cbb5c17de7484f9a702ef20597baef : SUCCESS in 56m 04s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/484dc1f8dac24a1f962f8af1e04b2747 : SUCCESS in 17m 46s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc94d335ea8b4f058d4e3402edf71ea7 : SUCCESS in 14m 06s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fba383b24925447f8bdd60657bd9332a : SUCCESS in 1h 36m 24s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/8efde813d2824eaa9baa4652d599627b : SUCCESS in 15m 59s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0b245557528420c82c4eb096dfe664d : SUCCESS in 15m 29s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/38c69d959d1147189b83794c07503f36 : SUCCESS in 23m 13s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f5813a00d5504bd0b67d934544f74386 : SUCCESS in 12m 09s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6fb41d25ba024b6ca531993e147ccf9d : SUCCESS in 10m 14s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c212057a0d1e43fcbc311d5efda667c2 : SUCCESS in 17m 17s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0580aac07a98482aadfea4f03734d003 : SUCCESS in 12m 32s","accounts_in_message":[],"_revision_number":1},{"id":"bc5e2997e9d863cb5565965ea15ff95d6c59105c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 13:45:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7a34bc2ca3e898e770a5000bf1528fa692af79b9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 13:46:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/669e38d03a264c1eb504373f30885bd9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/158fb9a863ba4963b4e39ce00254ea62 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c36262d932f5445aac079f75ae9c2d87 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cd0bdbdd2b9c31475cf51c2eb222d6ae19e6469f","revisions":{"cd0bdbdd2b9c31475cf51c2eb222d6ae19e6469f":{"kind":"REWORK","_number":1,"created":"2023-10-12 07:09:43.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/66/898066/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/66/898066/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/898066/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/898066/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/898066/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/66/898066/1"}}},"commit":{"parents":[{"commit":"c4a6883f1b2040c3553e0f29555264b44f15220f","subject":"Merge \"Use importlib.metadata instead of importlib_metadata\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c4a6883f1b2040c3553e0f29555264b44f15220f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-10 11:09:12.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-12 07:09:43.000000000","tz":0},"subject":"Add introspection rule to update deploy kernel location","message":"Add introspection rule to update deploy kernel location\n\nBack in the Ussuri release, we changed the Bifrost kernel location to\nuse ipa.kernel instead of ipa.vmlinuz. While this works fine for newly\ndiscovered nodes, any node added to Bifrost in Train or earlier will\nhave kept ipa.vmlinuz as its deploy kernel.\n\nThis can cause issues since upgrading Bifrost or building new deployment\nimages will update ipa.kernel, but not ipa.vmlinuz, resulting in nodes\nbooting with an old kernel and a new ramdisk.\n\nFix by adding a new rule updating the legacy kernel location and\ndocumenting how to update node information.\n\nChange-Id: I6ae44dcae78424b4638762d015cf24336444707f\n(cherry picked from commit 44af704836799e9b211e3eccfa5efa72f9713f32)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cd0bdbdd2b9c31475cf51c2eb222d6ae19e6469f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cd0bdbdd2b9c31475cf51c2eb222d6ae19e6469f"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
