)]}'
{"id":"openstack%2Fkayobe~898089","triplet_id":"openstack%2Fkayobe~master~Ia3550644e5fc3e83f792e46d42b4c35d5eec4705","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"35345":{"account":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"last_update":"2023-11-14 12:25:42.000000000","reason":"\u003cGERRIT_ACCOUNT_35345\u003e replied on the change","reason_account":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-11-08 13:40:17.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-11-08 15:35:53.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-11-08 16:15:42.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-11-08 16:15:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia3550644e5fc3e83f792e46d42b4c35d5eec4705","subject":"Bump up Ansible supported versions to 7.x/8.x","status":"MERGED","created":"2023-10-12 08:19:17.000000000","updated":"2023-11-14 12:25:42.000000000","submitted":"2023-11-08 16:15:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":35,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898089","meta_rev_id":"7db71c7e5ac2c31065b1596dfd5eb51e63ef789e","_number":898089,"virtual_id_number":898089,"owner":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-08 16:15:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2023-11-08 13:34:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2023-11-08 13:40:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-11-08 15:35:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-08 10:06:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2023-11-08 13:40:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2023-11-08 14:12:01.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"}],"CC":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-12 10:58:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-20 16:10:36.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-10-25 15:48:30.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2023-11-03 14:11:14.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2023-11-06 15:57:29.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2023-11-08 10:06:36.000000000","updated_by":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"reviewer":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"state":"REVIEWER"},{"updated":"2023-11-08 13:34:14.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2023-11-08 13:40:17.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"d73322676a45cfd43fcfd8131e940ea5946c9009","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2023-10-12 08:19:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d8a65105e14dcfcdd51194c00771893e2fc0ccee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2023-10-12 08:23:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8790d757366683e35f857e3e8eec3d6803592d7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 10:58:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/64e45013580347d28718fb7eb17d5d3f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4812d3bb52a94973a78cf2039e8a0ca3 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e4a2e5991064d488352f6fb23cbec0c : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da5cee81a733410283f420b0b9d60d8a : FAILURE in 2m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5cf3f84f3d1d42049b3edd176fe55608 : SUCCESS in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/98b295101e8843d9a7ca2cb4217fc71f : SUCCESS in 4m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/af4b7b43d7c44093be75c93c875d2027 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8657898af7694eb4a0ee6928bf4ea3d8 : SUCCESS in 8m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0d82b40d64e4efe8de0bbde8b54df11 : SUCCESS in 3m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9e3ecbc605454c81a13173fb91203343 : FAILURE in 4m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/578873c99fdd447382d5ec047e8ed868 : SUCCESS in 10m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/636ac26364c343c4af73daf2b7a281ce : FAILURE in 4m 50s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/af4206026271462d9288d368e84bf21c : FAILURE in 40m 19s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/f9f2a4f0e9554d2eb59fcc3b8ca99c80 : FAILURE in 43m 16s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a72b30ff18c2494697c0ce5195daf785 : FAILURE in 9m 21s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/640dfd0a5fae4f9b84557465862cabe3 : FAILURE in 43m 34s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a1216e9d87e947028f4cb8ff340b70de : SUCCESS in 15m 08s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/4c32ff345cc4411c9c567668c11c1f72 : SUCCESS in 16m 20s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f8ed7642721049a7bf04f1f509ea4cd2 : FAILURE in 9m 05s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d62c91110f884e4095d3d04fba0b7575 : SUCCESS in 2h 33m 58s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2bdec7822d4a4f99be7acfb6234a8ead : SUCCESS in 2h 14m 42s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a143d26e6cd54e4ab889dc4ca598c232 : FAILURE in 16m 09s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6c8410bed7546aaa046bdd22842878f : FAILURE in 9m 10s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/afed59e235b14fa4b9f1c6faa1b1eb8e : FAILURE in 18m 05s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5493d8aedbd442a49095f7fcf5e88461 : SUCCESS in 32m 14s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e972cf00db804b18a6e9fa373222173a : SUCCESS in 26m 36s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/25fbc956256d49e28e3c7593eed9fe12 : SUCCESS in 12m 35s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c5d5e422072d4946859b496b643c2409 : FAILURE in 8m 50s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7fdd5727891840f1b9a09b76c5e27182 : SUCCESS in 15m 41s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fde08ebb9a8846128caf6f52c0722b8d : FAILURE in 10m 14s","accounts_in_message":[],"_revision_number":2},{"id":"a445f78e4e3aca54f5b302d1baf673d889581ec7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2023-10-13 10:08:33.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"0efdfca18cdcbd767456e75bb4ad17389f76348b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-13 12:07:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/864399bb877b4890a532f3ff3e695e7c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1e8c779d65b46cbb3ca54b6851d6162 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09bd33586a964d41bc8b22aa82df6c3d : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5009d7c2bd4b48cc819b5602bdf5e6e5 : FAILURE in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/52191795db2f4ee6a3259e65456391d3 : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4411c886c26e4537801c292f5cae7985 : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d3e0a40d1f5e4d6d9529490c043b8983 : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c35678188954485b10ed0f9d5034698 : SUCCESS in 8m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3ce81cbae38420dac74a998aec97a97 : SUCCESS in 3m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4b46c293bfb64f049b88684230caeeed : FAILURE in 5m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/28c628eb13414da9b5509552e603d29f : SUCCESS in 11m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1dfec45c3a23467fba58134171891b12 : FAILURE in 5m 15s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/3033a7051f764f9d889141e3d6ac0a52 : FAILURE in 43m 52s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/49bdf34670744c8ab943acb17c3477cc : FAILURE in 47m 46s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5de3adce2cc846e0a2fc4ac3e0762282 : FAILURE in 8m 45s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ebebcfb772894dd4ac41b07df6a4768b : FAILURE in 51m 44s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/37f5e0d68c564c148faea8ae7e1ee52d : SUCCESS in 17m 57s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/038b5eb5971b45a3b354eb8ff25683b0 : SUCCESS in 15m 19s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5bfca7e3b372462d92e8cd976a87fdc2 : FAILURE in 8m 41s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/77a6e6811aca49338326e67ad5785006 : FAILURE in 1h 54m 15s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eaad879861d34645a7c73a5c721c2b41 : FAILURE in 1h 28m 57s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/89c05f612a484536825d6ca4449c9a3a : SUCCESS in 17m 20s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f0d8a41dc2f24838ab095ddb9fd92aac : FAILURE in 10m 10s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/c09285851f014e44b09c191aee9794f2 : SUCCESS in 33m 00s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/402429666af840e0a657bf077c03063c : FAILURE in 31m 55s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e892d2c1f5374ea2b6c8d20754151e67 : FAILURE in 25m 52s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6989ad65c9f34b56a0da5e2640692d1c : SUCCESS in 13m 25s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78ad038650154f4098bee43e920198a1 : FAILURE in 7m 28s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d4be30a8fa564d9fbe6b67ee1599a57c : SUCCESS in 19m 21s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e050c3f68164a6694d8138ead873511 : FAILURE in 8m 18s","accounts_in_message":[],"_revision_number":3},{"id":"8d4059f31e63a1dee6dc508b5cad5d7f22689b6b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-10-25 15:48:30.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"5b8eaadde3d878695ec25a31b62fe6344f099278","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2023-10-25 15:53:38.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"9c1cce35293298ac3573b6af9db9e4bbf529fb23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2023-10-28 09:03:21.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"d5feec3a5846ff50b341ff1987a3d9da0815ed28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2023-10-28 09:25:34.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"bac633c5ef7e0728aaf5d4956875518f57c8a166","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-28 10:55:26.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/54a2e1a8ac3e46288900dca1d013590b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28e13c63c3c346f6af762058123efc65 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cab5fb7d52e0428181b5203ae76326c6 : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cfd1be1ae4724641a674e9dc85ea7ebc : SUCCESS in 2m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ff1c2d88f7e740c58036491cfa7f871c : FAILURE in 4m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/291e0bf1d391499ca21b9456d096f159 : SUCCESS in 10m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/54761fdf8ce34a66bd05d51e14b70e50 : FAILURE in 4m 49s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0f315b38959c453ab953efaf1e8513bc : SUCCESS in 56m 55s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/205261d8d4b341eba80be760a4b56479 : SUCCESS in 49m 01s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/72cf65e71f114a9ebe06089441937e3d : FAILURE in 7m 09s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8fb83213f0824b73a2c6ad3e74815b0a : SUCCESS in 59m 41s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0bc8dcb5f90c46af8a0a7436e9df1a1b : SUCCESS in 12m 28s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/73ab3f5e67c34b7597d16ddc54375185 : SUCCESS in 12m 18s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7275809693084a7bbc6840db2d312f72 : FAILURE in 7m 59s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3a4cb5406ee640daa221abafdc4e7dc4 : FAILURE in 1h 26m 25s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9300d8b41c0c4469ac2a0793a65f6534 : FAILURE in 1h 11m 31s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7389d183abe8485a9789348aa33a2949 : SUCCESS in 15m 43s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa447ca7ca464946af5ae7e7ad88fefb : FAILURE in 7m 22s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/15ad34e26a7c4e71978d2618850b4bf8 : SUCCESS in 29m 11s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c9f1f56ea48648df915cd9cbbf55cbfe : FAILURE in 26m 17s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c9ed1b9b1434f9baa485426f17ccdac : FAILURE in 19m 55s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2e31124fec4744c1b0ffb523da55691c : SUCCESS in 12m 05s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb7306c197fe4949a226421b6f688fda : FAILURE in 7m 05s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/817c677b80a84332bc7839826dd94afb : SUCCESS in 16m 02s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/20b18e5947cf4194803e2dfd5c3f2bc9 : FAILURE in 7m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0a44698553a84f749ef2b73c16d00e42 : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9af7ee9f2eb446e69bfd85591dcd1bb2 : SUCCESS in 3m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/99aef909c09c42f49ce698e69e7fabe3 : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":5},{"id":"9d4ab76ee3a3baed604f71b7b748c7e6982bf357","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2023-10-30 10:28:20.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"2047dff54453aebacf9652c442d3ad8827c208b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2023-10-30 10:32:41.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"7c55f76691dc4b0a103c8acd9edb03dca9b7a31f","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2023-10-30 10:33:17.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"1cf425a9f1e45ee0e1dcb0d099843debe3d777a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2023-10-30 10:34:04.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"42193d95f53c4d9b38309d3bfea0999bd798c6b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2023-10-30 10:37:09.000000000","message":"Patch Set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"5094a4b1749a2d17f306e417ed759f862055a596","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2023-10-30 10:40:05.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"f0605c2cd8c09f664e4364ca86ddf2496cb68319","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2023-10-30 11:24:57.000000000","message":"Patch Set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"7bacf1ef9db171aa2c4bb32f3b6b8a6c261b93e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 13:30:53.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/847bdde36e9249c8a5ce51aa0ec6aa09\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5ccb6a30d304cf8b3c07730b5e49e62 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a97d1fe76ddb4e5f90e8b3ad5783e291 : SUCCESS in 8m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d09d21650a964e169773f9e92ad92d66 : SUCCESS in 4m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e34b1548c7e34afeb44803a50f1150b1 : FAILURE in 4m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d87d8b0f60d74955827d35a2ada4ac51 : SUCCESS in 10m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/386adfec352e4ba79fbfab76076bd4fe : FAILURE in 4m 36s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/98d28ed78e894365ad68de6d81477607 : FAILURE in 36m 21s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/56925170a97a4f58846c7cad73b66903 : FAILURE in 37m 22s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cfababdfe03d4bcf9f0403eef2b143a9 : FAILURE in 7m 01s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/31c08b5d4e1f4b83a025071dd238ce68 : SUCCESS in 58m 01s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/77aa34abe7f2479f8c93c1b42918e742 : SUCCESS in 15m 32s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/79d3512380194e41a96ef27054386eb5 : SUCCESS in 15m 32s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8c80b579cf646b88846cf9190c27d51 : FAILURE in 7m 42s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d077fb7c7a8a468d8674f3e687beae57 : FAILURE in 1h 42m 32s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a2c9cfa4c37f420dab461b351dd4bbe7 : FAILURE in 1h 26m 01s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/da7b18ea89ff40f1b033a6e5519c689e : SUCCESS in 14m 51s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/68b91472bec148b982996c2bd9230181 : FAILURE in 7m 02s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/3f9579f40c9e4392ba8c922a9cc72d22 : SUCCESS in 32m 48s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7e844d7289634d29a4938fbde10ddb66 : FAILURE in 31m 20s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bfdddb665c0145f4858d8c119b2a9606 : FAILURE in 23m 06s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a473ae82309b489daf89e681dfade085 : SUCCESS in 11m 49s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63863cae5870409a8310190f9a3869eb : FAILURE in 9m 11s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c03e14779eb04cbab1440054a173a93c : SUCCESS in 18m 33s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/608d33ab3c7f475eabe33188920a4785 : FAILURE in 9m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/11a015d10ff0417e824528f0b9c68c9b : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/06826da72b4b4e90b3b73637736fd64d : SUCCESS in 3m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b123b2229d8849b9ac007ea18788434d : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":11},{"id":"62d04d65c07ce3af04f3e839850d16997df6ad5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2023-10-30 14:30:17.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"3aa096a1963c7436a5e42202a9077993324749cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 18:02:59.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/861b76ddb14748ce84ff5db676890410\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/829cfccf15764a76a638216bb085b7ca : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00d1c733feab48028fbea43c50ce8035 : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7f9f3210b434681b9300eb2c5462d77 : SUCCESS in 3m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/835f1514722c4955b6397531876b7c2f : FAILURE in 4m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f1dd8f4ee01a4e4b859758ac41a7134a : SUCCESS in 9m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0a0e903a004c4d1c8ec9fdd75037b5c9 : FAILURE in 6m 32s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/48c224d31bf64ceebcc829e5a978438c : FAILURE in 35m 16s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/aa8aa69265fd418da41b07bf6dea3af4 : FAILURE in 40m 48s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5307588d86764ac28dd4e20aa3b02c03 : FAILURE in 31m 10s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d6a7c88a6a51480482f19b2c7fd061a3 : SUCCESS in 1h 04m 32s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/20bf3e6d8b2243ce8a8bc15f0e8346aa : SUCCESS in 15m 01s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/766bd0f3ef9441d98a3b772254a691e4 : SUCCESS in 15m 43s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8192af6874bb43c891b27fc00f4c21fe : SUCCESS in 14m 40s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d3650e971538446a9744c1e48459a857 : FAILURE in 1h 27m 30s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/545a503e8bf94a19a091f1070e1341f9 : FAILURE in 1h 35m 49s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2a416a553e354da0b9ac2525ed2a72ab : SUCCESS in 17m 09s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a0e70fdb99d4cc79a366ac0ba0ecb4c : SUCCESS in 12m 13s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/7785341827304116b6ed255654c779c0 : SUCCESS in 29m 59s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c25797faa45e4115879abf4fc2bc7276 : FAILURE in 33m 37s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b822890f2a8e476687f0e7a753c5e699 : FAILURE in 11m 36s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a8886a8fd9ee4ed29b30a32416f93225 : SUCCESS in 12m 13s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f4b63c958b4484bbdc19251d25f42e8 : SUCCESS in 12m 00s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/800c5412ded44378b9f593a7b5d66aa7 : SUCCESS in 17m 04s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/933df8398c2a48adb1c2e089ddc3e363 : SUCCESS in 13m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8fb153518b1b4351b7b57fa0d3ebcc7f : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bbdedd685a884874acfddbe74e20ffcc : SUCCESS in 4m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cd099243ef624463afa55a39a067fdad : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":12},{"id":"edfe6893f01e6c5a359986648a2760c3e8bcefc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2023-10-30 19:40:11.000000000","message":"Patch Set 13: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"4c65fdbdf0b828721566d44adda82781f87bd1e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 21:10:00.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e7c14de4b0774427bdf2828b2360c0e0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a23d57d245464f48a052689f99dfb48a : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6202b13057de4df583b798b367c5b9ed : SUCCESS in 7m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a74e5114f97541cdbbb839cbedc3ec37 : SUCCESS in 4m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/50604c285f1c48fa9b97c0dfccc774f7 : FAILURE in 5m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0bcb8de66091447bb6461aa72ef90258 : SUCCESS in 11m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f18b03552a3e4da59b0c1e35feb75d9d : FAILURE in 4m 47s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f1ae07c70f454f95bccf74ecdddc90b9 : FAILURE in 39m 44s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/e3e11ed0651d418a9640876cf7d013c6 : FAILURE in 38m 35s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/80159dfd68074b2cb22a62852aa65a41 : FAILURE in 34m 55s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d072bf2859ad4a82bcf39741b9c26741 : SUCCESS in 55m 58s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/fafdb39a6e7745b69a432ebe45430772 : SUCCESS in 13m 50s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/6f48376b6f5c44d8b367e9dd7d090a44 : SUCCESS in 14m 01s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad8f6dea5d26438ebae97e6b2360fcde : SUCCESS in 12m 00s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/84a85b332fcd4a6ca45270b1b18af868 : FAILURE in 1h 24m 52s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/72fd72e73f7c49f9904e4f8be4186202 : FAILURE in 1h 17m 13s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c0f39dcb8c5f405dbb735af15ec8c570 : SUCCESS in 13m 41s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e8721b2333df4b65ac1ac6b85e618d30 : SUCCESS in 13m 00s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/3d1151857a9e43c381fc1b8531adc4fa : SUCCESS in 27m 40s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d6431dfe29ba42bb8b6951c3fa346050 : FAILURE in 24m 36s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c2387215137c45f096cf255dfbcaa948 : FAILURE in 23m 06s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/710ede6f186049e4aacf856db864b80b : SUCCESS in 13m 13s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cbfb47a351604986b63475c90b42b56a : SUCCESS in 10m 17s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/97206c8fc57e412aa6d274d33c1aae10 : SUCCESS in 17m 18s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c7aae0274404c90aa1afcb594ce56a7 : SUCCESS in 14m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/556f477adfd24d0c866a2f4cc5861e2a : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/95d8ea540dfc48a68c4823f60849beeb : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/808ed622cba1438bbe960113660c1c16 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":13},{"id":"9d590f154dc6e1ba4a8e2468437a6638a56b2bf8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2023-10-31 15:16:34.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"2c42a2deb88dcb804c0640f40ad7e300a00a562a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 17:02:09.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/262c6339c42e4888a4296e6abf30ff1b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2e2c3c817914a60847757ee4d77b488 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aeb0c5b0d2d2458ebd3baf925bd56aed : SUCCESS in 7m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f01787502a2445d8b59107cdeda03771 : SUCCESS in 3m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5874f2558e5c488b8c49cfd721520087 : SUCCESS in 5m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/797efe8a060e4a72b849d254dc69f471 : SUCCESS in 13m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0aea2d44a18946beb9b3a4aa48315b66 : FAILURE in 6m 08s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2c49cf12dd064481a6035a4c040c4957 : FAILURE in 35m 29s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/1509f0357d014881b655db8fbeddc659 : FAILURE in 37m 57s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bcf4458f05a0447cbf72e67bd214668a : FAILURE in 31m 04s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/1dfaa5e8ec934a81b76d5fa74f92681a : SUCCESS in 58m 23s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/346952d9d36b44a5a0db4cdac121deff : SUCCESS in 15m 51s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/058e4d1b97a24a3e92a5d49f62fc6019 : SUCCESS in 16m 17s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47e4d27eaebd4e299d94fd090132d5b3 : SUCCESS in 12m 42s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5588255610334f67b3feb73ece26ea44 : FAILURE in 1h 34m 38s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/612ae23f03034cb685277a6a34dcbe4b : FAILURE in 1h 14m 31s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e4f8e1789b89489d933cfa1c81dc8b30 : SUCCESS in 15m 06s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e0b07f5849c54950ad49be00b919618a : SUCCESS in 12m 36s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f3007a626d6e4123a303c8e7bb7fd5f9 : SUCCESS in 33m 09s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3b28b314dcae4793a491692a2745fc29 : FAILURE in 27m 59s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1f6aa77245d54d958c201da3f1312bec : FAILURE in 28m 15s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/83c5b927ad354e95995e26a2d8e02e33 : SUCCESS in 11m 52s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/251c1f0f61b44c4199704e5c9b8d9bcf : SUCCESS in 10m 00s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5e10927a2b1f41f18762cf634720b989 : SUCCESS in 16m 34s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0d9199c9607842e98e6197385db15768 : SUCCESS in 14m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e984766f50945709b5015c0706abff5 : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ad1b2227e0714315ab82de04f9b35570 : SUCCESS in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f5f68b6b24fa4044a9d20292a09879e3 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":14},{"id":"2d14685ac1248e4b5e03cce24d29cf6287e15dc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2023-11-02 12:12:30.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"df805e38949d1f231eed73a1a37015abbe589857","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 14:15:51.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/48cb6823956e4e2b950677024fa9e321\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5eef26b24f8427aaa6fb460da98056d : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3acce24117d14bfab968fc2c770d9ba7 : SUCCESS in 7m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b1dbd5176d24a7698e97b1468c995ff : SUCCESS in 3m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/522e296cd7554f3d892d28d461263362 : SUCCESS in 4m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6a707146d2f54bb7acac98060b1aa23b : SUCCESS in 11m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/37fe734067e04b9386d28a0e6983bf10 : SUCCESS in 10m 38s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/7c17ff507dae40178e8f0cbbf5afd8b7 : FAILURE in 45m 13s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/e62fc40b4fa74b9d9a40978737100b61 : FAILURE in 39m 20s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b75f37cc5e3c42e79acd187232d777eb : FAILURE in 35m 16s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/039f70eb3e6948ecae923b833b2bc2b9 : SUCCESS in 1h 09m 34s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b9ebb306a8b042389f8f60bd5414780c : SUCCESS in 16m 28s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/ada574785d26428eaa1105731b651cc3 : SUCCESS in 17m 08s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/08b03bcc00624026b42d1ff168b3654b : SUCCESS in 16m 48s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/21bf78c6aea44b3bb918309766974f26 : FAILURE in 1h 57m 36s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ccdd9294145844929326cb50ab70aa09 : FAILURE in 1h 16m 02s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/42773f92fe1449ce9696716240f88010 : SUCCESS in 15m 59s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/13f0aa9d73974724ad35ea943eeffbf1 : SUCCESS in 17m 12s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/0b4b3712692d40b9accbbad9725295c1 : SUCCESS in 36m 42s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8e4ec03434a849789d915a9d87ad1cf0 : FAILURE in 29m 36s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea27057e03724b489b4261f6039606d2 : FAILURE in 26m 05s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ddf22383d96e4ff4bf1d987b102c1af0 : SUCCESS in 13m 15s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c047f365e31e4020b783891c889db4db : SUCCESS in 14m 16s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/706e8bcfac674c169da082997c0e3e8f : SUCCESS in 20m 09s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f38f3e85a5948ecbcb123bf9aebe4cb : SUCCESS in 15m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a83639a2108442c99a9d0c99cdd0ac1a : SUCCESS in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/480b5f40adf3400e89149f7584504c20 : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d80adc8af52b41ce8181c8069a5bf003 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":15},{"id":"8d84487cca299c357fde6dde5f4bbf20b5b06c25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2023-11-02 16:41:36.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"45fd70ab461ed4344f58897942800da3a329da0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2023-11-02 16:43:36.000000000","message":"Patch Set 17: Commit message was updated.","accounts_in_message":[],"_revision_number":17},{"id":"ad0bedcf384d6b2c6d8d1e9ad26c3243c569cba1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2023-11-02 16:46:38.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"ece860fc3608066bf162845e2ac479c3e633b35a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 18:11:33.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/712ec0c51c03412f88911c5cdf7d521b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7fc06d0add834aecb9e8e4687b4de08b : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ca8c37fcf8f48ef9172900958c7b96a : SUCCESS in 6m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bae472dd5c454854bf270ca335de6d4c : SUCCESS in 3m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1a0d1c975a824ecdb7b42ebc03e8c8ff : SUCCESS in 4m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ced2a892573849d6aa1a04cc29e09374 : SUCCESS in 12m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/607fa70c36744097a848f607285d48a2 : SUCCESS in 10m 34s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/7ac6d7e9065942d8b28d203e3ecc8a7e : SUCCESS in 48m 26s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/b453d3d419a849ad9ecda18272796882 : SUCCESS in 46m 12s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b2cbeedbb9314fc68660cb6b1e9d45e2 : SUCCESS in 1h 02m 25s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3cfe1d7a69004ef799c149fcac96d3f5 : SUCCESS in 1h 01m 14s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5e88344840e94457aa1a20d45c52c946 : SUCCESS in 15m 58s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/a27263a26de749e0981e78ed617508fd : SUCCESS in 13m 43s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c27b3481a82415e93c0a42cfeaa8a9b : SUCCESS in 12m 49s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/512b5e7457994200a2c6b7678c2ccee3 : FAILURE in 1h 22m 18s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e2873d3fc7d044eb8bc3a12b319d0b1c : FAILURE in 1h 07m 37s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/36f7f77232b847d58525bdce511e44da : SUCCESS in 17m 51s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c0fde5a48ae4ad5b9f220b76992e36f : SUCCESS in 14m 37s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/8c838ca3a5f1486d83ae283c6d9f8874 : SUCCESS in 31m 33s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/14536fd66b5f477395cdf25e759ea80f : FAILURE in 27m 22s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/afac1788ee9a4e688424901d40a48ea3 : FAILURE in 19m 43s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6f4b3b0252cc4926b4a2f4d571e63092 : SUCCESS in 11m 43s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99dcb6d81d724a62bcf42002457994e4 : SUCCESS in 10m 33s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6f1240eb62dc4f9d8b542b07ee7d02ad : SUCCESS in 17m 37s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/35e43cafa2bf407bb08b57dbcae57ae3 : SUCCESS in 15m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5cb1fa00651475399ef8c13524d0422 : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9fd571e8e4e54cc993c6f1cdc3bf7b27 : SUCCESS in 4m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6b12c493f8ef4786b56672396175f021 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":18},{"id":"120c5d461db4c19b99a73dcd1899d7610e5a44ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2023-11-03 08:52:54.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"fc47f1a1949180e28169010a0235292a845ab103","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 10:46:58.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0eee7f56c3124789a05df6205b3ab76e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bfddedbe525f43ea9e4810f3ecba645b : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c01bb0e186044f7b17a1eb19e294e68 : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe648975e95d47ef9093b6e98d13e47f : SUCCESS in 4m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dfbfc2fe9c3449a1979706804f39a706 : SUCCESS in 4m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/96b7491e65c34cf8a6278a2cac8e05c9 : SUCCESS in 11m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0819abbe1fad48aea71ca80280919485 : SUCCESS in 9m 13s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/9a251fb8e816400ebf1d8eb8cb7252f0 : SUCCESS in 1h 08m 43s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/2ae608dcdcb94a7ebc5566a9573d5852 : SUCCESS in 49m 53s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c38a07e5db3e43c587ff6619b5b46f72 : SUCCESS in 53m 27s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/4982f1a311bc4c9b82e10aa466abe8f9 : SUCCESS in 1h 07m 55s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5cb12f860caf46efad5141e632f6343d : SUCCESS in 16m 31s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/631b6edad6f745d2b5d2d21c84bdd5b8 : SUCCESS in 14m 32s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ddc610afc6c44214b587243e1ac5aeb4 : SUCCESS in 11m 21s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/31e6f15ff32844c69992c4df3747ec33 : FAILURE in 1h 47m 14s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/06a821c8ff7e41348d86f24a8404b861 : FAILURE in 1h 41m 55s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f9a2c4d8f0ab4f53bbaad3371d21fa35 : SUCCESS in 17m 43s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa71962a37a44443899eb2ea9ed206ff : SUCCESS in 10m 55s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/6660e8e3f33449d7a426f761ce7db169 : SUCCESS in 31m 04s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4fc993d2b85a490db6b54f48feeeb161 : SUCCESS in 25m 28s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/17cb3bc3e6be499bb66456c9788a962e : SUCCESS in 22m 14s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/368f2fbf2cdf4ad08605b9a1f6553fb7 : SUCCESS in 11m 32s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dcb685c2333e422298e24e53ebae352d : SUCCESS in 9m 18s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ccb74c8c3fc54c4fb65368b17ac8f3e2 : SUCCESS in 16m 54s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bdd92401705546cf98ed7cf2a1d8ed39 : SUCCESS in 12m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87b5ba4c1c1143eeb6ba3b882addb037 : SUCCESS in 6m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/66ba640059c240ddaeaa154933681979 : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9a09190a4e6742978d047eabc263f145 : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":19},{"id":"33cb87ef3583295785b64c640e7ad202eec33c2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2023-11-03 10:52:08.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"4f8922ef5c5ce85adeb1f16225e8555cd8e0f15b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 13:26:17.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69a145526e714cd29447718adc504094\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8612ccbc06a145eaa56f7229f1f89274 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc9d9dd683f34de8aed36f53b6c4c9b4 : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/722ca02c53ec462abd755a310dbf47ed : SUCCESS in 3m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c44f3549e4714c5791f4378e8e4abb50 : SUCCESS in 4m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/91bc7d442b0b473a844bbb2a8a05755e : SUCCESS in 8m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cd86e9b54d81495d9201dd8fa6126a6c : SUCCESS in 10m 34s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/fc07d68a921c4ed2ab3c0c7cd1c0fd03 : SUCCESS in 1h 01m 50s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/8915476a966b46cf86c94324944af9b5 : SUCCESS in 49m 23s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69c993ebd9474bf59de2b0b3fdc4eb73 : SUCCESS in 43m 32s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/7137d8762f4e42ecb76973b463163612 : SUCCESS in 1h 06m 08s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/52169548b2914ca48fdc3ac6b7b8a92e : SUCCESS in 16m 11s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/5732d0c515ce407cbedc0e62af9a22a1 : SUCCESS in 17m 47s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/225b293886e0486681a34bc949025cf7 : SUCCESS in 15m 23s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d25d0325f1ce46db9b63c563ab89076f : SUCCESS in 2h 28m 57s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f513457e188e46acbc348a8a0a157e5e : SUCCESS in 2h 12m 15s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/764159e8069d4e12b7b4cfee5e8a3dbe : SUCCESS in 16m 32s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c7b7d6183374482b8f776113c403c796 : SUCCESS in 14m 35s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/e406a065d4814b9397f9b51123278f69 : SUCCESS in 33m 32s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/72f43a4c9be243fb834f87441d77c543 : SUCCESS in 31m 35s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/88692f0476b94c9a8fa1b713b75bee50 : SUCCESS in 25m 37s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/aff62cda03d645ee907767ab0bae75a2 : SUCCESS in 13m 51s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b794a370f7aa49fcb558eb885498b411 : SUCCESS in 11m 03s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ad6b5d874c5147de965a01b7dafa2dfe : SUCCESS in 18m 27s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ecc8110ee8924071ba8f742d9f034821 : SUCCESS in 15m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f50e20fed79b4b8fb0203c491fbed421 : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/97f180b9b5be463bad6dd28172be4e23 : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cdf2e3b803824fb28ca1002f979da405 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":20},{"id":"860e902532bb8dc5ad4a503ca6506be8e7a5c4f7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-03 14:11:14.000000000","message":"Patch Set 20:\n\n(7 comments)","accounts_in_message":[],"_revision_number":20},{"id":"f48e8a053ba9bbc7397776cd70c131ff515151ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2023-11-06 09:55:54.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"3485ef99743a5b6d7405ee7874be218fa1f6c770","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2023-11-06 09:58:47.000000000","message":"Patch Set 22: Patch Set 21 was rebased","accounts_in_message":[],"_revision_number":22},{"id":"5712eee988a4dd9eeb7b01c01f054faecd5a6b9b","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2023-11-06 09:59:32.000000000","message":"Patch Set 22:\n\n(4 comments)","accounts_in_message":[],"_revision_number":22},{"id":"2a157d7b619a007496665b7f43ecabeeb74129c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2023-11-06 10:51:34.000000000","message":"Patch Set 23: Commit message was updated.","accounts_in_message":[],"_revision_number":23},{"id":"7ab6212dcc0ffb2a2bf447cf2f26da175539c7fa","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2023-11-06 10:51:46.000000000","message":"Patch Set 23:\n\n(3 comments)","accounts_in_message":[],"_revision_number":23},{"id":"8e4cfe1713aacad7f44f35743bfcef515b727d6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 13:30:20.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/87dcfb02081a4af2a872fec35257e94f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6575bc2a84b943558b3775b74c8ec5c1 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28d73b39a5714f03b591b5ef001cf471 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4df68af11fcd4597b7505c305c02ffcf : FAILURE in 2m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/814bb0970b054613939888171749723e : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0c11a3df48746bb97ff1a059bb46258 : SUCCESS in 7m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7d25b9d8e944cc9ae53a3b1f3bb55c5 : SUCCESS in 3m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c5cd9f556ef5410b9d615d1959f8773c : SUCCESS in 4m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5bdd4a951a2848c4aaa5cd0544b3b790 : SUCCESS in 11m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a78728d7800d496498a065c283968524 : SUCCESS in 9m 49s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/5c15a742019a4085bc63adc4a8310066 : SUCCESS in 1h 04m 32s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/f32f0caba42b4f2ba9252a0c2e368dc0 : SUCCESS in 1h 03m 35s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a389dab7b2c340ca8e6be881f253bf3a : SUCCESS in 48m 50s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/58ddb654cc1c47d99dc9343b8d991d6b : SUCCESS in 1h 07m 21s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c85df424b0e94297bd49b7d59b901189 : SUCCESS in 16m 18s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/40fdf6d74b6d4f6ea35d78c9a76e01ea : SUCCESS in 16m 09s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f4f8fb3688fb4d24a8659c46e59687aa : SUCCESS in 14m 51s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d5042c1d3883464187dc3ecbc26f08b8 : SUCCESS in 2h 33m 41s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f3bcb7dccc43442e97aa1db9707aec6f : SUCCESS in 2h 04m 55s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ab94988173c94c1aba88992a5706cab0 : SUCCESS in 16m 39s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/426ade1dd5d44c80970066e7025c5d45 : SUCCESS in 14m 32s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f3d2d16602ce4aebb3dbe64ba6a67c5e : SUCCESS in 33m 40s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2a9de3235d51444596b3bf8d79e0d40e : SUCCESS in 25m 54s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ed9420147064e50bb4b33154c80e2a0 : SUCCESS in 23m 20s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3f96fcc418914473aeebf4a65c3ef828 : SUCCESS in 11m 15s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a14c3d5297642f4b0074d76089a2111 : SUCCESS in 9m 17s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0daae96862a442c593582ded4111e0a3 : SUCCESS in 14m 51s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d68a5d6b66c648d29fbc3eac9f3cd9d1 : SUCCESS in 11m 38s","accounts_in_message":[],"_revision_number":23},{"id":"4406d1988a20875a3b84e95e41ac8849ca5e432e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2023-11-06 15:05:14.000000000","message":"Patch Set 24: Patch Set 23 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":24},{"id":"e614f72aa13ac8a76d2d797ac2e851c5427061ad","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-11-06 15:57:29.000000000","message":"Patch Set 24:\n\n(2 comments)","accounts_in_message":[],"_revision_number":24},{"id":"9e045a3efad72b4f396d18c0d2b538fa0cd364ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 17:38:54.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/523efaba0f864e4ea0c6e8e732c66896\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/542258ac63df4260ae16fa448853c178 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb67eb5990aa43eeba617ca189f07d58 : SUCCESS in 7m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4bbe8a20c4184b92960aef3633da230e : SUCCESS in 3m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f88a5ad7710b4cfa956b7ce44cf64764 : SUCCESS in 4m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e04cd0e100b84285a08967748ee081bc : SUCCESS in 9m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c4321391d0e844ba9158b9b6d557f8e6 : SUCCESS in 10m 44s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d57ead5744f448b7ad8c32234627c579 : SUCCESS in 1h 01m 49s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/776325159fb44ca698b7397d65133cf4 : SUCCESS in 50m 26s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/219661b194ba43d595b49d3c83a1b8cf : SUCCESS in 49m 02s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8320e976c16446dd9aba3d3319b04091 : SUCCESS in 1h 15m 13s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b031845531aa476190c23b93563f2579 : SUCCESS in 14m 08s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/5405f05821c2493292b5e842454f0bef : SUCCESS in 14m 26s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b24a6321443459eac6cd07831c2a049 : SUCCESS in 13m 24s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/64bea0c0c1d948eab9cc211e4fd40fb2 : SUCCESS in 2h 08m 03s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/61ed25e2a5c14acba0ac6e77abbeedf1 : SUCCESS in 2h 25m 04s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/85bd1a22bd394246b5400a94a8742c2c : SUCCESS in 13m 32s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93dcf39a21fa45e7aa34b39c15657221 : SUCCESS in 14m 38s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/82f5ad9e8bdb479db9f85a76d0340502 : SUCCESS in 31m 44s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c77279acdae0417f87dc1fa82589e2c6 : SUCCESS in 29m 21s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2929f949869c4c6faa04bfa4b91074f3 : SUCCESS in 27m 57s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bdd8d7043d6f4508aaf3001bf290ad27 : SUCCESS in 12m 11s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a9936e8319646a893d33ddfaa60e517 : SUCCESS in 10m 13s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e7b54c00c2084894877a3033bf893a6c : SUCCESS in 16m 59s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1ebeb802e71466bb33f7394cf6c6596 : SUCCESS in 12m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46910569893e43ac934dbcba07c70883 : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78f389f293c0432f8a029e3741427b9e : SUCCESS in 4m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/47742eeed0684423a6d4dbdea4349cf3 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":24},{"id":"8bb5cd6bbbe7e739690dd86b34448725213a3dae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2023-11-07 17:31:58.000000000","message":"Patch Set 25: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":25},{"id":"a33f1a2c6db8a9d3ca062fa15a36cad35f023db3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2023-11-07 17:33:07.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"127375dc9a2df0f035ca1a23dcbdc40b81cb4cc9","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2023-11-07 17:33:25.000000000","message":"Patch Set 25:\n\n(2 comments)","accounts_in_message":[],"_revision_number":25},{"id":"f546ae35d10ed68cf1e112cdb6c4a7cf5695f5f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 20:32:59.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1988a5e9fa94433ba41e43a0f89a701\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c9a8af023b84bf28e4af5e8e1d8b0ce : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a95802e0ecc499784b091d1d826c25e : SUCCESS in 7m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a5398b6ca1c42808cd9b5e5ccee6a12 : SUCCESS in 3m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/12b2e8601fcc4895b2f3d8a75f4b5e71 : SUCCESS in 4m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/beb0c17fa5df44d5b07092c4173a1215 : SUCCESS in 13m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/360c91ef3e18464eba3be7df77913ad2 : SUCCESS in 10m 29s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f579466d14014154b7fe28792031d467 : SUCCESS in 1h 09m 25s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/eeed63d64e6b4c68bb721d3f58480aff : SUCCESS in 54m 25s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d607d4f4f3e0474b8dc7cddf7f054a0c : SUCCESS in 55m 35s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/25b22e548e26442aa77f059f123b2cea : SUCCESS in 1h 07m 02s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0191a615944543f48f389a1eea7bd757 : SUCCESS in 16m 58s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/efd62186cb784159a57946244dc8bfc3 : SUCCESS in 14m 47s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6bb214509de24af4a8b0ac585bc8e8c5 : SUCCESS in 12m 42s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0220a9d5e267432c8e653ef840ccbd43 : SUCCESS in 2h 50m 29s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b4955c72675f423ba8f21fb1569e24be : SUCCESS in 2h 06m 01s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/1f4065f862f34117bf4a7afa5f673d04 : SUCCESS in 17m 29s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/90753cbb85214c218f83f1f3a6229b8a : SUCCESS in 15m 42s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/010717f062ca4798bd3940ae82c52858 : SUCCESS in 35m 03s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/65fd4bb80381464eae576e084663f08e : SUCCESS in 28m 23s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07eced75ff7b4426b84889f7e88ac049 : SUCCESS in 24m 45s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/340e603970934727a89292fcac204566 : SUCCESS in 11m 06s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e1c03ffb2a641ba9d1759f3e105c681 : SUCCESS in 11m 13s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e5306ecf6754477ab40af504eb3a23a2 : SUCCESS in 15m 00s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/983113be0ea74245b7c2052c7895f4da : SUCCESS in 12m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f2f6cbf7a51248c7ae447249f9699baf : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c082e8e7be6b4b26bf4c884224852135 : SUCCESS in 4m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/61d816e1257e4c599e52c1eb2e727cee : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":26},{"id":"b51440b27dd9b537f2a1e83b19ecc65bf8a1e303","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2023-11-08 10:06:36.000000000","message":"Patch Set 26: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"efa5441e27b32ab5ffa232abbd709e8b37a9affa","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-11-08 13:34:14.000000000","message":"Patch Set 26: Code-Review+2","accounts_in_message":[],"_revision_number":26},{"id":"a1e0b183fdd26849337300bf2bae0ccb8dd763ab","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-11-08 13:37:25.000000000","message":"Patch Set 26: Review-Priority+2","accounts_in_message":[],"_revision_number":26},{"id":"985ba34aa47920945fdeb1ee767778f0c9bd1a56","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-08 13:40:17.000000000","message":"Patch Set 26: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":26},{"id":"6f28f99e2f43e2bb417e10555adf8a844154c65f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 13:40:41.000000000","message":"Patch Set 26: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":26},{"id":"4a8a00ea1bbbebdab965a546e96116a76df505c8","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-11-08 14:12:01.000000000","message":"Patch Set 26: Review-Priority+1","accounts_in_message":[],"_revision_number":26},{"id":"306cadec6aad1c7823da1ebe5e9fe0601fdb5acf","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-08 15:35:53.000000000","message":"Patch Set 26: Code-Review+1","accounts_in_message":[],"_revision_number":26},{"id":"f483f4f802cc91c727207031679779bd6b682582","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 16:15:42.000000000","message":"Patch Set 26: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45b83722390842f0b47667c4a1c42728\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5542a794f80740039c0c634a6aa456b5 : SUCCESS in 7m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15ee11adbb0646c1a983fddac155799b : SUCCESS in 3m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dafd75f94e714045854db7f54952d88d : SUCCESS in 5m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/71a55128091b4907963035f98378ab2d : SUCCESS in 12m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/37048ac0ea2e4dad92d18b5f82753fa8 : SUCCESS in 10m 54s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/c1470ae0dcca44d1b51254b09ebe3741 : SUCCESS in 56m 25s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/075508bcd89f4be1939ce5b5bd49a1ca : SUCCESS in 52m 32s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/73a341fd155c487c89a5e027b45a4690 : SUCCESS in 1h 02m 46s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ea388fbe081646bd8ac7200d8a1e407e : SUCCESS in 15m 54s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/752bc669a13a4b9f98a0435ddad2e0d0 : SUCCESS in 13m 40s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6f204b66af5a4375b77b72b5f040b8cd : SUCCESS in 2h 30m 30s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c98e6cfec27042e48cc9f7619641737b : SUCCESS in 2h 14m 03s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/8bd5e7abca6c469891420a8921c81d63 : SUCCESS in 16m 57s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/03fca87cb714436692c50fc0b3225af5 : SUCCESS in 15m 49s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/01311850cc6d4f20a7cdc168266668e2 : SUCCESS in 30m 30s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0323956510984eae882cf9c94f6bb71b : SUCCESS in 28m 02s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/174d90b905284406a0b40818c8b76e3b : SUCCESS in 13m 30s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d3c0f0202ee4f289adbfb4e91ff4789 : SUCCESS in 11m 08s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/41169615b28142a49d106b29ec26cc6f : SUCCESS in 17m 40s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/20c3e285a25b4e12ab512ff3c61f9c74 : SUCCESS in 15m 19s","accounts_in_message":[],"_revision_number":26},{"id":"5c57c0689fe23a0a20fd7d898270c76176f4050c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 16:15:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":26},{"id":"6d0160cf48c0d5e3c2aa69a3e8dbd238fb065f11","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 16:16:47.000000000","message":"Patch Set 26:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de31eed95c8c4dd2a4b532adc2ee3300\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/980579419ac44a569c985dc85dd88792 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7f4efd166564ab4800295bbd6f1765a : SUCCESS in 44s","accounts_in_message":[],"_revision_number":26},{"id":"decd05e9e90e8b3c0175c067fcd85edbbc31eace","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-10 10:33:38.000000000","message":"Patch Set 26:\n\n(3 comments)","accounts_in_message":[],"_revision_number":26},{"id":"2c784b492d48c23197d003f3f7b7fafcbbccc864","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-11-10 11:53:28.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"f3f3639da5fd3ccee8b7770e17cf8d96fa0ebfcd","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-11-10 15:27:10.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"cb1db3f9c98f21ee11a4be748c5b4595bf5a77ae","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2023-11-13 09:41:27.000000000","message":"Patch Set 26:\n\n(2 comments)","accounts_in_message":[],"_revision_number":26},{"id":"93d8201805ef252c0a7711d579a153c09cbcdacc","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2023-11-13 09:43:09.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"bae312bf26d4eed9343e555d515f387a845cbee1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-14 12:22:07.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"7db71c7e5ac2c31065b1596dfd5eb51e63ef789e","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2023-11-14 12:25:42.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26}],"current_revision_number":26,"current_revision":"80f6db34cd686bf9b5851fc421bcc41e00723e65","revisions":{"4a74aaf3a1213f74ed721f0c47c58a48a6190fac":{"kind":"REWORK","_number":1,"created":"2023-10-12 08:19:17.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/89/898089/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/898089/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/898089/1"}}},"commit":{"parents":[{"commit":"44af704836799e9b211e3eccfa5efa72f9713f32","subject":"Add introspection rule to update deploy kernel location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/44af704836799e9b211e3eccfa5efa72f9713f32"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-12 07:58:48.000000000","tz":120},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-12 08:19:16.000000000","tz":120},"subject":"Bump up Ansible supported versions to 7.x/8.x","message":"Bump up Ansible supported versions to 7.x/8.x\n\nThis change bumps up the maximum supported version of Ansible to 8.x\n(ansible-core 2.15.4) and minimum to 7.x (ansible-core 2.14.10).\n\nMin/Max values are based on\nhttps://review.opendev.org/c/openstack/kolla-ansible/+/897926.\n\nChange-Id: Ia3550644e5fc3e83f792e46d42b4c35d5eec4705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4a74aaf3a1213f74ed721f0c47c58a48a6190fac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4a74aaf3a1213f74ed721f0c47c58a48a6190fac"}]},"branch":"refs/heads/master"},"e7f8b7ed9545566db43b33019f90a3bc8c60960c":{"kind":"REWORK","_number":2,"created":"2023-10-12 08:23:25.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/89/898089/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/898089/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/898089/2"}}},"commit":{"parents":[{"commit":"44af704836799e9b211e3eccfa5efa72f9713f32","subject":"Add introspection rule to update deploy kernel location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/44af704836799e9b211e3eccfa5efa72f9713f32"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-12 07:58:48.000000000","tz":120},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-12 08:23:14.000000000","tz":120},"subject":"Bump up Ansible supported versions to 7.x/8.x","message":"Bump up Ansible supported versions to 7.x/8.x\n\nThis change bumps up the maximum supported version of Ansible to 8.x\n(ansible-core 2.15.4) and minimum to 7.x (ansible-core 2.14.10).\n\nMin/Max values are based on\nhttps://review.opendev.org/c/openstack/kolla-ansible/+/897926.\n\nChange-Id: Ia3550644e5fc3e83f792e46d42b4c35d5eec4705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e7f8b7ed9545566db43b33019f90a3bc8c60960c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e7f8b7ed9545566db43b33019f90a3bc8c60960c"}]},"branch":"refs/heads/master"},"e95fdcecf478db19de156cd70f735a5ef7bb8fa1":{"kind":"REWORK","_number":3,"created":"2023-10-13 10:08:33.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/89/898089/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/898089/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/898089/3"}}},"commit":{"parents":[{"commit":"44af704836799e9b211e3eccfa5efa72f9713f32","subject":"Add introspection rule to update deploy kernel location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/44af704836799e9b211e3eccfa5efa72f9713f32"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-12 07:58:48.000000000","tz":120},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-13 10:07:51.000000000","tz":120},"subject":"Bump up Ansible supported versions to 7.x/8.x","message":"Bump up Ansible supported versions to 7.x/8.x\n\nThis change bumps up the maximum supported version of Ansible to 8.x\n(ansible-core 2.15.4) and minimum to 7.x (ansible-core 2.14.10).\n\nMin/Max values are based on\nhttps://review.opendev.org/c/openstack/kolla-ansible/+/897926.\n\nChange-Id: Ia3550644e5fc3e83f792e46d42b4c35d5eec4705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e95fdcecf478db19de156cd70f735a5ef7bb8fa1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e95fdcecf478db19de156cd70f735a5ef7bb8fa1"}]},"branch":"refs/heads/master"},"5b6ad204b260bc3c841845176937917e7734a4f8":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-10-28 09:03:21.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/89/898089/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/898089/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/898089/4"}}},"commit":{"parents":[{"commit":"86570531d2e0153b432f08837a78b148a25d2142","subject":"CI: Drop tox-py38 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/86570531d2e0153b432f08837a78b148a25d2142"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-12 07:58:48.000000000","tz":120},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-28 09:03:21.000000000","tz":0},"subject":"Bump up Ansible supported versions to 7.x/8.x","message":"Bump up Ansible supported versions to 7.x/8.x\n\nThis change bumps up the maximum supported version of Ansible to 8.x\n(ansible-core 2.15.4) and minimum to 7.x (ansible-core 2.14.10).\n\nMin/Max values are based on\nhttps://review.opendev.org/c/openstack/kolla-ansible/+/897926.\n\nChange-Id: Ia3550644e5fc3e83f792e46d42b4c35d5eec4705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5b6ad204b260bc3c841845176937917e7734a4f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5b6ad204b260bc3c841845176937917e7734a4f8"}]},"branch":"refs/heads/master","description":"Rebase"},"f2610b9893583aff4f784a407b7084dd4bd19f4c":{"kind":"REWORK","_number":5,"created":"2023-10-28 09:25:34.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/89/898089/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/898089/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/898089/5"}}},"commit":{"parents":[{"commit":"86570531d2e0153b432f08837a78b148a25d2142","subject":"CI: Drop tox-py38 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/86570531d2e0153b432f08837a78b148a25d2142"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-12 07:58:48.000000000","tz":120},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-28 09:21:57.000000000","tz":120},"subject":"Bump up Ansible supported versions to 7.x/8.x","message":"Bump up Ansible supported versions to 7.x/8.x\n\nThis change bumps up the maximum supported version of Ansible to 8.x\n(ansible-core 2.15.4) and minimum to 7.x (ansible-core 2.14.10).\n\nMin/Max values are based on\nhttps://review.opendev.org/c/openstack/kolla-ansible/+/897926.\n\nChange-Id: Ia3550644e5fc3e83f792e46d42b4c35d5eec4705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f2610b9893583aff4f784a407b7084dd4bd19f4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f2610b9893583aff4f784a407b7084dd4bd19f4c"}]},"branch":"refs/heads/master"},"302297052f6c0d6d6f09849c0480b2ba8df04b42":{"kind":"REWORK","_number":6,"created":"2023-10-30 10:28:20.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/89/898089/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/898089/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/898089/6"}}},"commit":{"parents":[{"commit":"86570531d2e0153b432f08837a78b148a25d2142","subject":"CI: Drop tox-py38 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/86570531d2e0153b432f08837a78b148a25d2142"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-12 07:58:48.000000000","tz":120},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-30 10:28:01.000000000","tz":60},"subject":"Bump up Ansible supported versions to 7.x/8.x","message":"Bump up Ansible supported versions to 7.x/8.x\n\nThis change bumps up the maximum supported version of Ansible to 8.x\n(ansible-core 2.15.4) and minimum to 7.x (ansible-core 2.14.10).\n\nMin/Max values are based on\nhttps://review.opendev.org/c/openstack/kolla-ansible/+/897926.\n\nChange-Id: Ia3550644e5fc3e83f792e46d42b4c35d5eec4705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/302297052f6c0d6d6f09849c0480b2ba8df04b42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/302297052f6c0d6d6f09849c0480b2ba8df04b42"}]},"branch":"refs/heads/master"},"3605c6850d74f4e9f69a00f3218570d7a2badfeb":{"kind":"REWORK","_number":7,"created":"2023-10-30 10:32:41.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/89/898089/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/898089/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/898089/7"}}},"commit":{"parents":[{"commit":"86570531d2e0153b432f08837a78b148a25d2142","subject":"CI: Drop tox-py38 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/86570531d2e0153b432f08837a78b148a25d2142"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-12 07:58:48.000000000","tz":120},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-30 10:32:33.000000000","tz":60},"subject":"Bump up Ansible supported versions to 7.x/8.x","message":"Bump up Ansible supported versions to 7.x/8.x\n\nThis change bumps up the maximum supported version of Ansible to 8.x\n(ansible-core 2.15.4) and minimum to 7.x (ansible-core 2.14.10).\n\nMin/Max values are based on\nhttps://review.opendev.org/c/openstack/kolla-ansible/+/897926.\n\nChange-Id: Ia3550644e5fc3e83f792e46d42b4c35d5eec4705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3605c6850d74f4e9f69a00f3218570d7a2badfeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3605c6850d74f4e9f69a00f3218570d7a2badfeb"}]},"branch":"refs/heads/master"},"712bd57b9fb6af2fb05d00cd85b3d622e7fa5f91":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2023-10-30 10:34:04.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/89/898089/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/898089/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/898089/8"}}},"commit":{"parents":[{"commit":"86570531d2e0153b432f08837a78b148a25d2142","subject":"CI: Drop tox-py38 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/86570531d2e0153b432f08837a78b148a25d2142"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-12 07:58:48.000000000","tz":120},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-30 10:34:04.000000000","tz":0},"subject":"Bump up Ansible supported versions to 7.x/8.x","message":"Bump up Ansible supported versions to 7.x/8.x\n\nThis change bumps up the maximum supported version of Ansible to 8.x\n(ansible-core 2.15) and minimum to 7.x (ansible-core 2.14).\n\nMin/Max values are based on\nhttps://review.opendev.org/c/openstack/kolla-ansible/+/897926.\n\nChange-Id: Ia3550644e5fc3e83f792e46d42b4c35d5eec4705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/712bd57b9fb6af2fb05d00cd85b3d622e7fa5f91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/712bd57b9fb6af2fb05d00cd85b3d622e7fa5f91"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b2592684dc890e6d71251bb257f15d1d6cea8ec6":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2023-10-30 10:37:09.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/89/898089/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/898089/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/898089/9"}}},"commit":{"parents":[{"commit":"86570531d2e0153b432f08837a78b148a25d2142","subject":"CI: Drop tox-py38 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/86570531d2e0153b432f08837a78b148a25d2142"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-12 07:58:48.000000000","tz":120},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-30 10:37:09.000000000","tz":0},"subject":"Bump up Ansible supported versions to 7.x/8.x","message":"Bump up Ansible supported versions to 7.x/8.x\n\nThis change bumps up the maximum supported version of Ansible to 8.x\n(ansible-core 2.15) and minimum to 7.x (ansible-core 2.14).\n\nChange-Id: Ia3550644e5fc3e83f792e46d42b4c35d5eec4705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2592684dc890e6d71251bb257f15d1d6cea8ec6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2592684dc890e6d71251bb257f15d1d6cea8ec6"}]},"branch":"refs/heads/master","description":"Edit commit message"},"1b48ec15f16d3e9ec7e192f955acc3e9d0dcfbff":{"kind":"REWORK","_number":10,"created":"2023-10-30 10:40:05.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/89/898089/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/898089/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/898089/10"}}},"commit":{"parents":[{"commit":"86570531d2e0153b432f08837a78b148a25d2142","subject":"CI: Drop tox-py38 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/86570531d2e0153b432f08837a78b148a25d2142"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-12 07:58:48.000000000","tz":120},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-30 10:39:48.000000000","tz":60},"subject":"Bump up Ansible supported versions to 7.x/8.x","message":"Bump up Ansible supported versions to 7.x/8.x\n\nThis change bumps up the maximum supported version of Ansible to 8.x\n(ansible-core 2.15) and minimum to 7.x (ansible-core 2.14).\n\nChange-Id: Ia3550644e5fc3e83f792e46d42b4c35d5eec4705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1b48ec15f16d3e9ec7e192f955acc3e9d0dcfbff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1b48ec15f16d3e9ec7e192f955acc3e9d0dcfbff"}]},"branch":"refs/heads/master"},"1931f6e3aac01f974cdc5c4df23e8e913a12c84d":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2023-10-30 11:24:57.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/89/898089/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/898089/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/898089/11"}}},"commit":{"parents":[{"commit":"86570531d2e0153b432f08837a78b148a25d2142","subject":"CI: Drop tox-py38 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/86570531d2e0153b432f08837a78b148a25d2142"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-12 07:58:48.000000000","tz":120},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-30 11:24:57.000000000","tz":0},"subject":"Bump up Ansible supported versions to 7.x/8.x","message":"Bump up Ansible supported versions to 7.x/8.x\n\nThis change bumps up the maximum supported version of Ansible to 8.x\n(ansible-core 2.15) and minimum to 7.x (ansible-core 2.14).\n\nDepends on: https://github.com/stackhpc/ansible-role-systemd-networkd/pull/10\n\nChange-Id: Ia3550644e5fc3e83f792e46d42b4c35d5eec4705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1931f6e3aac01f974cdc5c4df23e8e913a12c84d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1931f6e3aac01f974cdc5c4df23e8e913a12c84d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"030d4fd6338c5a27188caac125a2ef75e6c6cbf4":{"kind":"REWORK","_number":12,"created":"2023-10-30 14:30:17.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/89/898089/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/898089/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/898089/12"}}},"commit":{"parents":[{"commit":"86570531d2e0153b432f08837a78b148a25d2142","subject":"CI: Drop tox-py38 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/86570531d2e0153b432f08837a78b148a25d2142"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-12 07:58:48.000000000","tz":120},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-30 14:30:04.000000000","tz":60},"subject":"Bump up Ansible supported versions to 7.x/8.x","message":"Bump up Ansible supported versions to 7.x/8.x\n\nThis change bumps up the maximum supported version of Ansible to 8.x\n(ansible-core 2.15) and minimum to 7.x (ansible-core 2.14).\n\nDepends on: https://github.com/stackhpc/ansible-role-systemd-networkd/pull/10\n\nChange-Id: Ia3550644e5fc3e83f792e46d42b4c35d5eec4705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/030d4fd6338c5a27188caac125a2ef75e6c6cbf4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/030d4fd6338c5a27188caac125a2ef75e6c6cbf4"}]},"branch":"refs/heads/master"},"4687fb966db5a69fb56f6add095ad58d51fea63a":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2023-10-30 19:40:11.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/89/898089/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/898089/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/898089/13"}}},"commit":{"parents":[{"commit":"86570531d2e0153b432f08837a78b148a25d2142","subject":"CI: Drop tox-py38 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/86570531d2e0153b432f08837a78b148a25d2142"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-12 07:58:48.000000000","tz":120},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-30 19:40:11.000000000","tz":0},"subject":"Bump up Ansible supported versions to 7.x/8.x","message":"Bump up Ansible supported versions to 7.x/8.x\n\nThis change bumps up the maximum supported version of Ansible to 8.x\n(ansible-core 2.15) and minimum to 7.x (ansible-core 2.14).\n\nDepends on:\nhttps://github.com/stackhpc/ansible-role-systemd-networkd/pull/10\n\nhttps://github.com/stackhpc/ansible-role-dell-powerconnect-switch/pull/13\n\nhttps://github.com/stackhpc/ansible-role-mellanox-switch/pull/8\n\nChange-Id: Ia3550644e5fc3e83f792e46d42b4c35d5eec4705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4687fb966db5a69fb56f6add095ad58d51fea63a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4687fb966db5a69fb56f6add095ad58d51fea63a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b1a976cb0b464ba7d31099edd8ed4b84fa677a40":{"kind":"REWORK","_number":14,"created":"2023-10-31 15:16:34.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/89/898089/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/898089/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/898089/14"}}},"commit":{"parents":[{"commit":"86570531d2e0153b432f08837a78b148a25d2142","subject":"CI: Drop tox-py38 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/86570531d2e0153b432f08837a78b148a25d2142"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-12 07:58:48.000000000","tz":120},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-31 15:16:19.000000000","tz":60},"subject":"Bump up Ansible supported versions to 7.x/8.x","message":"Bump up Ansible supported versions to 7.x/8.x\n\nThis change bumps up the maximum supported version of Ansible to 8.x\n(ansible-core 2.15) and minimum to 7.x (ansible-core 2.14).\n\nDepends on:\nhttps://github.com/stackhpc/ansible-role-systemd-networkd/pull/10\n\nhttps://github.com/stackhpc/ansible-role-dell-powerconnect-switch/pull/13\n\nhttps://github.com/stackhpc/ansible-role-mellanox-switch/pull/8\n\nChange-Id: Ia3550644e5fc3e83f792e46d42b4c35d5eec4705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b1a976cb0b464ba7d31099edd8ed4b84fa677a40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b1a976cb0b464ba7d31099edd8ed4b84fa677a40"}]},"branch":"refs/heads/master"},"202ac987e3733191e0813461da26963987fcc266":{"kind":"REWORK","_number":15,"created":"2023-11-02 12:12:30.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/89/898089/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/898089/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/898089/15"}}},"commit":{"parents":[{"commit":"86570531d2e0153b432f08837a78b148a25d2142","subject":"CI: Drop tox-py38 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/86570531d2e0153b432f08837a78b148a25d2142"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-12 07:58:48.000000000","tz":120},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-11-02 12:12:05.000000000","tz":60},"subject":"Bump up Ansible supported versions to 7.x/8.x","message":"Bump up Ansible supported versions to 7.x/8.x\n\nThis change bumps up the maximum supported version of Ansible to 8.x\n(ansible-core 2.15) and minimum to 7.x (ansible-core 2.14).\n\nDepends on:\nhttps://github.com/stackhpc/ansible-role-systemd-networkd/pull/10\n\nhttps://github.com/stackhpc/ansible-role-dell-powerconnect-switch/pull/13\n\nhttps://github.com/stackhpc/ansible-role-mellanox-switch/pull/8\n\nChange-Id: Ia3550644e5fc3e83f792e46d42b4c35d5eec4705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/202ac987e3733191e0813461da26963987fcc266"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/202ac987e3733191e0813461da26963987fcc266"}]},"branch":"refs/heads/master"},"88e5c85362916327ae6b9eb08e0f5f7bb4c61ee4":{"kind":"REWORK","_number":16,"created":"2023-11-02 16:41:36.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/89/898089/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/898089/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/898089/16"}}},"commit":{"parents":[{"commit":"86570531d2e0153b432f08837a78b148a25d2142","subject":"CI: Drop tox-py38 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/86570531d2e0153b432f08837a78b148a25d2142"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-12 07:58:48.000000000","tz":120},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-11-02 16:41:18.000000000","tz":60},"subject":"Bump up Ansible supported versions to 7.x/8.x","message":"Bump up Ansible supported versions to 7.x/8.x\n\nThis change bumps up the maximum supported version of Ansible to 8.x\n(ansible-core 2.15) and minimum to 7.x (ansible-core 2.14).\n\nDepends on:\nhttps://github.com/stackhpc/ansible-role-systemd-networkd/pull/10\n\nhttps://github.com/stackhpc/ansible-role-dell-powerconnect-switch/pull/13\n\nhttps://github.com/stackhpc/ansible-role-mellanox-switch/pull/8\n\nChange-Id: Ia3550644e5fc3e83f792e46d42b4c35d5eec4705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/88e5c85362916327ae6b9eb08e0f5f7bb4c61ee4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/88e5c85362916327ae6b9eb08e0f5f7bb4c61ee4"}]},"branch":"refs/heads/master"},"5ea178b3c6fc4cbf61c3dd13f38176da764bc9c9":{"kind":"NO_CODE_CHANGE","_number":17,"created":"2023-11-02 16:43:36.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/89/898089/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/898089/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/898089/17"}}},"commit":{"parents":[{"commit":"86570531d2e0153b432f08837a78b148a25d2142","subject":"CI: Drop tox-py38 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/86570531d2e0153b432f08837a78b148a25d2142"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-12 07:58:48.000000000","tz":120},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-11-02 16:43:36.000000000","tz":0},"subject":"Bump up Ansible supported versions to 7.x/8.x","message":"Bump up Ansible supported versions to 7.x/8.x\n\nThis change bumps up the maximum supported version of Ansible to 8.x\n(ansible-core 2.15.x) and minimum to 7.x (ansible-core 2.14.x).\n\nDepends on:\nhttps://github.com/stackhpc/ansible-role-systemd-networkd/pull/10\n\nhttps://github.com/stackhpc/ansible-role-dell-powerconnect-switch/pull/13\n\nhttps://github.com/stackhpc/ansible-role-mellanox-switch/pull/8\n\nChange-Id: Ia3550644e5fc3e83f792e46d42b4c35d5eec4705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ea178b3c6fc4cbf61c3dd13f38176da764bc9c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ea178b3c6fc4cbf61c3dd13f38176da764bc9c9"}]},"branch":"refs/heads/master","description":"Edit commit message"},"29f35f02a70c2719d3fc13449ce585cc0a6e2028":{"kind":"REWORK","_number":18,"created":"2023-11-02 16:46:38.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/89/898089/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/898089/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/898089/18"}}},"commit":{"parents":[{"commit":"86570531d2e0153b432f08837a78b148a25d2142","subject":"CI: Drop tox-py38 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/86570531d2e0153b432f08837a78b148a25d2142"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-12 07:58:48.000000000","tz":120},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-11-02 16:46:28.000000000","tz":60},"subject":"Bump up Ansible supported versions to 7.x/8.x","message":"Bump up Ansible supported versions to 7.x/8.x\n\nThis change bumps up the maximum supported version of Ansible to 8.x\n(ansible-core 2.15.x) and minimum to 7.x (ansible-core 2.14.x).\n\nDepends on:\nhttps://github.com/stackhpc/ansible-role-systemd-networkd/pull/10\n\nhttps://github.com/stackhpc/ansible-role-dell-powerconnect-switch/pull/13\n\nhttps://github.com/stackhpc/ansible-role-mellanox-switch/pull/8\n\nChange-Id: Ia3550644e5fc3e83f792e46d42b4c35d5eec4705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/29f35f02a70c2719d3fc13449ce585cc0a6e2028"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/29f35f02a70c2719d3fc13449ce585cc0a6e2028"}]},"branch":"refs/heads/master"},"18540e547f686aa294ce0ac0ea0fad8b63804950":{"kind":"REWORK","_number":19,"created":"2023-11-03 08:52:54.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/89/898089/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/898089/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/898089/19"}}},"commit":{"parents":[{"commit":"86570531d2e0153b432f08837a78b148a25d2142","subject":"CI: Drop tox-py38 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/86570531d2e0153b432f08837a78b148a25d2142"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-12 07:58:48.000000000","tz":120},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-11-03 08:52:37.000000000","tz":60},"subject":"Bump up Ansible supported versions to 7.x/8.x","message":"Bump up Ansible supported versions to 7.x/8.x\n\nThis change bumps up the maximum supported version of Ansible to 8.x\n(ansible-core 2.15.x) and minimum to 7.x (ansible-core 2.14.x).\n\nDepends on:\nhttps://github.com/stackhpc/ansible-role-systemd-networkd/pull/10\n\nhttps://github.com/stackhpc/ansible-role-dell-powerconnect-switch/pull/13\n\nhttps://github.com/stackhpc/ansible-role-mellanox-switch/pull/8\n\nChange-Id: Ia3550644e5fc3e83f792e46d42b4c35d5eec4705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/18540e547f686aa294ce0ac0ea0fad8b63804950"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/18540e547f686aa294ce0ac0ea0fad8b63804950"}]},"branch":"refs/heads/master"},"ce227434972629d726d0e50b701f6a2ebd98e7be":{"kind":"REWORK","_number":20,"created":"2023-11-03 10:52:08.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/89/898089/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/898089/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/898089/20"}}},"commit":{"parents":[{"commit":"86570531d2e0153b432f08837a78b148a25d2142","subject":"CI: Drop tox-py38 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/86570531d2e0153b432f08837a78b148a25d2142"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-12 07:58:48.000000000","tz":120},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-11-03 10:51:54.000000000","tz":60},"subject":"Bump up Ansible supported versions to 7.x/8.x","message":"Bump up Ansible supported versions to 7.x/8.x\n\nThis change bumps up the maximum supported version of Ansible to 8.x\n(ansible-core 2.15.x) and minimum to 7.x (ansible-core 2.14.x).\n\nDepends on:\nhttps://github.com/stackhpc/ansible-role-systemd-networkd/pull/10\n\nhttps://github.com/stackhpc/ansible-role-dell-powerconnect-switch/pull/13\n\nhttps://github.com/stackhpc/ansible-role-mellanox-switch/pull/8\n\nChange-Id: Ia3550644e5fc3e83f792e46d42b4c35d5eec4705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ce227434972629d726d0e50b701f6a2ebd98e7be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ce227434972629d726d0e50b701f6a2ebd98e7be"}]},"branch":"refs/heads/master"},"ac034770c4b85cfa05c286d088814dd6dc14c5cc":{"kind":"REWORK","_number":21,"created":"2023-11-06 09:55:54.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/89/898089/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/898089/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/898089/21"}}},"commit":{"parents":[{"commit":"86570531d2e0153b432f08837a78b148a25d2142","subject":"CI: Drop tox-py38 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/86570531d2e0153b432f08837a78b148a25d2142"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-12 07:58:48.000000000","tz":120},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-11-06 09:54:11.000000000","tz":60},"subject":"Bump up Ansible supported versions to 7.x/8.x","message":"Bump up Ansible supported versions to 7.x/8.x\n\nThis change bumps up the maximum supported version of Ansible to 8.x\n(ansible-core 2.15.x) and minimum to 7.x (ansible-core 2.14.x).\n\nDepends on:\nhttps://github.com/stackhpc/ansible-role-systemd-networkd/pull/10\n\nhttps://github.com/stackhpc/ansible-role-dell-powerconnect-switch/pull/13\n\nhttps://github.com/stackhpc/ansible-role-mellanox-switch/pull/8\n\nChange-Id: Ia3550644e5fc3e83f792e46d42b4c35d5eec4705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ac034770c4b85cfa05c286d088814dd6dc14c5cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ac034770c4b85cfa05c286d088814dd6dc14c5cc"}]},"branch":"refs/heads/master"},"61329ceff19806dfd2c810ab3dc9b7d1106e1674":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2023-11-06 09:58:47.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/89/898089/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/898089/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/898089/22"}}},"commit":{"parents":[{"commit":"9146fdbc41b0e287bedc69a6552964aed5adb2c6","subject":"CI: Changing docker network driver in molecule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9146fdbc41b0e287bedc69a6552964aed5adb2c6"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-12 07:58:48.000000000","tz":120},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-11-06 09:58:47.000000000","tz":0},"subject":"Bump up Ansible supported versions to 7.x/8.x","message":"Bump up Ansible supported versions to 7.x/8.x\n\nThis change bumps up the maximum supported version of Ansible to 8.x\n(ansible-core 2.15.x) and minimum to 7.x (ansible-core 2.14.x).\n\nDepends on:\nhttps://github.com/stackhpc/ansible-role-systemd-networkd/pull/10\n\nhttps://github.com/stackhpc/ansible-role-dell-powerconnect-switch/pull/13\n\nhttps://github.com/stackhpc/ansible-role-mellanox-switch/pull/8\n\nChange-Id: Ia3550644e5fc3e83f792e46d42b4c35d5eec4705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/61329ceff19806dfd2c810ab3dc9b7d1106e1674"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/61329ceff19806dfd2c810ab3dc9b7d1106e1674"}]},"branch":"refs/heads/master","description":"Rebase"},"607d9fbcf6518243bcb2939e758d6b9f268bba49":{"kind":"NO_CODE_CHANGE","_number":23,"created":"2023-11-06 10:51:34.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/89/898089/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/898089/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/898089/23"}}},"commit":{"parents":[{"commit":"9146fdbc41b0e287bedc69a6552964aed5adb2c6","subject":"CI: Changing docker network driver in molecule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9146fdbc41b0e287bedc69a6552964aed5adb2c6"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-12 07:58:48.000000000","tz":120},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-11-06 10:51:34.000000000","tz":0},"subject":"Bump up Ansible supported versions to 7.x/8.x","message":"Bump up Ansible supported versions to 7.x/8.x\n\nThis change bumps up the maximum supported version of Ansible to 8.x\n(ansible-core 2.15.x) and minimum to 7.x (ansible-core 2.14.x).\n\nDepends on:\nhttps://github.com/stackhpc/ansible-role-systemd-networkd/pull/10\n\nhttps://github.com/stackhpc/ansible-role-dell-powerconnect-switch/pull/13\n\nhttps://github.com/stackhpc/ansible-role-mellanox-switch/pull/8\n\ngateway_ip was being set to empty string. This was triggering neutron\n\"\"Invalid input for gateway_ip. Reason: \u0027\u0027 is not a valid IP address.\"\nOmitting by default.\n\nInstalling openstack.cloud collection, otherwise upgrade process was failing as Ansible in kayobe-venv was not upgraded.\n\nChange-Id: Ia3550644e5fc3e83f792e46d42b4c35d5eec4705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/607d9fbcf6518243bcb2939e758d6b9f268bba49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/607d9fbcf6518243bcb2939e758d6b9f268bba49"}]},"branch":"refs/heads/master","description":"Edit commit message"},"68a69654eaef0d06716c064e5bd0f3231a4763ad":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2023-11-06 15:05:14.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/89/898089/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/898089/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/898089/24"}}},"commit":{"parents":[{"commit":"ed75bf83be1108297af6bd8b6193b2f950a7c375","subject":"Merge \"CI: Drop tox-py38 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ed75bf83be1108297af6bd8b6193b2f950a7c375"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-12 07:58:48.000000000","tz":120},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-11-06 15:05:14.000000000","tz":0},"subject":"Bump up Ansible supported versions to 7.x/8.x","message":"Bump up Ansible supported versions to 7.x/8.x\n\nThis change bumps up the maximum supported version of Ansible to 8.x\n(ansible-core 2.15.x) and minimum to 7.x (ansible-core 2.14.x).\n\nDepends on:\nhttps://github.com/stackhpc/ansible-role-systemd-networkd/pull/10\n\nhttps://github.com/stackhpc/ansible-role-dell-powerconnect-switch/pull/13\n\nhttps://github.com/stackhpc/ansible-role-mellanox-switch/pull/8\n\ngateway_ip was being set to empty string. This was triggering neutron\n\"\"Invalid input for gateway_ip. Reason: \u0027\u0027 is not a valid IP address.\"\nOmitting by default.\n\nInstalling openstack.cloud collection, otherwise upgrade process was failing as Ansible in kayobe-venv was not upgraded.\n\nChange-Id: Ia3550644e5fc3e83f792e46d42b4c35d5eec4705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/68a69654eaef0d06716c064e5bd0f3231a4763ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/68a69654eaef0d06716c064e5bd0f3231a4763ad"}]},"branch":"refs/heads/master","description":"Rebase"},"da9f2c63688c89e94d1bee8f7d7059dc3871a8cf":{"kind":"NO_CODE_CHANGE","_number":25,"created":"2023-11-07 17:31:58.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/89/898089/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/898089/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/898089/25"}}},"commit":{"parents":[{"commit":"ed75bf83be1108297af6bd8b6193b2f950a7c375","subject":"Merge \"CI: Drop tox-py38 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ed75bf83be1108297af6bd8b6193b2f950a7c375"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-12 07:58:48.000000000","tz":120},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-11-07 17:31:58.000000000","tz":0},"subject":"Bump up Ansible supported versions to 7.x/8.x","message":"Bump up Ansible supported versions to 7.x/8.x\n\nThis change bumps up the maximum supported version of Ansible to 8.x\n(ansible-core 2.15.x) and minimum to 7.x (ansible-core 2.14.x).\n\nDepends on:\nhttps://github.com/stackhpc/ansible-role-systemd-networkd/pull/10\n\nhttps://github.com/stackhpc/ansible-role-dell-powerconnect-switch/pull/13\n\nhttps://github.com/stackhpc/ansible-role-mellanox-switch/pull/8\n\ngateway_ip was being set to empty string. This was triggering neutron\n\"\"Invalid input for gateway_ip. Reason: \u0027\u0027 is not a valid IP address.\"\nOmitting by default.\n\nInstalling openstack.cloud collection.\nUpgrade failed as Ansible in kayobe-venv was not upgraded.\n\nChange-Id: Ia3550644e5fc3e83f792e46d42b4c35d5eec4705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/da9f2c63688c89e94d1bee8f7d7059dc3871a8cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/da9f2c63688c89e94d1bee8f7d7059dc3871a8cf"}]},"branch":"refs/heads/master","description":"Edit commit message"},"80f6db34cd686bf9b5851fc421bcc41e00723e65":{"kind":"REWORK","_number":26,"created":"2023-11-07 17:33:07.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/89/898089/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/898089/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/898089/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/898089/26"}}},"commit":{"parents":[{"commit":"ed75bf83be1108297af6bd8b6193b2f950a7c375","subject":"Merge \"CI: Drop tox-py38 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ed75bf83be1108297af6bd8b6193b2f950a7c375"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-10-12 07:58:48.000000000","tz":120},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2023-11-07 17:33:06.000000000","tz":60},"subject":"Bump up Ansible supported versions to 7.x/8.x","message":"Bump up Ansible supported versions to 7.x/8.x\n\nThis change bumps up the maximum supported version of Ansible to 8.x\n(ansible-core 2.15.x) and minimum to 7.x (ansible-core 2.14.x).\n\nDepends on:\nhttps://github.com/stackhpc/ansible-role-systemd-networkd/pull/10\n\nhttps://github.com/stackhpc/ansible-role-dell-powerconnect-switch/pull/13\n\nhttps://github.com/stackhpc/ansible-role-mellanox-switch/pull/8\n\ngateway_ip was being set to empty string. This was triggering neutron\n\"\"Invalid input for gateway_ip. Reason: \u0027\u0027 is not a valid IP address.\"\nOmitting by default.\n\nInstalling openstack.cloud collection.\nUpgrade failed as Ansible in kayobe-venv was not upgraded.\n\nChange-Id: Ia3550644e5fc3e83f792e46d42b4c35d5eec4705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/80f6db34cd686bf9b5851fc421bcc41e00723e65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/80f6db34cd686bf9b5851fc421bcc41e00723e65"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
