)]}'
{"id":"openstack%2Fkayobe~898091","triplet_id":"openstack%2Fkayobe~master~I3fa96f0276e08a6678e5d743399d01bc19a8dd1b","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"35264":{"account":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"last_update":"2023-10-26 18:48:49.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-10-25 21:39:24.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-10-26 18:48:49.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-10-26 09:15:44.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2023-10-26 18:48:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3fa96f0276e08a6678e5d743399d01bc19a8dd1b","subject":"Remove unsupported kolla_docker_registry_insecure variable","status":"MERGED","created":"2023-10-12 08:55:29.000000000","updated":"2023-10-26 18:49:45.000000000","submitted":"2023-10-26 18:48:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898091","meta_rev_id":"cd7703c13387ae095f0f75e6e69c7e4dbf33e96e","_number":898091,"virtual_id_number":898091,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-26 18:48:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"value":2,"date":"2023-10-25 16:46:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-10-25 21:39:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-26 09:15:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-26 09:15:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"all":[{"value":1,"date":"2023-10-25 15:45:31.000000000","_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-12 09:36:07.000000000","updated_by":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"reviewer":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"state":"REVIEWER"},{"updated":"2023-10-12 11:44:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-12 15:47:53.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2023-10-13 10:43:44.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2023-10-20 16:13:26.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-10-25 16:46:52.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"c6d552ba6559afa4b6f6eda83e3383cfa05fbb4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-10-12 08:55:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df42f4d284565e068dd9ab5c22a33aa369dccd5a","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2023-10-12 09:36:07.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7e20da427a9337d8dd9e2e925327efb147356654","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 11:44:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/158a99ca657c4d288e9f5524b3f089f2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b9885a796474cad944e0f05421b8b45 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb7fa3c03b9b4dbf889f3f388d204087 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b2777d1a1e543cbbdc25f7face7b4c3 : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69ec3b6693214713ab49dcc2c1703489 : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9deb013d6fd6401fa48d3bc1823838b5 : SUCCESS in 4m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5fd2749f595a4948b71435de65db80ab : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f563fdee91bb4dabbec2140c6ca067d4 : SUCCESS in 8m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/473d0e51ca4a4c588646db4411e91d9b : SUCCESS in 3m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3143baa78ef643a0b1108ac761e47d1e : SUCCESS in 4m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/00351776260d47ddb220e4cb5403a9ee : SUCCESS in 11m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1a51280da6874a49a93dbecc9cd5f6fc : SUCCESS in 10m 42s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0c90ececc6cf4a89b10e298703cd29ee : SUCCESS in 1h 02m 22s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/0be10b04709648c2a53bce367fe1a820 : SUCCESS in 1h 02m 43s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f89cf77d97e4050ae3b78d99a6fb564 : SUCCESS in 50m 43s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/da1d800dc37b456fbea3ab77d2fabf77 : SUCCESS in 1h 07m 04s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/52a429f1e94b46e2821d692769cc936f : SUCCESS in 16m 29s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/a1150e348774425582ece6ccf9d659d0 : SUCCESS in 17m 12s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/36ed1acaea814b45b6c6459d93f28e91 : SUCCESS in 14m 17s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/674b7956833444d2b7731c3a6c8362d3 : SUCCESS in 2h 43m 06s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97cdf31a8d3c4c5ca3c5cc65318aa327 : SUCCESS in 2h 36m 25s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/6022699016d74f86982748e3bc9233fe : SUCCESS in 17m 41s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e8914e9297645d3abfca34f76fd70f8 : SUCCESS in 14m 37s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/b3743d81a7654c59a74c21799d677774 : SUCCESS in 34m 06s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/66394a8dc8594555ba0aaba7ed807ae0 : SUCCESS in 31m 31s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/315075c2a5794ef7b9f13a5849650cf3 : SUCCESS in 25m 39s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4a8d5da377654f63b98a919bd37ff5ff : SUCCESS in 14m 08s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b31b5e1c1fd4f10974fe40ff9e07859 : SUCCESS in 12m 43s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e8b5e9e4295e4bd7b0910378d8b9fd4b : SUCCESS in 17m 50s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f18dc83371ec4c2bb34db95a7aeb22fd : SUCCESS in 14m 33s","accounts_in_message":[],"_revision_number":1},{"id":"471bf299f63b454cf9bfceacdee82f2d22a90c4e","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-10-12 15:47:53.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2f53850761d016d86e0562a67a7e66c5436d6e56","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-13 10:43:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e3f3c678e87c913e86a6da3bfc47c8687b45100d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-10-20 16:13:26.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3d3b542d59a8ad99ac74ead748a5b13f5d890576","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-10-25 15:45:31.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"28151a7b4070127b1844ac9e06877dcf3d8e412b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-10-25 15:45:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e8cf95d195e46b2fc96e9e23762934992514a29c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-25 16:46:52.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d612a9ccb21f69a9b31d3bb47111e6cbc32a8cb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 18:01:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83c5e5d4f3f64a5e8777c0c97b8c0be6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62e36ece0f0f48cda901e64902f18ebd : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9451ce3ad9d44ae6a291a8ffaf1fc9a0 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d66ffc78a7244428e5b30c4b459cc7c : SUCCESS in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/820cece7a70a450390eb37461d8933ac : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23f05288754b43d48bdf33e19894af91 : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5176b21d5f324aa09198c99fed48810b : SUCCESS in 3m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e3c4eb9485d047d6b3c922e99b737b3a : SUCCESS in 3m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/06c2ac03e51b401ab704c7dd7a155554 : SUCCESS in 11m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d9eb15d2efe04aa49a24171644b11cee : SUCCESS in 13m 55s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/9d891a005ffd4bf3b36031de7287dc14 : SUCCESS in 1h 06m 03s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/6e90dffed79d43b5aad3e40b624bff0a : SUCCESS in 1h 11m 26s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6918bdc884a24eb0bf343f84a2f24111 : SUCCESS in 47m 55s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/deb3a1d26862439eac6a6ea6ad356dcf : SUCCESS in 1h 21m 05s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7f4fb0193ec046b2b305d75acf758e27 : SUCCESS in 17m 25s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/1192a7248f1647afa2197700d47db841 : SUCCESS in 14m 32s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/583c0049e4824323bbf4b81fc26e44a7 : SUCCESS in 13m 42s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2a99f4ea73294af9b9344e2c9d7dd2c8 : SUCCESS in 2h 09m 50s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1357c8852298476e8ac13569b26459ea : SUCCESS in 2h 11m 15s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/37519e7e7c9b4deba27ad7c970ff6b69 : SUCCESS in 18m 19s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4833144441f34c75b389f5d006fc5418 : SUCCESS in 12m 48s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/2c79b2bc19cd4a80a69735ba39c2ff24 : SUCCESS in 32m 19s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/55ae1fc52c6142b798e9d2c1018db6ac : SUCCESS in 29m 31s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f12c87aec9ca436381af5782ef16bb05 : SUCCESS in 23m 17s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/913b84de9ea54f33abd9d8957ffbd2ee : SUCCESS in 10m 53s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66cad7c83f6044f0885ad1b52fa362a4 : SUCCESS in 10m 24s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/aa0d3a6c4d8740dfaae68f80b73e15b8 : SUCCESS in 19m 45s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0300945c34ff4ab89570c0bd81c8c615 : SUCCESS in 13m 07s","accounts_in_message":[],"_revision_number":2},{"id":"09605cb2a5904fa78aacefe6d77678f12ad7793e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-10-25 21:39:24.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b1d8fbb252c643aa8904045dbbede6b31755fd5c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-26 09:15:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8246942b62337d4167a47009b901ebb774bf60e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 09:16:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"80d78b047cf032e4777edc0e6c4abdcacccfc42a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 11:12:07.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a467751eb6134b78979f0a6eafb1cac0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fb56db1144c484ea7c0b68b58b26c19 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93090e6d5a964afc8c1dd2610d750d46 : SUCCESS in 5m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a586beabf05046dba9c921a70c09f6cc : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90ee970467e241bf998a1baf0a9cdb61 : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa7004309cdd4b289224cf6a215f1dd6 : SUCCESS in 4m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c2592c73e2e44d95aa7a2b9ac198555c : SUCCESS in 6m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5d25cb0c55994294b3db61c2b27580e0 : SUCCESS in 10m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d6dd0bb26abf446897aeac8c759233c4 : SUCCESS in 32m 12s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/6b2e409ab1a14d78a6c0b00118d2a35a : SUCCESS in 1h 03m 13s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cdf27ef73b094cc58b189cece54246a3 : SUCCESS in 44m 11s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/51c82508ebf04b8982fdbd8306de7909 : SUCCESS in 1h 05m 28s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4726d285865e49f48832ba0782b37352 : SUCCESS in 17m 16s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe8bc33d6e4e469baa3f90b4dbe60fc1 : SUCCESS in 14m 09s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ce58ba6306f04a08aec4b30087cf29e0 : FAILURE in 1h 32m 06s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/efdb0774b412488584937b9b70f68a29 : SUCCESS in 1h 54m 23s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/dce645d5ae6e4434914b65c3e644bffb : SUCCESS in 19m 51s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dcfacec0426a4b42b9a34d116dc85d17 : SUCCESS in 13m 03s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/67be33f2cfee48969eff4b2fb8e6e614 : SUCCESS in 30m 59s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ec59929cae464eaea66c1c01bcffca46 : SUCCESS in 25m 23s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d4a4395c3e204eb1b66d3edbf441d926 : SUCCESS in 12m 23s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f711fccad6944e4b27bfcc54e60178a : SUCCESS in 11m 37s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ffc1b2666af14b75a023d0bab01fe832 : SUCCESS in 17m 26s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30f00411a0cc4c62817c03b9e9e5ad3c : SUCCESS in 13m 40s","accounts_in_message":[],"_revision_number":2},{"id":"b767129a7983e090b1ed221cdf4cbf743a3bb8cc","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-26 13:19:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ba026e7f7b3532e2267a9a46d24c7a424e033e11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 16:04:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87499acc5ccd43d29f307dfbbdd6c922\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba9d68221c8f4559a1daa784bb1e5a20 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/913934304a224dada406e5019af18a29 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db6fe574ce7c42e9825f35d22104752f : SUCCESS in 3m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fa2d0dd79ef04f93aed8ca4a549ac5f8 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c78c0226c9c24177a56f70df75708a3b : SUCCESS in 7m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b02549eb97a349309ed44c7c2349474b : SUCCESS in 3m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2547f8f643ab4683a5ebc4dd73624930 : SUCCESS in 4m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/57ce010a614c4c4d8c361903c0a2823c : SUCCESS in 13m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/24ec712902d94363a10d009a0e57f3d3 : SUCCESS in 11m 23s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/3cf22f42dd54483cb8403dd696a8ea3f : SUCCESS in 1h 10m 10s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/abddbbee58fb44ecb82e231a6fcd8fe1 : SUCCESS in 1h 01m 49s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a83a1e0128f41c480bd57b1c3372304 : SUCCESS in 50m 37s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e6ab2a526f3a47fcb75c436e40083353 : SUCCESS in 1h 09m 04s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ce92741f4d0e466f8314587d1cdd5383 : SUCCESS in 16m 18s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/b453142006d2458fbc42ca12ad434b39 : SUCCESS in 15m 54s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/23a847477fdb4f3fb4e200d4cc6cd5fa : SUCCESS in 13m 23s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0abc328144c94297af01dd278e6ac7b7 : SUCCESS in 2h 35m 18s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0258e1ea4ba44adbfec949c83070628 : SUCCESS in 2h 13m 12s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/089844331b284eba9dcd79a0de05f9ad : SUCCESS in 16m 44s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59f1b22dd6614357b5ce32898f681249 : SUCCESS in 14m 30s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/025be5787c49449d97fbe0aa933aa8da : SUCCESS in 31m 02s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7281ca14cdff407393bd0ab8618cdfb5 : SUCCESS in 30m 12s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c522b2668dc74c61a861d7467e2bae46 : SUCCESS in 24m 24s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d41f31baa9314a35960cfa623829d7ff : SUCCESS in 13m 23s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/44000c5043574d0bb7be18d383533ffd : SUCCESS in 10m 41s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3e9e3af98235443b980e4212278048d2 : SUCCESS in 19m 06s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c2ad4c94a9c8455bb0baf7905cbc7caa : SUCCESS in 14m 56s","accounts_in_message":[],"_revision_number":2},{"id":"cab70f0067c715274f542a9a90b2a22c0f9b7bad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 16:05:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6252ab7c08cd29dfc8b8449f9126278b4e839e03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 18:48:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a165c1ee38a64a588ac16cda6e9e79cc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b85fffad5f1342e7bae65342bd010303 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cbfa9299bfb745818c57167f29b6f068 : SUCCESS in 4m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6a0a44eecbf44a1aa315743854363f87 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/358f658cb75a4d5091c3168de64a5716 : SUCCESS in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a077b2de9d2643f983eba5ec52780157 : SUCCESS in 3m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2ac11b13b32a4d08bdb29ff1b4fef70a : SUCCESS in 4m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9b55659d22ed4dbc9ac957ef0d7ba2ae : SUCCESS in 11m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cc6dbc68cc514ff898a30fa950da67f5 : SUCCESS in 10m 34s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/dc1eda5792c444a583fcbe9df3fc56c8 : SUCCESS in 1h 12m 18s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf1e8474ae28466d985c189fce5dfb4a : SUCCESS in 57m 03s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/12c9053838b94585a4abeeb610847ee1 : SUCCESS in 1h 10m 18s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0979fda902094f2b9a7fc5b34ee5991a : SUCCESS in 16m 14s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/727734de38164fb4aea451b5f33d5cee : SUCCESS in 12m 38s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a085183c319643be870c5692725510db : SUCCESS in 2h 36m 51s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a65f7138de454c06945048bec2b5c99a : SUCCESS in 2h 11m 16s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/fe77a7f51b634c028da5676aab2f3133 : SUCCESS in 15m 40s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/726cd795d40d4a5fa88a278c2b274fbb : SUCCESS in 14m 14s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e0280331b9344a5c829779a8e3e05bc9 : SUCCESS in 30m 22s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0153ab6b5b5f42d28239026f49c811e1 : SUCCESS in 27m 53s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d35a228a2b27499b938528184785c9e6 : SUCCESS in 13m 28s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/06e0497fe6f6434ab35b16a480495422 : SUCCESS in 12m 30s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/820f145f277447b5af0cd07f82d81db4 : SUCCESS in 17m 03s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b44c8c0238549b1b1f94b53d979d55f : SUCCESS in 13m 38s","accounts_in_message":[],"_revision_number":2},{"id":"5c4c953ef66ed6441152ab87b60fde8959a155a2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 18:48:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"cd7703c13387ae095f0f75e6e69c7e4dbf33e96e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 18:49:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/684a447b0f0f455189edbb55549e65a6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3c798a9cfa6461581b6687ade1e4420 : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f73c711a1e4c4a43865d9bd5d7acb851 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1dce99094b8a2803a9e7e005f95de29f576f7b1a","revisions":{"95532c921db7c640c33635d46a4dc0a3883d7771":{"kind":"REWORK","_number":1,"created":"2023-10-12 08:55:29.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/91/898091/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/91/898091/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/898091/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/898091/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/898091/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/91/898091/1"}}},"commit":{"parents":[{"commit":"44af704836799e9b211e3eccfa5efa72f9713f32","subject":"Add introspection rule to update deploy kernel location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/44af704836799e9b211e3eccfa5efa72f9713f32"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-10-12 08:52:35.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-10-12 08:55:18.000000000","tz":60},"subject":"Remove unsupported kolla_docker_registry_insecure variable","message":"Remove unsupported kolla_docker_registry_insecure variable\n\nThis variable is not supported since\nI61a61ca59652b13687c2247d5881012b51f666a7, but was not removed from\netc/kayobe/kolla.yml in that change.\n\nThis change also adds the replacement variable docker_registry_insecure\nto etc/kayobe/docker.yml.\n\nTrivialFix\nChange-Id: I3fa96f0276e08a6678e5d743399d01bc19a8dd1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95532c921db7c640c33635d46a4dc0a3883d7771"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95532c921db7c640c33635d46a4dc0a3883d7771"}]},"branch":"refs/heads/master"},"1dce99094b8a2803a9e7e005f95de29f576f7b1a":{"kind":"REWORK","_number":2,"created":"2023-10-25 15:45:31.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/91/898091/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/91/898091/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/898091/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/898091/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/898091/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/91/898091/2"}}},"commit":{"parents":[{"commit":"44af704836799e9b211e3eccfa5efa72f9713f32","subject":"Add introspection rule to update deploy kernel location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/44af704836799e9b211e3eccfa5efa72f9713f32"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-10-12 08:52:35.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-10-25 15:45:31.000000000","tz":0},"subject":"Remove unsupported kolla_docker_registry_insecure variable","message":"Remove unsupported kolla_docker_registry_insecure variable\n\nThis variable is not supported since\nI61a61ca59652b13687c2247d5881012b51f666a7, but was not removed from\netc/kayobe/kolla.yml in that change.\n\nThis change also adds the replacement variable docker_registry_insecure\nto etc/kayobe/docker.yml.\n\nTrivialFix\nChange-Id: I3fa96f0276e08a6678e5d743399d01bc19a8dd1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1dce99094b8a2803a9e7e005f95de29f576f7b1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1dce99094b8a2803a9e7e005f95de29f576f7b1a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
