)]}'
{"id":"openstack%2Fkayobe~898434","triplet_id":"openstack%2Fkayobe~master~Ib847d3420dee374cec067cd8af519b510be04120","project":"openstack/kayobe","branch":"master","topic":"configdrive-gateway","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-10-16 23:11:50.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-10-23 13:45:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib847d3420dee374cec067cd8af519b510be04120","subject":"Resolve slow fact gathering when seed SNAT is off","status":"MERGED","created":"2023-10-16 15:26:29.000000000","updated":"2023-10-23 13:46:35.000000000","submitted":"2023-10-23 13:45:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898434-configdrive-gateway","meta_rev_id":"09af8ad8ee1ee9388aabb7a371d62caa064bbad2","_number":898434,"virtual_id_number":898434,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-23 13:45:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2023-10-17 11:17:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2023-10-16 23:11:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-10-23 11:01:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-10-23 11:01:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-16 16:56:50.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-10-16 18:04:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-17 11:17:10.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2023-10-23 11:01:45.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"740f7f8054199069428568099b84e245ab85ed8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-16 15:26:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"519b840ddd0fe8746ce4fd4af3cffcb87b1e10f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-16 15:30:52.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1d7058577f56eb9ec58ad5c338133910d5de852d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-16 15:31:18.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"434693ef13b11900c2fa0c4e2bc1badb2bc0b0fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 18:04:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9520c988bc644dff96d0464ed9acd5ff\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39649101410d448da9fdd89ffbf082da : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/008227076f174da5acb41e3bf56afbd3 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc31dfd09d2d43f8a00a49e99ff5828d : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef8ffe671a524103a96285fef3b5b33f : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/64b8a275a51b4da0b3f8227afcd22574 : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c67a42139f694a50bf7b3853f82a9a42 : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0132b2144604a45b0a948c5a30f43e4 : SUCCESS in 7m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2661df3da7d4350a5be418d2af9d369 : SUCCESS in 3m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e8311e141a2446f9a4f664ace5b72e93 : SUCCESS in 4m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/deee9a8cafaa40aeaeba400f4918cf57 : SUCCESS in 8m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/408a45ab13444ffa83306fab9a4f0ae3 : SUCCESS in 11m 29s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/96e7ceed70254369a6cbbfbd03c987d5 : SUCCESS in 1h 10m 00s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/ec12f21056e34a1a9742f6bab781da9b : SUCCESS in 46m 27s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b53b87ed60954034818b8c787c5a73d5 : SUCCESS in 54m 49s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/f514292ab3984301a08c93e7156d37ad : SUCCESS in 1h 02m 46s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a01899a92ede47a9a7b0f9a312809ee0 : SUCCESS in 17m 08s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/3c2e87e48fe34807845a48934ed2023c : SUCCESS in 15m 59s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c21f09df15c4daaa55832fbc838f0d8 : SUCCESS in 14m 47s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5a8a9fcff4cc4e23ae0942f27ddcce1b : SUCCESS in 2h 27m 43s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a219588ba29e43b09da0c725da930414 : SUCCESS in 2h 16m 08s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/77a53b8463ba4bd7ae46c278b6ad669b : SUCCESS in 17m 32s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1f24ca7a87164fca857a5af1a0791189 : SUCCESS in 13m 57s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/7b3353a8401b48929c02df05b1607208 : SUCCESS in 32m 25s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/eab66f463f294ca1a6a45eff5be9f68b : SUCCESS in 31m 19s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/addce7e4be494aac80792c53504e80aa : SUCCESS in 21m 41s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bd6eb441e3544b7181bf6b007bf380bb : SUCCESS in 12m 02s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6aa0c56932844329ac4f0e8a5a9aea10 : SUCCESS in 10m 19s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/15faebe507e342c6a7b36e3c6eb26830 : SUCCESS in 18m 02s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a78a330783046f4995a20937339ca3f : SUCCESS in 13m 57s","accounts_in_message":[],"_revision_number":3},{"id":"bebdea83429ea9f089b76549773af632da4056c4","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-10-16 23:11:50.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"e6721e6ffaf65618042b41eeb71268905b363b60","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-10-17 11:17:10.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"b726c24c5dbd2b845b9235ca6621da91193c5ef3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-10-23 11:01:45.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c3acc6109db500e9721acd9e4f14d24f1411a766","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 11:02:09.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fe724474aa6263b83e9f9b01584649391b9e2774","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 13:45:29.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0ce6280d8864bc68f15c543e30fc1b6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7244849c9ad9407eb19f65eb168b3ec1 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9c1d65dfcb94534a01c272054589b21 : SUCCESS in 4m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ef5cd6f38b0b46758fe294eead9fb3ee : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe3e31c0eea5422f8acf4aee8326f198 : SUCCESS in 7m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43b34c6a79734ec78df1afc2733f0176 : SUCCESS in 3m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cc640d1286944724b84b2e85f2a8b84b : SUCCESS in 4m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8c938b639e7d4db6830593e621f9afab : SUCCESS in 9m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/01989d671758466f9e3c18a9e6478865 : SUCCESS in 9m 11s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/b853d42e404648a8ad748630bfea1db4 : SUCCESS in 1h 19m 41s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0ea85ab69d64567b559fb14a4ce795b : SUCCESS in 44m 19s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/bcf986bf86d74f7282eb1293889947c4 : SUCCESS in 1h 08m 51s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a8a6d26df5aa4474a4dfd61090329da2 : SUCCESS in 16m 32s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e927974706f5400b8eb8a4058eb4b098 : SUCCESS in 12m 08s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8e88ba1667b8453ab26a28bf8f6a2b33 : SUCCESS in 2h 38m 53s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9edbb97c8d04ecca9fd0ba213ce7469 : SUCCESS in 1h 55m 42s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c0a8a4a19cda49f1b073efb7656bb957 : SUCCESS in 17m 10s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c80821ad154c4a1ea5f82f5f4a212236 : SUCCESS in 13m 46s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f74fdfab927d458cacc6fb93e2ebf098 : SUCCESS in 29m 39s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d734fc41a52a4ea5ac0d732f751f4258 : SUCCESS in 21m 43s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c1a7dfa404704e84b14d61075441031b : SUCCESS in 15m 53s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9652c1b62c0a497aac401d54f63ace1e : SUCCESS in 9m 53s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1f1bcc7c233f420ba6de0bb94d4bea95 : SUCCESS in 20m 34s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3146590b65c44b3eafac1416251da0a1 : SUCCESS in 15m 08s","accounts_in_message":[],"_revision_number":3},{"id":"b5c176623ef6a1a59b1622a613f0f75c63ed1dba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 13:45:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"09af8ad8ee1ee9388aabb7a371d62caa064bbad2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 13:46:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/599c8bac96b2440d928915189b60a1e5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3c6254f81ce4c2c8cd061dfe06d1bab : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a02c3e0143874c298086b0b0bd4eae91 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9980a8f891edac8f21d0cde9ee1e756a40c7be3b","revisions":{"dcbfa5c04cbc7b594009b4400ad0bbc59f4198b5":{"kind":"REWORK","_number":1,"created":"2023-10-16 15:26:29.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/34/898434/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/34/898434/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/898434/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/898434/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/898434/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/34/898434/1"}}},"commit":{"parents":[{"commit":"65e8f09e23fe25d863898b8d3e1e630763970b96","subject":"Merge \"Add option to skip kolla docker registry login\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/65e8f09e23fe25d863898b8d3e1e630763970b96"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-16 15:26:21.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-16 15:26:21.000000000","tz":120},"subject":"Resolve slow fact gathering when seed SNAT is off","message":"Resolve slow fact gathering when seed SNAT is off\n\nStop configuring the seed host as gateway when seed SNAT is disabled\n(which is the default since Xena), otherwise overcloud hosts may take a\nlong time to gather facts until their network configuration is applied\n(probably because hostname resolution is not functional at this stage).\n\nCloses-Bug: #2039461\nChange-Id: Ib847d3420dee374cec067cd8af519b510be04120\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dcbfa5c04cbc7b594009b4400ad0bbc59f4198b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dcbfa5c04cbc7b594009b4400ad0bbc59f4198b5"}]},"branch":"refs/heads/master"},"c00271d0025a69ae0c085aeaa94c0ed49c113f9d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-10-16 15:30:52.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/34/898434/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/34/898434/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/898434/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/898434/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/898434/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/34/898434/2"}}},"commit":{"parents":[{"commit":"65e8f09e23fe25d863898b8d3e1e630763970b96","subject":"Merge \"Add option to skip kolla docker registry login\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/65e8f09e23fe25d863898b8d3e1e630763970b96"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-16 15:30:19.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-16 15:30:19.000000000","tz":120},"subject":"Resolve slow fact gathering when seed SNAT is off","message":"Resolve slow fact gathering when seed SNAT is off\n\nStop configuring the seed host as gateway when seed SNAT is disabled\n(which is the default since Xena), otherwise overcloud hosts may take a\nlong time to gather facts until their network configuration is applied\n(probably because DNS servers are thought to be reachable until the\ngateway is removed).\n\nCloses-Bug: #2039461\nChange-Id: Ib847d3420dee374cec067cd8af519b510be04120\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c00271d0025a69ae0c085aeaa94c0ed49c113f9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c00271d0025a69ae0c085aeaa94c0ed49c113f9d"}]},"branch":"refs/heads/master"},"9980a8f891edac8f21d0cde9ee1e756a40c7be3b":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-10-16 15:31:18.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/34/898434/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/34/898434/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/898434/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/898434/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/898434/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/34/898434/3"}}},"commit":{"parents":[{"commit":"65e8f09e23fe25d863898b8d3e1e630763970b96","subject":"Merge \"Add option to skip kolla docker registry login\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/65e8f09e23fe25d863898b8d3e1e630763970b96"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-16 15:31:05.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-16 15:31:05.000000000","tz":120},"subject":"Resolve slow fact gathering when seed SNAT is off","message":"Resolve slow fact gathering when seed SNAT is off\n\nStop configuring the seed host as gateway when seed SNAT is disabled\n(which is the default since Xena), otherwise overcloud hosts may take a\nlong time to gather facts until their network configuration is applied\n(probably because DNS servers are thought to be reachable until the\ngateway is changed or removed).\n\nCloses-Bug: #2039461\nChange-Id: Ib847d3420dee374cec067cd8af519b510be04120\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9980a8f891edac8f21d0cde9ee1e756a40c7be3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9980a8f891edac8f21d0cde9ee1e756a40c7be3b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
