)]}'
{"id":"openstack%2Fkayobe~898437","triplet_id":"openstack%2Fkayobe~master~I86d8f13062ff8e664919cd5d63bc17bdafb32e0c","project":"openstack/kayobe","branch":"master","topic":"skip-registry-login","attention_set":{},"removed_from_attention_set":{"35264":{"account":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"last_update":"2023-11-01 16:16:58.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-10-18 15:01:29.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-11-01 16:16:58.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2023-10-18 15:20:57.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"I86d8f13062ff8e664919cd5d63bc17bdafb32e0c","subject":"Add seed_deploy_containers_registry_attempt_login","status":"MERGED","created":"2023-10-16 16:18:24.000000000","updated":"2023-11-01 16:18:11.000000000","submitted":"2023-11-01 16:16:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898437-skip-registry-login","meta_rev_id":"1c58c1772d2e9eb1b84bd09289d7bd852343aa06","_number":898437,"virtual_id_number":898437,"owner":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-01 16:16:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-10-26 13:20:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-10-18 15:01:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-11-01 13:36:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-18 15:20:57.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-11-01 13:36:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-16 18:54:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-17 11:07:51.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2023-10-18 12:12:36.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-10-20 12:09:41.000000000","updated_by":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2023-10-26 13:20:11.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-11-01 13:36:31.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"2ddc7170c8020172fb98b2cc024d1bcd94fde799","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2023-10-16 16:18:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5841e934e376c03b75dd04c70e1fa126f84bc9b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 18:54:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f465d05ebea245d79a06cc8bb013acad\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8e27613fc78484a81ec8da0c21ea27b : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c096160df94c4a3cb08dee1ec58c8136 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc8771b271d9493e94f5111a76b4462a : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4fa49b0c2d374d5aafab53984ac83cc7 : SUCCESS in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4b0f31ec444b4a9696e251babc8686a1 : SUCCESS in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7475b84826c14fe0a98827ae3e1396d7 : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/020c1ab8d6a74258ac2d73f9193fbaec : SUCCESS in 6m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9d36aec46fe54d9583f2f03792a565e3 : SUCCESS in 4m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0258c762f7e64898b336450bc1441eb3 : SUCCESS in 11m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/23f91909d78849d0be7441b8f9ca016d : SUCCESS in 10m 36s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e4e9af142eef49199ee218665f17e8fb : SUCCESS in 1h 03m 31s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/9b88066ae0a44672b7d28db9d7cca29f : SUCCESS in 59m 42s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/acccc318da75470c9e200c9e2801e460 : SUCCESS in 52m 22s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/237c34b3c4834a38be09aff8a3868cdf : SUCCESS in 53m 56s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/715c0141b1154e3eab1aca148657d2f4 : SUCCESS in 16m 53s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/ba8c0974ae6048cfbd7faadbc6937eb9 : SUCCESS in 16m 14s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cc2a9e4877124d868b06fc62c21c1118 : SUCCESS in 13m 25s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7a6da04994774ae0bf70e0e5f8b6226c : SUCCESS in 2h 31m 08s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a2744110985d44f0892cbfc5d11b537e : SUCCESS in 2h 14m 33s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c1cd5ed1636c491cb04ecb5c94685f13 : SUCCESS in 16m 49s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ece48e72380f4a27871d8b24f9077576 : SUCCESS in 14m 50s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/471a03e0d1ad45958d9be41580bc91db : SUCCESS in 32m 26s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2e32b11f10d647acb5d916ab2ddc3368 : SUCCESS in 29m 56s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/351fa1e3699a42fb8a2a8c204c746cad : SUCCESS in 31m 29s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ee67e9d72352441288121ce246d607df : SUCCESS in 13m 37s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47d79182daa848528c73ee2c3d4eba3d : SUCCESS in 10m 28s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e018d521393d4aa69dc6a037282ff189 : SUCCESS in 18m 18s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f35db23351c42d5bf772f4eb54b5e20 : SUCCESS in 12m 24s","accounts_in_message":[],"_revision_number":1},{"id":"85fa0a4b1f6faadd49ca7a67708b2b96b6ed2016","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-10-17 11:07:51.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8bbfa733bccce2cded1b4fb9c0ca21f916305cd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2023-10-18 12:10:58.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"0e3265bb6fd0d656e0e133a23b748ef7dd489ca1","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-10-18 12:12:36.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"48e8454c1f3af1555efde89c258e505e32774cf9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-10-18 12:15:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a4fa63da3809dae86489d6e152fc5be76b22da3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2023-10-18 13:59:10.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"44b459a3f582418047463a5a653d9512582e12b7","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2023-10-18 13:59:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"152429bce650f1bca356badd405f6e44099bc97c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-10-18 15:01:29.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"fb29a1ec2343a9390e357e7937dc8256064bd5d5","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-10-18 15:20:57.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"191f10ab9bfe63c3820f99c22bb17bca39427a21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 16:29:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/891cd155f8834bdf911c0181db4b97c2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc39a1b6e0234bf39e30f11efa8fa6d4 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97058f613c824877b44d5c05fbf527a4 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad32b20e0d5840b4b9d2a8ceec3d1d31 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21e5872b893242cc8d9eb5b66af0efe4 : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/71d5f47c46bf41fe9aa5fd525b2e0338 : SUCCESS in 3m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4b081d4e6c5e44aab01546fed7399eee : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16f61c1f9a2543b3a669929f96e53db4 : SUCCESS in 7m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/70b276391a8d4c749ec090cf6497fb60 : SUCCESS in 4m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/77e05e01b76a43159133acba31d2ea59 : SUCCESS in 9m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/35435b72cb1e4e3284cd4309cb6c3ce4 : SUCCESS in 11m 15s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/29c34d5d45eb4d1e9c5b88e63adeb774 : SUCCESS in 1h 04m 41s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/5c86ca91bcd549889e8a96d04c9e60e2 : SUCCESS in 1h 11m 13s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bed7341ff5804452b475a28748b22f4d : SUCCESS in 56m 06s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a666d99b2ebe4c8081bb44069143ecef : SUCCESS in 1h 15m 57s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ec73d9441854414d98fa1843c4380948 : SUCCESS in 15m 21s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/898b583522aa4f29992b96dbd3d3d091 : SUCCESS in 17m 17s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db85a6ee75bd4c7e95ed130d6fa86e6b : SUCCESS in 14m 02s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ec04ad224ec14032a309d3d6d9a29205 : SUCCESS in 2h 29m 46s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b3bf87c4c6542f5b68133c2cdfc42a0 : SUCCESS in 2h 15m 06s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/62967b6dc9ca4c048700eefb282e15ce : SUCCESS in 16m 30s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/68ae824a335c4cbca1846fe2f132c9eb : SUCCESS in 13m 32s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/94ea49fdea3649ce9aa04540121f77b8 : SUCCESS in 29m 15s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7eac2607aa614a9194385c4f4c1c73fc : SUCCESS in 36m 24s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b4c039acef449228cecb7d98c959f01 : SUCCESS in 24m 51s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7294f03cb62246fea690f764e02bc3e1 : SUCCESS in 13m 51s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/70adea3557084233ac946d4f59d8d3a9 : SUCCESS in 8m 02s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9ab22d755e804203a613234f1b23232b : SUCCESS in 17m 39s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f52e0e3859f74bdfb2c88ab7b0cc11ac : SUCCESS in 14m 50s","accounts_in_message":[],"_revision_number":3},{"id":"e7fd6434c6ddf8b13a5bb08ec6fc1a314cb91ae0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-26 13:20:11.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"4fbc60716be54badcc11bed8c5265af266f6c5e8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-01 13:36:31.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"009ccb716a30e07bce6e5f0e2349b325fe0a167c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 13:36:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8d2afa03d81a4e8f4956ef9d9caad1024a18b3d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 16:16:58.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee6563d3f20c4ddcb1f0848f69792d47\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7b390651ebc4abf87bd4686e4f760d9 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe4b7538653244f5b5cde8d08ab1dbd9 : SUCCESS in 3m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cdc02a74b7cd40d6ad387de446d13d12 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb48b0186de642a19d51a0168a7ff1a9 : SUCCESS in 6m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6d74a12e1ce0491e8a15dd5baf24aff8 : SUCCESS in 4m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b272ba6eb6d847248c571fd2f4d57203 : SUCCESS in 11m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7d64f6b9bd8243d8bb5492a9e23374ea : SUCCESS in 9m 38s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2f211f1dbd444bdc85d7b5bdd281efdb : SUCCESS in 1h 03m 53s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/328f29d8cd0841289faa3395ac950bb5 : SUCCESS in 54m 21s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/33b6ff7cfe3941c792381cf6c3c943fb : SUCCESS in 1h 04m 43s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f5be7f773f244707a49ea02256a48958 : SUCCESS in 16m 38s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e80585cb742c4af4af8246a7e629729e : SUCCESS in 12m 55s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/47aff03d98a24d2fba5a2badbfd260ab : SUCCESS in 2h 34m 04s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/54fd8789a485475db803d37206567863 : SUCCESS in 2h 06m 27s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/1c169478d79f4f26b43c084fca0cd576 : SUCCESS in 16m 33s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b0e4c6373a304125ba2d070b231a6f2c : SUCCESS in 11m 59s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9ceece068d3a48f5b925091904472329 : SUCCESS in 26m 41s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4dbc6014ec7741c69976bf0dabbf823d : SUCCESS in 24m 59s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d3af5afa764f40d89fd12aa716075dd3 : SUCCESS in 13m 35s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39595d105ef5434bb7dad732b1fa99b7 : SUCCESS in 9m 48s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2b2e9f067e2e4e8f8c1926c53bd6bd4d : SUCCESS in 17m 16s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/15821326fb284500bc9a19e5775defb4 : SUCCESS in 11m 20s","accounts_in_message":[],"_revision_number":3},{"id":"ef46c1d60ccd0c650bc6736695e8bc9e1fb84f68","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 16:16:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"1c58c1772d2e9eb1b84bd09289d7bd852343aa06","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 16:18:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9734464e8cb342df9a8287c982de3d87\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60c1ec1b81bf4c7b92914acfc63f9ef7 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1476b9a68bc7bda524ea6ec94d701b3aaeebf68f","revisions":{"6db02c3ec71d02f30fee657d9199f7dbd9da305b":{"kind":"REWORK","_number":1,"created":"2023-10-16 16:18:24.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/37/898437/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/37/898437/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/898437/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/898437/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/898437/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/37/898437/1"}}},"commit":{"parents":[{"commit":"65e8f09e23fe25d863898b8d3e1e630763970b96","subject":"Merge \"Add option to skip kolla docker registry login\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/65e8f09e23fe25d863898b8d3e1e630763970b96"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2023-10-16 16:15:05.000000000","tz":60},"committer":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2023-10-16 16:15:05.000000000","tz":60},"subject":"Add seed_deploy_containers_registry_attempt_login","message":"Add seed_deploy_containers_registry_attempt_login\n\nAdds seed_deploy_containers_registry_attempt_login variable to seed.yml,\nwhich acts as a redirection of deploy_containers_registry_attempt_login\nso that the variable is more descriptive, declared in a better location,\nand extensible to other groups\n\nChange-Id: I86d8f13062ff8e664919cd5d63bc17bdafb32e0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6db02c3ec71d02f30fee657d9199f7dbd9da305b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6db02c3ec71d02f30fee657d9199f7dbd9da305b"}]},"branch":"refs/heads/master"},"9a907e9a4173d5204a796043bb38ab653f414d38":{"kind":"REWORK","_number":2,"created":"2023-10-18 12:10:58.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/37/898437/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/37/898437/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/898437/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/898437/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/898437/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/37/898437/2"}}},"commit":{"parents":[{"commit":"65e8f09e23fe25d863898b8d3e1e630763970b96","subject":"Merge \"Add option to skip kolla docker registry login\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/65e8f09e23fe25d863898b8d3e1e630763970b96"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2023-10-16 16:15:05.000000000","tz":60},"committer":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2023-10-18 12:10:17.000000000","tz":60},"subject":"Add seed_deploy_containers_registry_attempt_login","message":"Add seed_deploy_containers_registry_attempt_login\n\nAdds seed_deploy_containers_registry_attempt_login variable to seed.yml,\nwhich acts as a redirection of deploy_containers_registry_attempt_login\nso that the variable is more descriptive, declared in a better location,\nand extensible to other groups\n\nChange-Id: I86d8f13062ff8e664919cd5d63bc17bdafb32e0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9a907e9a4173d5204a796043bb38ab653f414d38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9a907e9a4173d5204a796043bb38ab653f414d38"}]},"branch":"refs/heads/master"},"1476b9a68bc7bda524ea6ec94d701b3aaeebf68f":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-10-18 13:59:10.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/37/898437/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/37/898437/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/898437/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/898437/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/898437/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/37/898437/3"}}},"commit":{"parents":[{"commit":"65e8f09e23fe25d863898b8d3e1e630763970b96","subject":"Merge \"Add option to skip kolla docker registry login\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/65e8f09e23fe25d863898b8d3e1e630763970b96"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2023-10-16 16:15:05.000000000","tz":60},"committer":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2023-10-18 13:56:23.000000000","tz":60},"subject":"Add seed_deploy_containers_registry_attempt_login","message":"Add seed_deploy_containers_registry_attempt_login\n\nFollow on to Ie17ef9ce1147cbaec2e42db932c7d59293b49b1b\n\nAdds seed_deploy_containers_registry_attempt_login variable to seed.yml,\nwhich acts as a redirection of deploy_containers_registry_attempt_login\nso that the variable is more descriptive, declared in a better location,\nand extensible to other groups\n\nChange-Id: I86d8f13062ff8e664919cd5d63bc17bdafb32e0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1476b9a68bc7bda524ea6ec94d701b3aaeebf68f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1476b9a68bc7bda524ea6ec94d701b3aaeebf68f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
