)]}'
{"id":"openstack%2Fkayobe~898561","triplet_id":"openstack%2Fkayobe~stable%2F2023.1~I4609494d009c6e1f97d833a9a11529d107b3216f","project":"openstack/kayobe","branch":"stable/2023.1","topic":"bifrost-host-vars-deprovision-stable/2023.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-10-18 13:44:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4609494d009c6e1f97d833a9a11529d107b3216f","subject":"bifrost: Populate bifrost host vars on deprovision","status":"MERGED","created":"2023-10-17 13:33:28.000000000","updated":"2023-10-18 13:45:52.000000000","submitted":"2023-10-18 13:44:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898561-bifrost-host-vars-deprovision-stable/2023.1","cherry_pick_of_change":897751,"cherry_pick_of_patch_set":3,"meta_rev_id":"66a5d781d8fa8d42bfb9c900153605d086f7eb5a","_number":898561,"virtual_id_number":898561,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-18 13:44:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-10-17 16:40:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-10-17 20:19:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-17 13:59:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-10-17 16:40:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-17 13:59:46.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2023-10-17 15:22:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-17 20:19:07.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"b998e5ff2d68d7ded4863f9369cc2bbde12f2d4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-17 13:33:28.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"456b8dd69579b85d3d5484c04a4efe887f89fde7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-17 13:59:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"01683b74a2dd3270af9a8c88ca62103766a891e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 15:22:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b426b3fa2e6b47dd9a475aee1acfce9f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7670a366312f49eea8fe216cb141b916 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ccf3917d73449598864fb32e6af6798 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/332f6a6b0a6c4f549062d7d8ca49de37 : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4b178023c086470698831e23f4ef7245 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4068e097641f4bf8abb31c72d1c9143e : SUCCESS in 7m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6fe0acd66474566ae8df42768ad122d : SUCCESS in 3m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/27adc70c364c46738cd7baba50747443 : SUCCESS in 5m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/381dfc27854b46b4a8815cdf130d5663 : SUCCESS in 11m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8ab2577079ee4d428d13e916176d4c59 : SUCCESS in 15m 51s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/6e48c3d8b8484178ab838168716724e0 : SUCCESS in 53m 28s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/c78277d3e4574678a6697fc6003680c7 : SUCCESS in 59m 38s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cac1fd32ec514a5fac8bb60798250d85 : SUCCESS in 56m 41s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/5670e2084a1940769a05aeef7fddbe9e : SUCCESS in 1h 00m 23s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d13723a42f054e0085b1a3b36cecfcfe : SUCCESS in 17m 44s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/bd620ebb55d64f3ab82375f6cf41a913 : SUCCESS in 19m 24s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e96d425a61b4fae8937794e0c860ecf : SUCCESS in 15m 49s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ecc7e12de4364f4b83d5b3411948c2e4 : SUCCESS in 1h 36m 21s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/45ed7e6032994f8586b4d2c97363a1b5 : SUCCESS in 1h 38m 30s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/83c9c9abbb0c43bcb45ac1411b3d8bed : SUCCESS in 13m 47s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d84fa0e3a45f45c9b86b5bd36c12dbe5 : SUCCESS in 15m 37s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/7123b000d7674de3b94aa9a94592c497 : SUCCESS in 29m 57s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/eeb6b8252d8c496296cd210fc9f685f8 : SUCCESS in 31m 35s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e89c72b065247d9ab1a07d934e7690d : SUCCESS in 27m 33s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d0f4144e7ece44978d7826696ca5f0f6 : SUCCESS in 13m 57s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a054ef67c4d4aa095b8ac625aa2bc9c : SUCCESS in 12m 18s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f4ec6972a13d4d7ea35289303c969a4e : SUCCESS in 18m 27s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e564f37fa98e4178aa8d841b68efd07e : SUCCESS in 14m 20s","accounts_in_message":[],"_revision_number":1},{"id":"55e79e356a6b88349c825bf3c1570edf9313ae20","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-17 16:40:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a30a810dc1530f7cb0d8947f36b6b536e744fd65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 16:40:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"114f779e1bb24191d6a1400c5619d958cb68c87a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 18:18:07.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dac4dcf4c96742caae57dcb7a6e0e55a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2daed7681754c18be3e82cc81a340aa : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1fa36772e8bb4092b10e34748163a330 : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ee414fa448a46a382f6e78bd192498a : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b626a48a7b4f4732b4ea6650a2e3cf5a : SUCCESS in 7m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/838b984451294c5783771a73b2a63909 : SUCCESS in 3m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f99cd8643a3d494a8f9c977127b101b4 : SUCCESS in 5m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9b42df5599624c08902c7b7671e3806a : SUCCESS in 9m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/713ebf1929864490ad70f2b63fab7e94 : SUCCESS in 13m 44s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/92a453243dd042cabe6789c85f1cb4df : SUCCESS in 56m 06s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97e93198bb404f4a90ed8f6c66141219 : SUCCESS in 52m 19s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/799c3cc248fa48968d55ba07840a004f : SUCCESS in 1h 07m 50s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/daece20335f446aba31f7a300faa9ae3 : SUCCESS in 16m 20s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/14ad877c61014c5d8fdbc73e7cfa3c4f : SUCCESS in 13m 55s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/10321ef2682d4d2b8911070fe1988df7 : SUCCESS in 1h 28m 05s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d5db06cf37fe438d86067935035773f1 : FAILURE in 44m 06s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/fe33636d98d945f3bcccdbdaff9a84a1 : SUCCESS in 17m 00s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b9f787707d3249af974a95b1ac5b896b : SUCCESS in 14m 53s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c90b4e6c33ad43e7ad6c79c575641488 : SUCCESS in 29m 24s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d22763ec7d5c41f5823c5f4f02cdbe3c : SUCCESS in 24m 41s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/545a594750794b6ca6c151792c6efeda : SUCCESS in 13m 54s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/72b0858843364929af26cffdce62dca7 : SUCCESS in 12m 17s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f3b578073dc54710a9f79f99e92043e8 : SUCCESS in 17m 53s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b23aa56f534a4d2790b0b7ea95807896 : SUCCESS in 14m 58s","accounts_in_message":[],"_revision_number":1},{"id":"07f5a29c01569457f6218e5ef6339b0f3ce2cc66","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-10-17 20:19:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4a0a9af20c6839504b562663a505f7af319022be","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-10-17 20:21:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"10b5ad40ac5a6eb9e54c9b23ede02d13cdda97c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 22:00:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b589fab400be4c769174c51d30c590d2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8893c673055f4210aed26b1db1178d77 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/135a4eeb7c0f42f19632094cba8d0238 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d02efaaa6514020b9919707849fa4c6 : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b588d0fc57ad416996dc1694e464b2d5 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f364f647b6b04321891fcad6405d3038 : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f8f124fdafd43089acf459fb70d6e29 : SUCCESS in 3m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/babae41f19a640b4b67d19b07a283d7d : SUCCESS in 4m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8fab7a08dcc34cd2b93271c01f5356fc : SUCCESS in 8m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9ed60e486d4f4160b39fb07492bf93ef : SUCCESS in 11m 59s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/8d81f8f07b4e40d38651700c3394fbba : SUCCESS in 43m 13s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/0233105cda854d778339e452e5232acd : SUCCESS in 45m 01s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8563e277f39646faaad5c90678e55b92 : SUCCESS in 43m 27s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/da05f38dba164cc6ae5367d549cec2ed : SUCCESS in 48m 57s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/840ba5d45e7f41d3ba2311aa8ab4eb60 : SUCCESS in 15m 17s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/cf7d5f315e6e43a983160033e4458d1b : SUCCESS in 14m 48s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e30c664dfacc4366adc7f91bc29779ab : SUCCESS in 13m 11s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ff0a914ae7b748338c565089b9026da2 : SUCCESS in 1h 35m 02s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9f3e20b8a094f65b5af0c01c9c11f61 : FAILURE in 36m 33s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c1d6e65ab19f49159be3406713f5d921 : SUCCESS in 17m 31s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7940023c3f46436a92b509a044e8d095 : SUCCESS in 12m 03s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/ceb8fbc1b0d046129f4af88222b03956 : SUCCESS in 32m 46s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6907a53f71d94bccb10400d83f6ffe5e : SUCCESS in 29m 09s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67c5fa7d626a4e7a8a1177329f66c7cc : SUCCESS in 20m 40s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/235edcd20c7b4b0c95ea82b5a99ef9cc : SUCCESS in 12m 48s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7f78a2e8d5c46859370f143bc2b7c7e : SUCCESS in 9m 27s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d675afd20ae449c9bf0c221154117110 : SUCCESS in 17m 51s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7eb37ae15fd445749bb3768cb05c240d : SUCCESS in 12m 27s","accounts_in_message":[],"_revision_number":1},{"id":"16bd37a431efb7859c1689775812e46a5ffb65b2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-18 07:52:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c495c6ffb26bb10236a921aa3ddbad717fc57297","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 09:31:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9ac1d33a152f4dcd9e227a570a132a64\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae7a7994ea5b4a1eb5dafedce022a3fa : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8b72a710643425480ceafec45edeeca : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd65508928dc478abfc5482fdc261ca2 : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fe433eb1aa34440ab19a43d9df7f2eb7 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf061d183b594bb5bd9c79e1ee23f153 : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28e88eaeb59b482a8859d9e1d7f51c8c : SUCCESS in 3m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d3156d89ee0640ea8fdad58ef6f3471a : SUCCESS in 5m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e1d71b8de1f9488eb8336cacb0530632 : SUCCESS in 8m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8e386c70417f45788d3716a3dad78b09 : SUCCESS in 14m 29s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/48b53b1dc1984b748f215561b3283baf : SUCCESS in 57m 49s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/eb33ca5c66bd4efc95997faf6be1518e : SUCCESS in 1h 01m 06s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab121184d2674b019dcefeee8fe68b0a : SUCCESS in 49m 41s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/67a6d44eaae048e39c204850918d42bb : SUCCESS in 54m 04s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b92070b39c1649c59ca58db27625ad94 : SUCCESS in 16m 45s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/6b2a7c61c03a43e8ab8ae91cf4294274 : SUCCESS in 19m 39s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/42af2990306e4bd8937f2bc5103bba95 : SUCCESS in 14m 02s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/bfc362996145438e90add79b1c37bad5 : FAILURE in 25m 03s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3abe8a21ac6e4720b7de2eed5a550981 : SUCCESS in 1h 32m 27s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d6502a15fb7341c588a3663775f34dd7 : SUCCESS in 17m 15s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f85efc31e89e436c97888cc814e11e39 : SUCCESS in 19m 17s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/168a32ad789e406bb651d0964ad8dc0d : SUCCESS in 33m 40s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9c6f0f7d062446f883436e5770bdb35d : SUCCESS in 21m 02s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87b84dde64cb41d1a36b2e11e64593fc : SUCCESS in 23m 00s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2c26cae4587749dcbfc438b012fdfa32 : SUCCESS in 12m 54s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c6d45d46cd8f43b4ae9b4d1310f7bff2 : SUCCESS in 8m 14s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/12dc577a09294c0c9eda37f22be7a918 : SUCCESS in 15m 24s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69a475bbae02471794f4242b67c2d5ad : SUCCESS in 16m 01s","accounts_in_message":[],"_revision_number":1},{"id":"fdba801c6d5e10dfc79238c9e8d27ee241d3e08c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-18 10:20:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6f6540a3f6bf6431fef9f43091be71839b9fd1e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 11:59:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b98a3db4256147a4a9061f0aad1f648d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea73573984f6435eae75ec94da305ce7 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f46babc3aa84ef68524200d79c23e99 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7db485abc0954710831339fa4624ba8e : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/714d283a5d0d49ada47eabda1301ef72 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3381f9bec5314e1e88e2408890ecfddf : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e4abbc8c1c34eb7ab2235c0737f9c82 : SUCCESS in 4m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b60e42eab0f540d8b53ee0fa85854b27 : SUCCESS in 4m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5d9a5bf70e9443029dd08141f36454f5 : SUCCESS in 11m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6d35b889f11f49dcb42e748cbd369a9a : SUCCESS in 13m 59s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/50c76c13a65c435bb901bb112b9ed059 : SUCCESS in 1h 05m 32s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/b0db3c73c5d74bc493e751482ec597ae : SUCCESS in 57m 15s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d8b4980d882f4a6c928f50d73524d3d8 : SUCCESS in 55m 54s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/be36d47cfec44e7a81538629e34abda4 : SUCCESS in 1h 01m 41s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/dfc7ce43c7044345bea65a8c415a4162 : SUCCESS in 17m 15s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/204efffa754e471cbf7b03c1a958f0b8 : SUCCESS in 15m 35s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c02d92be16d47628d6b912b8c73fd77 : SUCCESS in 13m 31s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1428c5b8a5784800888b701ad95df6a3 : SUCCESS in 1h 34m 50s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4753f03dbf0c42018b719bc1389bf2a7 : SUCCESS in 1h 33m 47s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/de72fb36771b4be292ae5912a440f3cc : SUCCESS in 17m 02s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55e5d3899bec48fd9c19cd31d88f4064 : SUCCESS in 14m 22s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/b43e2785f48247c58eef8bdda569a417 : SUCCESS in 29m 10s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c2d3f9d46a994ef9bfa55c378441173c : SUCCESS in 21m 17s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2fc1f6c9c7554181b4547e6a4e2f1206 : SUCCESS in 26m 11s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f66038b868884da0bac5f3d55669a6f5 : SUCCESS in 10m 04s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5af17f3ec4cb486caf739492aae754ee : SUCCESS in 9m 22s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d64380eb2afe42f09625ebee04a7439e : SUCCESS in 18m 51s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac1d355ca0124ec7b6441609d5491ec4 : SUCCESS in 14m 32s","accounts_in_message":[],"_revision_number":1},{"id":"f2dcb487c4d05448c8e7d4385c2161a09e4813f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 12:00:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d5bc8f9711642b13a175bcc06329ff14f7a30a07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 13:44:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bc5a38a4ca3438397b7cf7973b15de5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aad6808138e6443b90c7f245e7851c50 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4e583556f9142418d8a4203593aedda : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dd36cae03f654696aaa916feaffee5e2 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5235106200654ca08ed118ab77927f3e : SUCCESS in 7m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d32524b9d2c4f15bcf2a5529c87dc55 : SUCCESS in 4m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bc726c4bff4c4973b96b8dea1d73f379 : SUCCESS in 4m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c0ddf1b022e34c55bcb057157b4b49e0 : SUCCESS in 10m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ed07f51b33a24e12b28edaa76020dc01 : SUCCESS in 14m 46s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a7affd22883b493ead748157dd36932f : SUCCESS in 1h 01m 59s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e3fc0a62faa441aba9386845498c9bba : SUCCESS in 56m 29s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e17efeafe2fc434e8427610456d691e6 : SUCCESS in 52m 15s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c7c07a0a697d4e749a46b2ee8a3393de : SUCCESS in 18m 58s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8b606c567a54f49a362d54705ac4314 : SUCCESS in 13m 39s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/83cf36560fc34524870af9cb3938a020 : SUCCESS in 1h 38m 29s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/168fd9ce4f054ae6851380d2d984ee92 : SUCCESS in 1h 43m 41s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/92864809f1724358b7c586e49634b5ca : SUCCESS in 18m 00s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8907baa3dd44a3fbdf6cd9c8a09aa5c : SUCCESS in 16m 13s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2404d001269943d38b4712ddfb0b57f1 : SUCCESS in 29m 08s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a84f490276b14c7bb1d36975dd1bb80e : SUCCESS in 27m 50s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c31e48d2e4c941d1af894c99e542e9f6 : SUCCESS in 13m 16s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a2ed64a9e9f4d94aea8ecfe075398f7 : SUCCESS in 13m 06s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8e72572466134c7f93ea3cbec0e8e7cd : SUCCESS in 17m 06s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4fd7520f5d1a473090b2b2073db4ffe8 : SUCCESS in 15m 49s","accounts_in_message":[],"_revision_number":1},{"id":"0c095e07b4042d80dee0b527a411c9fbd66defae","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 13:44:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"66a5d781d8fa8d42bfb9c900153605d086f7eb5a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 13:45:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10b992bb63284e2aa3805f56fb3c0159\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccf571e158854ea593443700e700365b : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4bfad067c9f141b0b55a0c6c4a0eee2e : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7825b28750dbe3e3a9096448ddc4662be81820dd","revisions":{"7825b28750dbe3e3a9096448ddc4662be81820dd":{"kind":"REWORK","_number":1,"created":"2023-10-17 13:33:28.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/61/898561/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/61/898561/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/898561/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/898561/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/898561/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/61/898561/1"}}},"commit":{"parents":[{"commit":"16a993be6859ad42fdbdb2b68b3b9cc27ffba631","subject":"Use importlib.metadata instead of importlib_metadata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/16a993be6859ad42fdbdb2b68b3b9cc27ffba631"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-10-10 06:03:47.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-10-17 13:33:28.000000000","tz":0},"subject":"bifrost: Populate bifrost host vars on deprovision","message":"bifrost: Populate bifrost host vars on deprovision\n\nCloses-Bug: #2038889\nChange-Id: I4609494d009c6e1f97d833a9a11529d107b3216f\n(cherry picked from commit 672d6fa24a978009588e66dd19c623ee244a8f8e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7825b28750dbe3e3a9096448ddc4662be81820dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7825b28750dbe3e3a9096448ddc4662be81820dd"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
