)]}'
{"id":"openstack%2Fkayobe~898562","triplet_id":"openstack%2Fkayobe~stable%2Fzed~I4609494d009c6e1f97d833a9a11529d107b3216f","project":"openstack/kayobe","branch":"stable/zed","topic":"bifrost-host-vars-deprovision-stable/zed","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-10-17 16:40:30.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"I4609494d009c6e1f97d833a9a11529d107b3216f","subject":"bifrost: Populate bifrost host vars on deprovision","status":"MERGED","created":"2023-10-17 13:33:36.000000000","updated":"2023-10-17 20:19:16.000000000","submitted":"2023-10-17 18:57:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898562-bifrost-host-vars-deprovision-stable/zed","cherry_pick_of_change":897751,"cherry_pick_of_patch_set":3,"meta_rev_id":"59d078acac0ef9308d3734b517949474de6898ba","_number":898562,"virtual_id_number":898562,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-17 18:57:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-10-17 16:40:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-10-17 20:19:16.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-17 14:00:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-10-17 16:40:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-17 14:00:18.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2023-10-17 14:55:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-17 20:19:16.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"9b131901cb6962c4ab558a709b10d4449f0569aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-17 13:33:36.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"bd1d9ec44b0728c76d612784f5ead128ecaf9217","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-17 14:00:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0ab0dbe3cb8b517a15536ae6f1fe4315c16724d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 14:55:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db6d064903ef495c9dc95911d88d6f85\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e4a0f44d5a8b498b9b18595ade1c13e7 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ec34951ccd34ace934123e3fd9ec3bb : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d7c43d4b5804992a42d6300fe5ffff0 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7489dfeb8c341378fe65ec65b9a1217 : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/64316d257a754f089eaa17ca82f8ba21 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c0f1b9bc1d2493d9d3dc9087b4a9b97 : SUCCESS in 7m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4bc2ecab8d084bb086303d4994537b91 : SUCCESS in 3m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5925df0d5b294d74848e77ad2e4001ce : SUCCESS in 4m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f1846646d23a4cc6a4b1842568f9c5b4 : SUCCESS in 10m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/00d1c41afa20499a93090632254f9126 : SUCCESS in 12m 58s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/9ae717252cc1441c984011d40ba9718f : SUCCESS in 50m 09s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/d86b7a7bc184422d92e685cefdb1f37e : SUCCESS in 49m 53s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/34dd10fc06b242c9b2671b33ee1d3029 : SUCCESS in 50m 52s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a519d00133844ca28e3eff3e59c42966 : SUCCESS in 53m 33s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b2b08e1dfa404226ad4d38ec30144157 : SUCCESS in 18m 37s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/68bc0ea8d8f2495888b216844e96ee08 : SUCCESS in 19m 17s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d9010bfa8bb4b42812d09f6f6ea1f9a : SUCCESS in 13m 26s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c87cf3726a64b4fbe226ce017a169ea : SUCCESS in 1h 12m 48s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/09212c36d8714b73be71bfca5852ca60 : SUCCESS in 16m 41s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97f3b3dcf0bf48a6a88426f6b08e0eff : SUCCESS in 17m 52s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/14dc9f17543740c78bfb37698d38b8c5 : SUCCESS in 31m 26s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4556a78e75b74bab8b3bb3f510bf7da5 : SUCCESS in 27m 08s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b5e98a6e9972412a9f0eba76f75e6d50 : SUCCESS in 12m 11s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e639424f47746d7ade53aff7a82c2c0 : SUCCESS in 10m 42s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0ac0c1a46bac45efba6d5b1379d9a943 : SUCCESS in 15m 28s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1936350f62ba41349eaf8b81528ac97c : SUCCESS in 14m 19s","accounts_in_message":[],"_revision_number":1},{"id":"b747a462824ef50d5ace02317e7ae5b0188b5469","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-17 16:40:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"87b41ec096a63bee5ff3a0550533ae24ff25dd80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 16:41:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"289a9fc1c169a7a4bd54aa3c4480f6c460b11270","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 18:57:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/713b4f0d4adf4ed595f4cd79bee8e93f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/611d00f4fdce4911ac6a2bbf8298d661 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be2055811d8c49fbb6df69f1e03cab9b : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d31abbbe018140fb922560788e8f2504 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7618142acbdb48fca0505d7b4d14c3cd : SUCCESS in 7m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ff96b50e8cc49d091930a4d3a1b0087 : SUCCESS in 3m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d217dff54017434f9ee6b18848c93277 : SUCCESS in 4m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/daf3fb36c64341da95d931e052d84cbf : SUCCESS in 10m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/28169f61eaf54d13a3147a2c458debd9 : SUCCESS in 13m 40s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a4e05effcd814739a27af981d3d1a0c4 : SUCCESS in 54m 05s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a07b5fe989e045ada29b1efa6b2939a8 : SUCCESS in 1h 01m 59s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/0b91d3497cbc48dbb21f8f4078e70655 : SUCCESS in 59m 28s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/463ff522384c44c3b7d198bccc55db7e : SUCCESS in 18m 09s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a601d71112e4a0d8520b349408869ac : SUCCESS in 18m 01s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b813bc613d0a4b369def9e0bce1b689b : SUCCESS in 1h 27m 04s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/329006dac2b24731aa9f3b7b8cd5e0d0 : SUCCESS in 18m 11s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a4b3c40869242249aff37854779b023 : SUCCESS in 15m 49s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6631aa1294034796b6c3a48402fa070e : SUCCESS in 26m 35s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/718e7ec42f9a4596804c247260849535 : SUCCESS in 13m 41s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1759c85492464e3e87993a8170ae7230 : SUCCESS in 11m 33s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0291e9fcf7294456b908cf9668868ee6 : SUCCESS in 17m 07s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/256e4ceb9fe74700806b446b3e75a314 : SUCCESS in 12m 04s","accounts_in_message":[],"_revision_number":1},{"id":"52b2bff2a26b69e084308f2aacc1cb2c33157c42","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 18:57:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6231cb10b2e3c7ea4284733069a66423a2829cea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 18:58:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b134e139cea941909dac40f97767fd34\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d1f9f533b9f48058d6bcfa9beb24ac9 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f345f3f82174979bb527c4bd437960b : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1},{"id":"59d078acac0ef9308d3734b517949474de6898ba","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-10-17 20:19:16.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a1a1f8d3700fa7d274edb47a7e9467c17b923c88","revisions":{"a1a1f8d3700fa7d274edb47a7e9467c17b923c88":{"kind":"REWORK","_number":1,"created":"2023-10-17 13:33:36.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/62/898562/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/62/898562/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/898562/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/898562/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/898562/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/62/898562/1"}}},"commit":{"parents":[{"commit":"c4a6883f1b2040c3553e0f29555264b44f15220f","subject":"Merge \"Use importlib.metadata instead of importlib_metadata\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c4a6883f1b2040c3553e0f29555264b44f15220f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-10-10 06:03:47.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-10-17 13:33:36.000000000","tz":0},"subject":"bifrost: Populate bifrost host vars on deprovision","message":"bifrost: Populate bifrost host vars on deprovision\n\nCloses-Bug: #2038889\nChange-Id: I4609494d009c6e1f97d833a9a11529d107b3216f\n(cherry picked from commit 672d6fa24a978009588e66dd19c623ee244a8f8e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a1a1f8d3700fa7d274edb47a7e9467c17b923c88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a1a1f8d3700fa7d274edb47a7e9467c17b923c88"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
