)]}'
{"id":"openstack%2Fkayobe~898563","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~I4609494d009c6e1f97d833a9a11529d107b3216f","project":"openstack/kayobe","branch":"stable/yoga","topic":"bifrost-host-vars-deprovision-stable/yoga","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-10-17 23:07:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4609494d009c6e1f97d833a9a11529d107b3216f","subject":"bifrost: Populate bifrost host vars on deprovision","status":"MERGED","created":"2023-10-17 13:33:44.000000000","updated":"2023-10-17 23:08:03.000000000","submitted":"2023-10-17 23:07:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898563-bifrost-host-vars-deprovision-stable/yoga","cherry_pick_of_change":897751,"cherry_pick_of_patch_set":3,"meta_rev_id":"7144a2eba2eb9d71c6287cd734d718a27d016504","_number":898563,"virtual_id_number":898563,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-17 23:07:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-10-17 16:40:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-10-17 20:19:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-17 14:00:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-10-17 16:40:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-17 14:00:21.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2023-10-17 14:56:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-17 20:19:22.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"be1f2e923297dbee73686b1d00fa76722b97d8ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-17 13:33:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"95894f55a59b9c7dc197d77d61a33161f1615ca5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-17 14:00:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6738241b5b80490872142b6c82ecfb953581181c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 14:56:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a71798c9d2b4ed19cc68334b55ac22e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f29d83f451644210afb879bfa264a357 : SUCCESS in 7m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/487d23dc9ec54d55b8b214417220d739 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/505797a0bdf74543a2b4fc88669b07d0 : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a1cdaec19b14c40a574b1a4a74da5bf : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6917f3d5a7934bcd802f70f273c95f87 : SUCCESS in 9m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c07654e5ef3343f8a35d74250f5ed323 : SUCCESS in 3m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/844f038ca74444e5996efecb228d7a1e : SUCCESS in 6m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ca1aad7a197048ebbd8254aaa3b04fef : SUCCESS in 14m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8bf1372ced3b474f85da37bcf16f97fc : SUCCESS in 18m 20s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/fde2b94189e84296bfeaafd0f1c36f25 : SUCCESS in 1h 08m 47s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/92385127f80d4a6b985df646907684b5 : SUCCESS in 51m 44s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/e5fb4f9e9265444c99a813d92218cf1f : SUCCESS in 1h 08m 37s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/6ff5348b4d94413db2274fc6fa000ef6 : SUCCESS in 48m 20s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14a91503cd624fa7a543f75a1d8a6890 : SUCCESS in 53m 40s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d7ab29c1c5364760b575500cf4b6766c : SUCCESS in 54m 35s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/8ece6b57b7d64124ac406c1c3cdcafa9 : SUCCESS in 27m 45s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/a2a7f7253e614a5a9eeeaf3631e2634b : SUCCESS in 20m 29s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/5e0419423d4749b3afd072b386920e9d : SUCCESS in 25m 09s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5be95d518943475aac1693214e7e46b5 : SUCCESS in 20m 46s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6dbb1efcfb374bf0a01b86b726f962b5 : SUCCESS in 17m 25s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c7a3f60be81456f8e5bffcba6b6ea0c : SUCCESS in 14m 38s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/88b172d895e64671ad2551e8a279ece1 : SUCCESS in 1h 16m 29s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/725b034d298e40ce83796bea43d687e8 : SUCCESS in 58m 45s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a3994b1843c043518555e72d3e8ae6e2 : SUCCESS in 23m 28s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/268fe3bdbeb9429489e3cbdbfc6eced5 : SUCCESS in 19m 58s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/0d624199af8841cf81e85d0d637ba120 : SUCCESS in 20m 04s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65653bbc7dda4e399215bdad529f786a : SUCCESS in 17m 17s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e099fe48d9f48048d6e1be57676ab7b : SUCCESS in 17m 40s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/6d370a6f497740c4a3d3a1f613b186c8 : SUCCESS in 20m 08s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/d9a37f7061d24624bfd9d93c9080dff7 : SUCCESS in 19m 39s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d962c5adf927496f8e48faedf6f81c78 : SUCCESS in 16m 27s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4fbac66b37f2472dbfbda731f91f3576 : SUCCESS in 15m 20s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6fdc675bcb549fcb984ab462f898805 : SUCCESS in 15m 13s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/f5e544a1b2df446e96d0f3ae8772cb92 : SUCCESS in 25m 12s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/77022186f57a486ca45b17960a2e771f : SUCCESS in 22m 48s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3ec76321c19c4c398be6933f866f01b0 : SUCCESS in 19m 32s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4926a14a9024e7e83e277cb436e4d86 : SUCCESS in 18m 20s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3caddf94eb2e47ababc6ba20fdc88785 : SUCCESS in 16m 32s","accounts_in_message":[],"_revision_number":1},{"id":"520e7d0958d7670d33466505edc12b5ef1dbdb83","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-17 16:40:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e2470851d5bcf74f82e683b116eca4ee55cbe117","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 16:41:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b4f0299cdb11ae6836e3e0d7eafe0443eb5c5da7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 18:57:44.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/74dd6cadae3d4b2683e00e5951172294\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1561c06bde4542b6bf14f8cf124a3afc : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5b6633a9f524557a666d3788210a3ca : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bad7601aa4a74c2499ac492d633a8f88 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02236798128b4c5ab74a763ae0ef929d : SUCCESS in 9m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2388fc1bd58f42d9bccb3bd6894bf12d : SUCCESS in 3m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0ef9a4bc6fc34b17924f76cf4c19f5d1 : SUCCESS in 6m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/06ae7973a78a4f21aca4dff70f2c816d : SUCCESS in 13m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1708f90f9eb049afb3713d552695e15b : FAILURE in 8m 20s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/09cfd8c3060943fba5e6c2dd17442e4c : SUCCESS in 1h 12m 50s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/6975448e9e1643ffbf65a5318d906bf7 : SUCCESS in 1h 10m 49s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/5d9d534fa3e84e288d280d66317b2228 : SUCCESS in 41m 25s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4724fbb637954326bc7717d41f16f6ab : SUCCESS in 52m 24s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/32fd8f0106fb4ef395346446a8dc45bd : SUCCESS in 48m 21s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/88cc8e6689b142c798d56998732f33c7 : SUCCESS in 25m 35s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/5d1975edd9ca4209a159b49156834501 : SUCCESS in 21m 14s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/449a2a2ae0ea42fc9505979ff1464708 : SUCCESS in 21m 22s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e099ed03720d49a9ac68b2c9fecfb8ec : SUCCESS in 17m 07s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6855c95919094fdd976561907ae57fe5 : SUCCESS in 16m 20s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5e92195fe1ee432c91765b50e0a6ca67 : SUCCESS in 1h 04m 38s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/4308f9032c7e4667818bef096f0c5d87 : SUCCESS in 56m 11s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d7b266b82fe74cceb3c4cf125dec3687 : SUCCESS in 23m 26s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/411253cc65014c0884a49a8ea10ecbda : SUCCESS in 23m 52s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/fe5da6db028a4235b97d845b8fc4dc41 : SUCCESS in 20m 19s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd498829573b46999895daf369977f79 : SUCCESS in 17m 55s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/50ad165c42f64563ab17077c285d15cb : SUCCESS in 22m 21s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/15f780150e904817a7897edd7f569d2d : SUCCESS in 17m 53s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/46b7dbfebf374d14a4f0ce910573695f : SUCCESS in 17m 30s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d2589a54c4b9441fa20915272445014e : SUCCESS in 16m 25s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4fdfd77c94504917afd9b721e121857a : SUCCESS in 13m 30s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b8231786b57a450ebdabd6b7f7d3b80b : SUCCESS in 12m 50s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/3fca80f5c1d54ce5a8696c8151bfaf5e : SUCCESS in 23m 14s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/4d839dd2605b4e10a02e0a058f13813a : SUCCESS in 19m 15s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/dcc673bea5304967979f16b8783753ed : SUCCESS in 16m 56s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c857768c82c841f5bd03792fa48619f6 : SUCCESS in 17m 58s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa0433efc2914e34b744ec48fbd5dd09 : SUCCESS in 12m 46s","accounts_in_message":[],"_revision_number":1},{"id":"de9c11e34cffd4f765f69e5bec78c034d685df82","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-10-17 20:19:22.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cef0ea624457d56c766f8330b5114844c0ef18ff","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-10-17 20:21:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"03c14fb8bc2756a32db3dca7f81565af3df26907","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 21:41:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22ee8461076f475f9fa0a2a56d1b7aec\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/727644e83c344c108a6f22720223edf7 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4413b6b3d6c34bc2b39ffc6a6b87f14e : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85f148f4aff64ebaa3539c3c47ecea7e : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10f219682dee41f68a20d24d877d8140 : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c303d14e72745088b5327aafe11e8bd : SUCCESS in 9m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e115cb12d2d48ffb4f625b3826de299 : SUCCESS in 3m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/063c3c6d1a444156a596c52495d51dd1 : SUCCESS in 4m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/222296e716d24ef48235b6371d673ce7 : SUCCESS in 9m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d6c877b7054542bb913e62c33daec603 : SUCCESS in 12m 55s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/13686ac8701140abb4284864c6d9f876 : SUCCESS in 56m 03s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/f1e72a1484fc4fa9b34943cb675f02dd : SUCCESS in 54m 36s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/51ac1ff3685c4e939679e6a8540427a3 : SUCCESS in 1h 01m 11s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/39d09291cb454a4eac343f3255915da7 : SUCCESS in 51m 12s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7c701f6e2fa4d91a546548a94dee340 : SUCCESS in 43m 33s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a96905666e27484c98c2427e314bf183 : SUCCESS in 44m 51s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/13b95036a34f4198b86a3ac59be1fb26 : SUCCESS in 32m 45s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/db8dd4a7be4449e48218f6d0913a2312 : SUCCESS in 17m 22s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/7a843713af3e49e9aee1cb1381000abf : SUCCESS in 26m 47s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/08d83c48b7b248acb821e62af1c8c77d : SUCCESS in 20m 19s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6faf037040084c6c9e3c0f8943a1e063 : SUCCESS in 11m 50s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5500b149f7c3462ab0a2599531561d1c : SUCCESS in 15m 24s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/58e467e3711c4af7ba256992cd3a4824 : SUCCESS in 1h 15m 14s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d05113234b5d4dda80ce93f6d373a468 : SUCCESS in 49m 03s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7989ab681b244bd7a2dab3ddf782ab59 : SUCCESS in 22m 38s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/7214cfa316c64d7faa57ef5a32fc8f01 : SUCCESS in 20m 33s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5c1a36ce36c5455daef3f01b18946d82 : SUCCESS in 18m 18s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/490d5cdc65624d798d2a98764d78b6d0 : SUCCESS in 16m 55s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/74b722f9f983492a86fe5abd8c019d0e : SUCCESS in 18m 56s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e31ebd03d06c4f3ab72ba25a729b9312 : SUCCESS in 18m 05s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/ead41c42c8044b5daccb879de3593bef : SUCCESS in 17m 22s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e0711bd4a32b41aaa33a38830ed4dcf6 : SUCCESS in 16m 09s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/614be952e3a34021a7feb8ab60fa9023 : SUCCESS in 11m 43s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f559c66ea0c447bc8d8a5fa49632589b : SUCCESS in 14m 49s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/20b2681a1f95489fafc7e4b2ed8d953c : SUCCESS in 17m 33s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/cfd99228d1f4423da521514ad8600548 : SUCCESS in 20m 40s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6a93140fb7324835866bc53c50037e8f : SUCCESS in 19m 01s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b2fbaa1871549279c296485a3cb466c : SUCCESS in 13m 40s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a477cce2507451a8de74d80e82f8495 : SUCCESS in 12m 13s","accounts_in_message":[],"_revision_number":1},{"id":"d37a74407ecac942df6f44bf27156115fb2b0854","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 21:42:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"11a111771ee300293b0f496141e0c0ca51c19e4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 23:07:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/730aca893d0540a6aa19da70a2cae75b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/543cd01334bd45ffa78d1020a0754295 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4785d5c87faa4047aa34f3ad91329955 : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d1bad0c55a98440b849cd5c850ad24f2 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d27a8704ce442f98a4c4ca7ac7fe447 : SUCCESS in 11m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8eb0690ac502498fa1e036f15f6653ce : SUCCESS in 3m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9113452b8cd348bcbfb1b25248abe73c : SUCCESS in 5m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7fc90e3109614343a80cbf62669e93f9 : SUCCESS in 11m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/57ed584780634de581f8e3b56727bfd3 : SUCCESS in 16m 32s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/647d1ac9926d4d9981b400b9f80319ae : SUCCESS in 1h 18m 25s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/5714bc00227d4947b8efbedf2223f9e8 : SUCCESS in 1h 03m 40s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/eadd4ce24b7a4e95ad5abb5c55e1cd1e : SUCCESS in 50m 47s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/83452e8fe41d4ec6a7a1eac9878ce0b2 : SUCCESS in 49m 22s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f43ed68a50448b9ae2b87046ab38907 : SUCCESS in 38m 19s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/99a40984f5ef48f5a1f580ed667c0fb6 : SUCCESS in 24m 25s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/6a6b5d100b8246619b38593a1b68ab9e : SUCCESS in 22m 01s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/2fadde6b76614c8cbc9a33f24bbea289 : SUCCESS in 20m 36s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ffc615ad8d24553a7727011eefa6e32 : SUCCESS in 15m 29s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f0862c761f1943ff9c616bd361765a70 : SUCCESS in 15m 54s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/395a30099e7a48c1b99a8470a6b4809f : SUCCESS in 1h 03m 03s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/dee72142018d41baaf4c203e407f1491 : SUCCESS in 42m 37s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/1e58278e192348218710ef987349af99 : SUCCESS in 22m 14s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/11c377591f5f4790ab9888e7fa7e548d : SUCCESS in 21m 39s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f79138018d9e466a9840e511cd8247dc : SUCCESS in 13m 53s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1b6133b63184e45b52149aa3ba4cc3c : SUCCESS in 16m 30s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/83d08715fb614f598678590144fb0a6f : SUCCESS in 14m 30s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2df4b2d824b146b09afb9df8bef85bf4 : SUCCESS in 18m 00s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/6be358afb0da4d989de5a1a1f73658c0 : SUCCESS in 18m 00s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/07df28a6c70649e99abb814a2ade8ebd : SUCCESS in 15m 56s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db408c387ef948f896f27479ad157144 : SUCCESS in 12m 18s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c617868cdeb46479783afaaf3d68eab : SUCCESS in 13m 19s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/42021b11acde433bb81edfafdee9c062 : SUCCESS in 22m 37s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/5aa66f4850b145a4a568d14bf2b8c2af : SUCCESS in 21m 50s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d84143bb3c0c440c8ae725cbdbd3a31b : SUCCESS in 18m 13s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0159b6eeef049039a93da68f77bc972 : SUCCESS in 15m 24s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55a22b34a906440bb6e9733932f06e2c : SUCCESS in 15m 25s","accounts_in_message":[],"_revision_number":1},{"id":"6702bcf93939fec490192b8c939a7ac711c15355","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 23:07:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7144a2eba2eb9d71c6287cd734d718a27d016504","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 23:08:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aef762f1cc6e4a9490f5d8d680401b20\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/190feff678ee469da0caf8d3878f0a16 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d5bfd29c9bf46fb988e737f78b65075 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f609ae27a073aac351e703751b7731df9230cc20","revisions":{"f609ae27a073aac351e703751b7731df9230cc20":{"kind":"REWORK","_number":1,"created":"2023-10-17 13:33:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/63/898563/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/63/898563/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/898563/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/898563/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/898563/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/63/898563/1"}}},"commit":{"parents":[{"commit":"390fbf1d416f64d9e75856ef17632c9ccd71483d","subject":"Fix data file path detection with new pip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/390fbf1d416f64d9e75856ef17632c9ccd71483d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-10-10 06:03:47.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-10-17 13:33:44.000000000","tz":0},"subject":"bifrost: Populate bifrost host vars on deprovision","message":"bifrost: Populate bifrost host vars on deprovision\n\nCloses-Bug: #2038889\nChange-Id: I4609494d009c6e1f97d833a9a11529d107b3216f\n(cherry picked from commit 672d6fa24a978009588e66dd19c623ee244a8f8e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f609ae27a073aac351e703751b7731df9230cc20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f609ae27a073aac351e703751b7731df9230cc20"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
