)]}'
{"id":"openstack%2Fkayobe~898566","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~I9d7d71d42026f3c12da6de9c5ca55dc647554fd0","project":"openstack/kayobe","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-11-06 11:47:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9d7d71d42026f3c12da6de9c5ca55dc647554fd0","subject":"Fix last return value in _get_base_path","status":"MERGED","created":"2023-10-17 14:00:10.000000000","updated":"2023-11-06 11:48:52.000000000","submitted":"2023-11-06 11:47:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898566","cherry_pick_of_change":896414,"cherry_pick_of_patch_set":2,"meta_rev_id":"cd145aca8cec977f99d4cbaab07fb715986941d2","_number":898566,"virtual_id_number":898566,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-06 11:47:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-10-19 07:32:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-10-20 16:06:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-11-06 10:29:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-11-06 10:29:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-17 15:16:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-19 07:32:55.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-10-20 16:06:39.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-11-06 10:29:03.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"661beb3a21b170b23784a9e6ab93408ca25da0c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-17 14:00:10.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b57c8ba4b281f75b9049cf970b739855df05ca91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 15:16:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cc1373e4a9f4349a8a5e44578f5fe6c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab04820c4fe9449a843006d76b0627fb : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/238a88aea0c54a36b455f8a59e0f9679 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/abe51e394bbd44ea840f7ccaed95f94e : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0cf665731fd4c80b3d560d0b71993d7 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b49b6b137f2d493c9619e95785691470 : SUCCESS in 11m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/76d3c02695c2421ba56481ce265219a1 : SUCCESS in 5m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6adea8f4ad844e4bb6c2c515688b8693 : SUCCESS in 13m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0c6a16009a58452ab6c0e62a59910921 : SUCCESS in 16m 37s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5797f3a9003547ffb5678c9fd1d835eb : SUCCESS in 1h 07m 34s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/22c4e6aec0cb421eb8acd0d43521d059 : SUCCESS in 57m 19s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/60a4ca84f3164dddb89c530cd178a744 : SUCCESS in 1h 01m 26s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e37f931c20584cb49eca97bc0d879b25 : SUCCESS in 46m 42s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/902a0fcf2af34e9886360e4453586d58 : SUCCESS in 53m 04s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b6c5ad2733f14de9987fa7c86d42ed50 : SUCCESS in 45m 12s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a4d5dc47162d4062a6ee4b5877000c93 : SUCCESS in 27m 03s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/6e9f192c7ade4cd09f2b90eb1c6c2801 : SUCCESS in 19m 40s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/7c61fed7a21443c999d6a91412ceebcd : SUCCESS in 24m 21s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4315705fd8e14fc0aa3f5522674bdb3a : SUCCESS in 20m 34s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e605ecce26054b799db9d88b2b2accb8 : SUCCESS in 18m 35s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d5bbdfd89d746e1b21a9093dc65750f : SUCCESS in 14m 01s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/607f4db1b0474d5d9705246a94cc0a97 : SUCCESS in 1h 07m 03s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ed140497c30841bcaccb451a6b72f1ce : SUCCESS in 46m 07s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3ba8e41efdd4426a8dd9d0414a0e48ae : SUCCESS in 20m 22s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/dc68c74ab33048f6b08944d96623eb95 : SUCCESS in 19m 43s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/dd3981fbb4074e4ead22838968326f6b : SUCCESS in 20m 10s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0518ba70349445448f9c96225d278cd4 : SUCCESS in 17m 00s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1fb8ff6b53e04359acc5f4d593a5fc58 : SUCCESS in 16m 04s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/02c4642dc1734e279dea354261012da1 : SUCCESS in 17m 01s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/5fb4c3c585ec42c69e7f3245bb82c8a5 : SUCCESS in 17m 56s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/98e751f17f2a422f8a1bfb6532af7c92 : SUCCESS in 16m 48s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92feedc1b01949d3bf048fa1f6bde41b : SUCCESS in 11m 53s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b638c3cf9e2e49e38e71adc0c8d58a70 : SUCCESS in 11m 07s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/95609390b3c2414c807e47341469f173 : SUCCESS in 19m 52s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/606ef258705c48fc86f9523260329e77 : SUCCESS in 20m 46s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c4a329d4aef949fb9a7d509c96e457af : SUCCESS in 16m 46s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c784b7d25914293b43598be77643ee6 : SUCCESS in 14m 11s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a4e0d8e0e5b4867af66f3b07287b041 : SUCCESS in 13m 36s","accounts_in_message":[],"_revision_number":1},{"id":"122faa352ca727d0fa6dc5e72d523c479bb40ed8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-19 07:32:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"34d2b7be5d78b8c7480b7ff88aa781de4d9b92b8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-10-20 16:06:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"139a30d213c02ee01f376bb328abc77005d50f7d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-06 10:29:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d614164b393851498892363f7645ab4105d09a09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 10:29:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c6d6a31a911e3d400cd66fbe04875a2708382174","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 11:47:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3c53434501b48a38f98cff728ee867f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91b05436f60841af9bba898a30cc10f1 : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40697b0395984625a27439cb6a680f6c : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09d0b18588b84e4985cde4ca6849280b : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b12fc0a9ffb4165a52e7900d8876bac : SUCCESS in 9m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/511fb19948ac4a54b8ca7ac1e93517ff : SUCCESS in 5m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8c1ec1c412d740ee816e637511ce608e : SUCCESS in 14m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/82fa09216bfc4b3cbc2937632793f3c8 : SUCCESS in 15m 29s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/325e34ce25944b97bcd4e6457fb8bacb : SUCCESS in 1h 04m 27s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/45f33e78f462431ab8a5450a0f889c35 : SUCCESS in 1h 06m 27s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/c43bd404675741d99ebe159bac03eb39 : SUCCESS in 54m 41s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57d7c36ae31a4a6888840253b8597775 : SUCCESS in 45m 16s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1fd6485cc0884b5ba4a06e58c86670e8 : SUCCESS in 46m 00s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/32792219c95d42eb96b0840980d2c850 : SUCCESS in 22m 15s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/cabf6708c77f4ccda97fa2d4451a41cf : SUCCESS in 24m 09s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0e84a1f6a6144dffb42c69b5f3b8027e : SUCCESS in 19m 05s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66cf78f9f8364c9c9477dbb005d6c7ad : SUCCESS in 15m 50s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f8a969519c84756ba2257f2f280cd49 : SUCCESS in 16m 14s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c0490df323f94cdc81eadba39b1f9f2e : SUCCESS in 1h 13m 37s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ba7b9641b6b6476391864f3f0d5b2ee4 : SUCCESS in 57m 44s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/4d7bcd0ff3b5455b817a291ef68176f4 : SUCCESS in 24m 06s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/d13ee60371fb437a9666332bde484528 : SUCCESS in 20m 56s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f8bbc10b5cd447ddbeb076732c29e69b : SUCCESS in 20m 20s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8227f3f5183142ce9e09f8b021b50283 : SUCCESS in 20m 24s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc11d1a5472242ca9173fd30fc8fc1e3 : SUCCESS in 17m 24s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/64bcf1e10b054291a15b148afda5d332 : SUCCESS in 18m 51s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/6c46b56db5994eb881a8e486fdbe3c04 : SUCCESS in 17m 28s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1497a927b2bb4efbaba663029c4429fa : SUCCESS in 16m 23s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9982e9dc4d647f99eaddddfcb944c44 : SUCCESS in 13m 19s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ebb1f513998c4436b59424e4bcfd60bb : SUCCESS in 13m 22s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/a2efed1515044aa5a7aaf66695dac80d : SUCCESS in 18m 21s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/5ab64ca039ca499b85c106e73bf0580f : SUCCESS in 19m 30s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/634adbd2093b430ba3b032fcd0def12a : SUCCESS in 20m 53s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b65158b831541c6bb2aee3222816f98 : SUCCESS in 16m 54s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d2061f4b35b49f69aa72fcbaf419db5 : SUCCESS in 14m 45s","accounts_in_message":[],"_revision_number":1},{"id":"a0df48a0a332b6db2bbccc1455896d9accd4f87f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 11:47:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cd145aca8cec977f99d4cbaab07fb715986941d2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 11:48:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/986f7881cb5b4b04b7e43c1db4d4228f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b436fa0a4d07497782168f12c09c62cf : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cf78d2e7664143e1318dceafaa6a3e80da456fab","revisions":{"cf78d2e7664143e1318dceafaa6a3e80da456fab":{"kind":"REWORK","_number":1,"created":"2023-10-17 14:00:10.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/66/898566/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/66/898566/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/898566/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/898566/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/898566/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/66/898566/1"}}},"commit":{"parents":[{"commit":"8e93ecc986b65e73e728f4050f0635023b79c093","subject":"Add introspection rule to update deploy kernel location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8e93ecc986b65e73e728f4050f0635023b79c093"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-09-25 16:43:07.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-17 14:00:10.000000000","tz":0},"subject":"Fix last return value in _get_base_path","message":"Fix last return value in _get_base_path\n\nThe return value was set to the following path which is not valid:\n\n    kayobe/kayobe/utils.py/..\n\nThe intent must have been to use dirname to produce:\n\n    kayobe/kayobe/..\n\nChange-Id: I9d7d71d42026f3c12da6de9c5ca55dc647554fd0\n(cherry picked from commit 810e021d8bdca17b1e217569f9bf24f917f6810d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cf78d2e7664143e1318dceafaa6a3e80da456fab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cf78d2e7664143e1318dceafaa6a3e80da456fab"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
