)]}'
{"id":"openstack%2Fkayobe~898776","triplet_id":"openstack%2Fkayobe~stable%2Fzed~Ie17ef9ce1147cbaec2e42db932c7d59293b49b1b","project":"openstack/kayobe","branch":"stable/zed","topic":"skip-registry-login-stable/zed","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-10-20 20:06:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie17ef9ce1147cbaec2e42db932c7d59293b49b1b","subject":"Add option to skip kolla docker registry login","status":"MERGED","created":"2023-10-20 08:11:13.000000000","updated":"2023-10-20 20:07:47.000000000","submitted":"2023-10-20 20:06:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898776-skip-registry-login-stable/zed","cherry_pick_of_change":896655,"cherry_pick_of_patch_set":3,"meta_rev_id":"8b788331783e6afbd61a95aebc39054a7ad35cc9","_number":898776,"virtual_id_number":898776,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-20 20:06:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"all":[{"value":1,"date":"2023-10-20 12:11:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"value":2,"date":"2023-10-20 18:25:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-10-20 16:03:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-20 13:38:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"value":1,"date":"2023-10-20 18:25:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-20 08:11:13.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"state":"CC"},{"updated":"2023-10-20 09:47:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-20 12:11:08.000000000","updated_by":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"reviewer":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"state":"REVIEWER"},{"updated":"2023-10-20 13:38:37.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2023-10-20 16:03:43.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-10-20 18:25:00.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"608b1f2887c9f141cfdd158afc98a0c54bf14587","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-10-20 08:11:13.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b1321968edfc49860c51d331d2bafd8718d34339","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 09:47:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a1f4d58295748efb43434479492094c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d4a6ec10ea240af99c93f1c808e3fd6 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79f0f7c129b7422d8324cd21f9f142e0 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d325cb1eaf1841d6b4d23e55f7b75f81 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7651552429d849148fb6c256008597d2 : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4f006ce5444e49a39c10a31b2335564b : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d72c21ea0d5411895e0e4332d2df229 : SUCCESS in 7m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60644bec3f4b4216865b4e9d03949019 : SUCCESS in 3m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/95a549b738cf4fdeac24695bb487c66a : SUCCESS in 4m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/17a82129c2084696aec3ee3acd1fa783 : SUCCESS in 9m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5775efa68f42466fb36a16e1605765b3 : SUCCESS in 14m 16s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/60e18b8ffd644c38a0e04fbc7562f7d8 : SUCCESS in 59m 23s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/fa737c2baeb841f2a865d1440f64914d : SUCCESS in 42m 59s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eaf03898a95343489a0ea69375c69218 : SUCCESS in 55m 47s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d468194afdf84780ae27b8a28689c79d : SUCCESS in 54m 06s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/cc4f23ded2a0441eaf8c005de23a54f6 : SUCCESS in 17m 18s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/0c6cb770da384f3987b7d7c3f0dd5518 : SUCCESS in 18m 00s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e85c5351e1f74a5bbbb74889555173f1 : SUCCESS in 18m 18s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28d71cd3ad8e45f3944a3ee2a6c15ddc : SUCCESS in 1h 31m 18s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a3ac306418ba4ea08546553fb00bcde6 : SUCCESS in 19m 04s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/01c7ec549dd9446da3c8bef464923833 : SUCCESS in 15m 08s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/4ffae25606fc41bcac6c6fb53f1e6c59 : SUCCESS in 33m 23s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f6db0fb31f84ddba33e6e21dc9e2bfc : SUCCESS in 26m 50s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d1a93993409340b8b531b22ecbc855a6 : SUCCESS in 13m 06s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d7f0c1600694d96984543c0de27cfd5 : SUCCESS in 10m 34s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/815d30102e5441adb017ca2798ba66b0 : SUCCESS in 14m 45s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a80fad90a54424ba064c38bc734d4d4 : SUCCESS in 14m 09s","accounts_in_message":[],"_revision_number":1},{"id":"068db0162a4a05b5df497cfc1e87cc2cffc66eda","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2023-10-20 12:11:08.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"728bcebea0d3e3ab4fd9dc30c0529064f27c36dd","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-20 13:38:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"17e52015891353dd4873d5cb70bc5236b264c476","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-10-20 16:03:43.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ca1b50dc810d49ec18d203b3ff56ada7de3724fe","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-20 18:25:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"646393adadebb4732439b20fbeb63eadf74cf471","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 18:25:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f681d1123ee7d4b3a9f60c13ac8e4e15b9e5d31a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 20:06:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a2d4b905c104bb8a777155f05722441\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/728ddac783b34d4dabacfa99eb52a6e8 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee8a5bd6055641a1a81c3a0efd7b98d0 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3942e2d8147d4671ac46c17e8460577b : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cc974836a4b416f90692968c4432819 : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6beabb207d4b4ac0a869223ebba78a41 : SUCCESS in 3m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6dd31b8ef360496490a49d13ce10a24c : SUCCESS in 4m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/964be0c9d4ca456aa03c91e932e7f96e : SUCCESS in 11m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f4846642ec48442d86cb286f4e4d5ef4 : SUCCESS in 13m 34s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/b773170de0b94844ba7d00f96b71d44f : SUCCESS in 38m 37s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a6e7ca3a793434795d7718ca7593971 : SUCCESS in 49m 14s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9e69c7b214ed4e9591df73edd7ef1eb5 : SUCCESS in 54m 40s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/02ab39ee182a4d1180fa394367649852 : SUCCESS in 17m 26s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1704ce1f6f324a71b31d5f8d6f24d587 : SUCCESS in 14m 58s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9fede81b29104d31a5f857d2db473cfe : SUCCESS in 1h 27m 39s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/65361fed95094bb284ded3e0083ee350 : SUCCESS in 16m 53s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fdd70d1ed8294daa8b5bc558ef5b8cec : SUCCESS in 16m 51s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7847636128d94be2adea1fae91c991d0 : SUCCESS in 26m 55s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/484b86cea61b4d91ab65f076cd1b7c05 : SUCCESS in 12m 14s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0352b589d634e61969797be3cab9e8f : SUCCESS in 10m 38s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/23d8b70d0c484ea3ac8a5e4b7501d3d0 : SUCCESS in 16m 17s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0d3e853395c448a8a6641e8063f92d49 : SUCCESS in 13m 26s","accounts_in_message":[],"_revision_number":1},{"id":"344c0b459bfd2aae0e75e5762c698dff918795d3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 20:06:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8b788331783e6afbd61a95aebc39054a7ad35cc9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 20:07:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/676de872156e42f4816440a223e8981d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6dfaf1867724be38a155e5055ecad0e : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bdce5b9ef37e434d8ef60fc3359dc1d9 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0638862a574613d1b6a87afb70d0d3ba985f24de","revisions":{"0638862a574613d1b6a87afb70d0d3ba985f24de":{"kind":"REWORK","_number":1,"created":"2023-10-20 08:11:13.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/76/898776/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/76/898776/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/898776/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/898776/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/898776/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/76/898776/1"}}},"commit":{"parents":[{"commit":"a1a1f8d3700fa7d274edb47a7e9467c17b923c88","subject":"bifrost: Populate bifrost host vars on deprovision","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a1a1f8d3700fa7d274edb47a7e9467c17b923c88"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2023-09-27 13:19:21.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-10-20 08:11:13.000000000","tz":0},"subject":"Add option to skip kolla docker registry login","message":"Add option to skip kolla docker registry login\n\nThis change adds a variable that can be used to disable kolla docker\nregistry login attempts, even when the registry username and password\nare set.\n\nThis is required for deployments using a non-standard containerised\nregistry deployed on the seed during the deploy-container step,\nsince it takes place after the registry login attempt.\n\nChange-Id: Ie17ef9ce1147cbaec2e42db932c7d59293b49b1b\n(cherry picked from commit 57d7764114052b6f69c4eb58a1b6b206067cce74)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0638862a574613d1b6a87afb70d0d3ba985f24de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0638862a574613d1b6a87afb70d0d3ba985f24de"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
