)]}'
{"id":"openstack%2Fkayobe~898905","triplet_id":"openstack%2Fkayobe~master~I23366f4de7842e7c2fe40e431fac76f26e9892de","project":"openstack/kayobe","branch":"master","topic":"nm-bridge-vlan-mtu-workaround","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-10-20 18:24:34.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-10-20 21:41:23.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-10-23 14:12:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I23366f4de7842e7c2fe40e431fac76f26e9892de","subject":"Fix MTU of NetworkManager bridge VLAN interfaces","status":"MERGED","created":"2023-10-20 08:46:22.000000000","updated":"2023-10-23 14:13:16.000000000","submitted":"2023-10-23 14:12:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898905-nm-bridge-vlan-mtu-workaround","meta_rev_id":"6901c309a4c2a840e2d52cd0d24a13d17d94d920","_number":898905,"virtual_id_number":898905,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-23 14:12:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-10-20 18:24:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-10-20 21:41:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-23 06:15:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-23 06:15:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-20 09:01:28.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2023-10-20 12:15:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-20 15:35:04.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-10-20 18:24:34.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-10-23 06:15:09.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"17dc0c87000289b6cc8a84aa9855d354b135f2a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-20 08:46:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20f6ebab119a851269a99a3770de10e73576a868","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-20 09:01:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f770cd9e5242170889fe4b110eb601a539934fd4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-20 09:08:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4f01c5191198860f59e06d469a67da187becc860","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-20 09:23:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c9a0301e9d7ec4f348242acfea6b5209a6484612","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-20 09:23:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4cfeb4e54cef646136a922b6d20d12ee21c03c4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 12:15:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eea98423ca414669a19d5eb6acd66ed4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b2fc0288bee4cd3b910008ab184fc54 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb1fa1ac27314451908c7f7b5f7dd247 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/730cd869a2a34570b08c3108dc2fbf94 : SUCCESS in 4m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4e816265265b4d1f8585d32eaa8aad68 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a7e5875e95045ff8c81e429e0565ad6 : SUCCESS in 6m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d89a46abc17f438aa2b105c790d55ec1 : SUCCESS in 4m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/47bc02123f214ba68f85bfa67eb6abb8 : SUCCESS in 9m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2296d0f1ac714f67a65c7780635a70e4 : SUCCESS in 10m 18s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/c00a8bafbde145b69fee2005c792ac6e : SUCCESS in 1h 02m 16s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/88938b18dc0c47d790127663ec7126e8 : SUCCESS in 54m 08s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/98a994eca8194d97abe38ead8378b7cf : SUCCESS in 50m 25s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/f78c7be3cc7f444b91f118c0111b16e2 : SUCCESS in 1h 15m 44s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0511cd3902f544eebd2047fdf47c3ab8 : SUCCESS in 16m 56s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/10272b082f5848bfaf7522ba4443aabe : SUCCESS in 17m 06s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b2fa7b6b6d94ab499b4029c6d939a6d : SUCCESS in 14m 57s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/cc5fc3309ab94626acb825aae96f8a9e : SUCCESS in 2h 21m 00s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/984a5c049e17402291981c6c23618172 : SUCCESS in 2h 30m 17s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5a79aee461af4f50a1b4a4c364973c32 : SUCCESS in 16m 48s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/981b44be46cd426d96e380422ec8a09e : SUCCESS in 13m 24s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/7538979d1ead441b96e9519a7f8c8f36 : POST_FAILURE in 57m 13s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f22563e8512c41a09804ad4ae1a9d67a : SUCCESS in 27m 56s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2bb1dc7d54334bab84623004c950bdc3 : SUCCESS in 26m 16s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f062e0407b6e4e73bea56c9d0450f74d : SUCCESS in 13m 04s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b2c48f14568047e880ef4dcbc3066cbb : SUCCESS in 11m 33s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8c7b04cabed04ff6a665aa6256adb181 : SUCCESS in 18m 17s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1379ef6a9873456b8396080b509e453f : SUCCESS in 14m 27s","accounts_in_message":[],"_revision_number":2},{"id":"37ebc2b70947dc12bdaa67c7db1b9c6e287a54c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-20 12:50:17.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"21a751a11e91a1684e0aeb67acf93e5a0e752859","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-20 12:51:19.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"7a151fe80fa11dba49a3b4a503d290b7f59941f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-20 12:51:51.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"8c81995bf0860821ae4846db684d4cd47bc241d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 15:28:12.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ce6bf0598a614e9f94f4af74c7bd0199\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a58add7532a4099934bde6d43ac1ae9 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0d33865b2bb4749ab16c80484b0512e : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd2ca7d722084a7d841cdf9bfd66a08c : SUCCESS in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0827a6cebacd4ae9803bdd5588fe3dec : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34c8d820288f4394a5d28f9a92d1bb23 : SUCCESS in 7m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d1415049b6149f1ace0628f3dc0fe00 : SUCCESS in 4m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1b497f67922649b9bfbc098c5e8aec17 : SUCCESS in 4m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5c8661456b114c4cbeda74f76e5084ec : SUCCESS in 11m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/92906913fa2b44abb276139a5dba30c4 : SUCCESS in 11m 15s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/077e03f31b374aebade277d96798dae2 : SUCCESS in 47m 25s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/34fb8fdb9710435699b071de8e6d5cdd : SUCCESS in 1h 00m 11s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6489cd7053e94be497e1dab89e119de7 : FAILURE in 16m 08s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/6b219f0de3a6402b85bd6836871403cf : SUCCESS in 1h 07m 42s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9e97b54c61394bcfba981fa70774f716 : SUCCESS in 16m 18s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/036e2ed5378e4ac88b5e8e7c9af9bc9c : SUCCESS in 15m 01s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe0469b47fa34cf6a841047017088843 : SUCCESS in 13m 04s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b09e6376613c42919d14bde8a46c28dd : SUCCESS in 2h 31m 34s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99ffd6d6fe254f39beab350daa70411c : SUCCESS in 2h 22m 04s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e81e06aa27644aa2a36bf5ae39afaca5 : SUCCESS in 15m 27s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c7befc83d5541e29d647d9e97653a95 : SUCCESS in 13m 12s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/a85fa513fdf64dbbab92251c2ec067e0 : SUCCESS in 40m 51s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/764c06190c114f99882cd13942e16c3d : SUCCESS in 30m 11s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f88282684eee439e8b6f529fd9c8109d : SUCCESS in 23m 57s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/df5434ea18e4414d97ec3fcdd7972f53 : SUCCESS in 12m 18s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f42e806860324ad1a5a1cee877709c98 : SUCCESS in 11m 23s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0755af54bac54d6c8f0368d473cce957 : SUCCESS in 16m 56s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9798e6cb6caf43d7bc622c0c4301edb2 : SUCCESS in 15m 35s","accounts_in_message":[],"_revision_number":5},{"id":"dc9fc1e758f5f55e039f2bed78d29947257ffcac","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-20 15:32:27.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"67ac95c5b14615b75eda5220b7cdcb270e7ebd97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 17:52:46.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8dde10d2b2324d8597051083ef80d9fa\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a4764f4b88b4d5a84a851879ed33b7f : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b5ae19e30434d7184ccdf79681c32b7 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60c36c5597ac4f42b6aa8a3bae9e965b : SUCCESS in 4m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ed698acf2f6047069bebf18fe238f8de : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a6838fe6a9a4f7bbed2ff70e58e42ff : SUCCESS in 7m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db6ee0d30f424cbaab05bdfec23af8b3 : SUCCESS in 5m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8cfa108e1a274f3c81f4af816125c8d5 : SUCCESS in 4m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/30e9b9458b1d4f2bac40bde901a96370 : SUCCESS in 11m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/48a21a543fc44ea883f6dc1b262ce2dc : SUCCESS in 9m 22s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ab4acde13b694e14b46f5b6e01952d26 : SUCCESS in 1h 02m 06s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/73e7dba007ba4e7d9c3bc6f5bbef75ab : SUCCESS in 57m 04s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30ac756f602f441ea395cdcce40251de : SUCCESS in 55m 55s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/eb16b0ade5a848aeadb0e26e3de34a71 : SUCCESS in 1h 12m 13s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9ab17a0fe2584714aafed221dc6be6ad : SUCCESS in 17m 36s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/18aa5be5d5f7472f998cbc45a77b99d1 : SUCCESS in 15m 02s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e115d30ecd04d64b1dfcc5d0d3e1294 : SUCCESS in 9m 45s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d537a09980a5404fbb4cdb086818b07e : SUCCESS in 2h 19m 46s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09491024122f41e885f8bb7eb3f393d7 : SUCCESS in 2h 01m 51s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b9d9244fd400423db3c4c4170879b663 : SUCCESS in 16m 37s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6963ef871a645b0a63d8bcb7f7cdb53 : SUCCESS in 14m 30s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/74d0c1142ed941b385060b6be8926420 : SUCCESS in 36m 10s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/dbb4f14ca3ea4b68bff3a782496d35d3 : SUCCESS in 34m 30s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b35f2fc95be400eaf6b1c0c53dea63f : SUCCESS in 24m 31s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1424093e5ffb40188c2299e92fd410a9 : SUCCESS in 12m 04s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/11ad99e6c38847e5a0b9f5e20322dd68 : SUCCESS in 10m 29s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4eaa908cb834434da6193242bde453e9 : SUCCESS in 18m 15s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c5531099fe3c4b84bbf41d90e0f55ff5 : SUCCESS in 14m 22s","accounts_in_message":[],"_revision_number":5},{"id":"f27f1859fff7618abca1dec2df063ab945c0a59a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-20 18:24:34.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"f97f59fcf82aa59220b61651ee1267cd0bdf5d3a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-10-20 21:41:23.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"036d9c70a9fb16bad3a43f82aaf161425cca3390","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-10-23 06:15:09.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"0459164a4b01bb06c23200f2b14dd98b99b74e89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 06:15:35.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"88bd25d812d8b7524ec5e84af71ef12a29376bc7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 08:46:31.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1246cb16b59143cb9dc057527b87c7b6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8dfeca9504464d30a2cb23d2c4489d4e : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe1dd3d2ceda4580a0def96058e1d263 : SUCCESS in 4m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d95674143b294c4a9a958d5984bc2cac : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23b1da8e0fa641139561de2e957c6876 : SUCCESS in 8m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c22243eca3e649998a28e2d290c7dd68 : SUCCESS in 3m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c8aee9689b8b4acc9030b39e7765b477 : SUCCESS in 4m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f4e548725077410a944f061c2a16577f : SUCCESS in 11m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1405d6fc139b40fe9baa8bc7bfa350e2 : SUCCESS in 10m 49s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/26da81da5b0b4b15a8cde5b17ab42faf : SUCCESS in 1h 08m 45s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3dde1b10ec5346bca64e38567099be28 : FAILURE in 51m 19s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8f3012d7e5a24b9cbe5800f04e365bba : SUCCESS in 1h 10m 10s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/045bc74e0b7e47e7b3edd3ac49ac9c74 : SUCCESS in 17m 10s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/62086399bc574313ae31c7f63a06237f : SUCCESS in 11m 46s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/37ab7fab7b9644f7b35b8593332853ae : SUCCESS in 2h 26m 54s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/61916a605d1a48b0bbac8e3d52baed05 : SUCCESS in 1h 50m 28s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7824b0fc0cad41ad9e03541593954108 : SUCCESS in 12m 29s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/941a230971404077ac6899eb65a58602 : SUCCESS in 11m 25s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/575989c4afb8441fbaf37db2c8da8fe0 : SUCCESS in 21m 54s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9388de4e247c4b46869c6f6a597b7285 : SUCCESS in 23m 30s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/277dfca6eb32469493571e72da54854f : SUCCESS in 13m 54s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/13abcb515997416b94d528956a6a526f : SUCCESS in 7m 35s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/53f508a07fdb4adcab2f8164253216c3 : SUCCESS in 15m 21s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c114e04fe6104f45a1341fac44fab068 : SUCCESS in 15m 35s","accounts_in_message":[],"_revision_number":5},{"id":"5d9df67172bb1c879dedca8a7078751578721c87","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-10-23 08:59:28.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"37b5fd127dd5d5d7b1ee7f2a529d2b2e854969f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 11:43:25.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/855cb811aafb49f1a8ac2da42de8b547\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f93643480434e4da282aeda9547590f : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ab4ad8a536e4e74ba2ea329cf808b25 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eec413bd663f495f8744d8ca801048f3 : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6b3d314a60de4fe7b808c1dd893d55bd : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b87d5b0725e4469da44cb2fda72174d7 : SUCCESS in 7m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e9cb4a419a142228e89f81c8ae3829c : SUCCESS in 5m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/916eb1def82d4c2e91d0ff3868d571fa : SUCCESS in 4m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3445b837fa2f43ba944aa8a6aa4a401c : SUCCESS in 11m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6b56a871aaa840bcb3cfb913ad1e1a73 : SUCCESS in 10m 59s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/07b2e0656d864a00831c3ca0f5d647c9 : SUCCESS in 54m 46s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/4fb6c7e8eab447ff9ec5566754fa3116 : SUCCESS in 1h 03m 45s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be1a37d3a1e341069dbeb32d0b01ee72 : SUCCESS in 42m 03s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/88277921d44f4616808d1804c6146ec9 : SUCCESS in 56m 14s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c23abfa18c9a4b39bafb8cac29200955 : SUCCESS in 15m 40s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/3ddea8389b1245c3a5f531c571695367 : SUCCESS in 17m 53s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b565c85c273d41afb70601693609c6aa : SUCCESS in 12m 10s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6ee1ed7ef3ad426e9396ba0338ca7b23 : SUCCESS in 2h 37m 13s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/edf476e6e9d44ffa8a17fe7026fa93ce : SUCCESS in 2h 13m 55s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/84b3a9280a374d1f8e490ca429f4a119 : SUCCESS in 15m 52s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c312e0ccd39449b8aa18eb853f1303f : SUCCESS in 11m 23s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/d6418d3308934e19bfb241fe18144715 : SUCCESS in 28m 22s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/73409e6e364645d8bd3a6c9b0bce5e70 : SUCCESS in 22m 46s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8504b4f5cdfd4301968b51bab01b35c3 : SUCCESS in 22m 48s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e1a23daac8944700a4fc56dd44b64db3 : SUCCESS in 10m 20s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e6f54a110d3949a4af73aee7f95fa165 : SUCCESS in 8m 27s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4ed89e85259a4a78ac87d3f3b6c085d0 : SUCCESS in 15m 00s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/986af74a38cc4091bbaf9905c05408dc : SUCCESS in 11m 57s","accounts_in_message":[],"_revision_number":5},{"id":"01b2e469e1ba52e6b687fb7a279378b5894f051c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 11:43:57.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"9aa0cd2dd24072f2f98c6c6bc990f51feddb25f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 14:12:06.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2786dbb21ba94a40ab7c0f2af60ec478\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b772a15d6c14c6588b51def607e8d50 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa810324480444e782193880112fae97 : SUCCESS in 3m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7f688804f9ed4e9394e75c9127b28279 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62bcd2bd2212447a8a1f797dc6bebeeb : SUCCESS in 6m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7830fce0045948e680722f862f6bf866 : SUCCESS in 3m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5f06fc912e864ad1a34f68017ff965e9 : SUCCESS in 4m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/86d50c70d6d84cec9df009f44ce4ccce : SUCCESS in 11m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/52a02bfe617b4640b3502ab30647d3c6 : SUCCESS in 10m 33s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/8bcfe4156dec4434952fb16bebc3d9ca : SUCCESS in 1h 07m 41s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e0786a9d4dc430597b08502bf5aa803 : SUCCESS in 52m 33s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/b3883022fe7d4983b6bf452b5752e93b : SUCCESS in 1h 07m 23s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/110c39995e0646f89812ec9f57d4eece : SUCCESS in 15m 46s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a5145b8cff24b1a9aa8e966a83245fa : SUCCESS in 13m 54s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7d2ce208c06349f098ecca155697a86f : SUCCESS in 2h 20m 26s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b7034bf4e674187bacce961e119dbc4 : SUCCESS in 2h 23m 09s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d3dbd45192bb476fa60d7056e3f305dc : SUCCESS in 18m 05s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9cd778bc6de94018a7f5472bb51071df : SUCCESS in 13m 41s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0cb82a62e550462f8752defad114b11e : SUCCESS in 30m 18s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bdc5c0a4db474ba78da7e6e8e8f5e2c5 : SUCCESS in 20m 58s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5f4480f1467d407aa8a40dd3673b21d8 : SUCCESS in 12m 56s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6d70b40382ce4073af64d4460368bea0 : SUCCESS in 10m 02s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8df9fdf475854c1589ad8e47be56aadb : SUCCESS in 17m 08s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d09354b0cb8c4e7c8796c0ec2785eeec : SUCCESS in 14m 21s","accounts_in_message":[],"_revision_number":5},{"id":"50d96fa0fe4b5cd2c82e0c110f74aa620cb8055a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 14:12:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"6901c309a4c2a840e2d52cd0d24a13d17d94d920","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 14:13:16.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a31a3c4ef224accb2b9202ca7536be9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d65b8baa16e478ebed16c91e524f8ad : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d26a8bfb981c4fe7896974c17ad14e4b : SUCCESS in 41s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"ee63b3253dedaf62cc321f5b29080d7e7c459a99","revisions":{"4e25df56d66ae46e4ca6a4e137a529999d2ac17c":{"kind":"REWORK","_number":1,"created":"2023-10-20 08:46:22.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/05/898905/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/898905/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/898905/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/898905/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/898905/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/898905/1"}}},"commit":{"parents":[{"commit":"4e2a0d2e8d3a92946e1b0eb0266670c190721a06","subject":"Merge \"Fix last return value in _get_base_path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4e2a0d2e8d3a92946e1b0eb0266670c190721a06"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-20 08:09:41.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-20 08:09:41.000000000","tz":120},"subject":"Avoid MTU-1 on NM bridge VLAN interfaces","message":"Avoid MTU-1 on NM bridge VLAN interfaces\n\nWhen configuring MTU for bridges, NetworkManager first sets the MTU to\nMTU-1 before setting it to the correct value [1]. We are seeing an issue\nwith VLAN interfaces on bridges which are set to MTU-1 after a reboot.\n\nWe can work around this issue by not explicitly setting the MTU\nconfiguration on the VLAN interface if it is the same as the parent\nbridge. It will be automatically inherited from the parent.\n\n[1] https://gitlab.freedesktop.org/NetworkManager/NetworkManager/-/commit/864872e9a8c694ee7472154bf45f869040310b71\n\nChange-Id: I23366f4de7842e7c2fe40e431fac76f26e9892de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4e25df56d66ae46e4ca6a4e137a529999d2ac17c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4e25df56d66ae46e4ca6a4e137a529999d2ac17c"}]},"branch":"refs/heads/master"},"fa93f937b4a4f185cfb8f7a4c020304a50dd1f8d":{"kind":"REWORK","_number":2,"created":"2023-10-20 09:23:05.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/05/898905/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/898905/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/898905/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/898905/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/898905/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/898905/2"}}},"commit":{"parents":[{"commit":"4e2a0d2e8d3a92946e1b0eb0266670c190721a06","subject":"Merge \"Fix last return value in _get_base_path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4e2a0d2e8d3a92946e1b0eb0266670c190721a06"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-20 09:21:07.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-20 09:21:07.000000000","tz":120},"subject":"Avoid MTU-1 on NM bridge VLAN interfaces","message":"Avoid MTU-1 on NM bridge VLAN interfaces\n\nWe are seeing an issue on Rocky Linux 9 where VLAN interfaces on bridges\nhave their MTU set to MTU-1 after rebooting. This is likely related to\nhow MTU is set by NetworkManager for bridges: it first sets the MTU to\nMTU-1 before setting it to the correct value [1].\n\nWe can work around this issue by not explicitly setting the MTU\nconfiguration on the VLAN interface if it is the same as the parent\nbridge. It will be automatically inherited from the parent.\n\n[1] https://gitlab.freedesktop.org/NetworkManager/NetworkManager/-/commit/864872e9a8c694ee7472154bf45f869040310b71\n\nCloses-Bug: #2039947\nChange-Id: I23366f4de7842e7c2fe40e431fac76f26e9892de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fa93f937b4a4f185cfb8f7a4c020304a50dd1f8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fa93f937b4a4f185cfb8f7a4c020304a50dd1f8d"}]},"branch":"refs/heads/master"},"be0388185503d0d67912d5e1c39309ee9ced3d7d":{"kind":"REWORK","_number":3,"created":"2023-10-20 12:50:17.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/05/898905/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/898905/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/898905/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/898905/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/898905/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/898905/3"}}},"commit":{"parents":[{"commit":"4e2a0d2e8d3a92946e1b0eb0266670c190721a06","subject":"Merge \"Fix last return value in _get_base_path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4e2a0d2e8d3a92946e1b0eb0266670c190721a06"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-20 12:49:41.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-20 12:49:41.000000000","tz":120},"subject":"Avoid MTU-1 on NM bridge VLAN interfaces","message":"Avoid MTU-1 on NM bridge VLAN interfaces\n\nWe are seeing an issue on Rocky Linux 9 where VLAN interfaces on bridges\nhave their MTU set to MTU-1 after rebooting. This is likely related to\nhow MTU is set by NetworkManager for bridges: it first sets the MTU to\nMTU-1 before setting it to the correct value [1].\n\nWe can work around this issue by not explicitly setting the MTU\nconfiguration on the VLAN interface if it is the same as the parent\nbridge. It will be automatically inherited from the parent.\n\n[1] https://gitlab.freedesktop.org/NetworkManager/NetworkManager/-/commit/864872e9a8c694ee7472154bf45f869040310b71\n\nCloses-Bug: #2039947\nChange-Id: I23366f4de7842e7c2fe40e431fac76f26e9892de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/be0388185503d0d67912d5e1c39309ee9ced3d7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/be0388185503d0d67912d5e1c39309ee9ced3d7d"}]},"branch":"refs/heads/master"},"59116f03d12edb1ec941d1a6ca57bb56351e4bcd":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-10-20 12:51:19.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/05/898905/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/898905/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/898905/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/898905/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/898905/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/898905/4"}}},"commit":{"parents":[{"commit":"4e2a0d2e8d3a92946e1b0eb0266670c190721a06","subject":"Merge \"Fix last return value in _get_base_path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4e2a0d2e8d3a92946e1b0eb0266670c190721a06"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-20 12:51:07.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-20 12:51:07.000000000","tz":120},"subject":"Avoid MTU-1 on NetworkManager bridge VLAN interfaces","message":"Avoid MTU-1 on NetworkManager bridge VLAN interfaces\n\nWe are seeing an issue on Rocky Linux 9 where VLAN interfaces on bridges\nhave their MTU set to MTU-1 after rebooting. This is likely related to\nhow MTU is set by NetworkManager for bridges: it first sets the MTU to\nMTU-1 before setting it to the correct value [1].\n\nWe can work around this issue by not explicitly setting the MTU\nconfiguration on the VLAN interface if it is the same as the parent\nbridge. It will be automatically inherited from the parent.\n\n[1] https://gitlab.freedesktop.org/NetworkManager/NetworkManager/-/commit/864872e9a8c694ee7472154bf45f869040310b71\n\nCloses-Bug: #2039947\nChange-Id: I23366f4de7842e7c2fe40e431fac76f26e9892de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/59116f03d12edb1ec941d1a6ca57bb56351e4bcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/59116f03d12edb1ec941d1a6ca57bb56351e4bcd"}]},"branch":"refs/heads/master"},"ee63b3253dedaf62cc321f5b29080d7e7c459a99":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2023-10-20 12:51:51.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/05/898905/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/898905/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/898905/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/898905/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/898905/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/898905/5"}}},"commit":{"parents":[{"commit":"4e2a0d2e8d3a92946e1b0eb0266670c190721a06","subject":"Merge \"Fix last return value in _get_base_path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4e2a0d2e8d3a92946e1b0eb0266670c190721a06"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-20 12:51:37.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-20 12:51:37.000000000","tz":120},"subject":"Fix MTU of NetworkManager bridge VLAN interfaces","message":"Fix MTU of NetworkManager bridge VLAN interfaces\n\nWe are seeing an issue on Rocky Linux 9 where VLAN interfaces on bridges\nhave their MTU set to MTU-1 after rebooting. This is likely related to\nhow MTU is set by NetworkManager for bridges: it first sets the MTU to\nMTU-1 before setting it to the correct value [1].\n\nWe can work around this issue by not explicitly setting the MTU\nconfiguration on the VLAN interface if it is the same as the parent\nbridge. It will be automatically inherited from the parent.\n\n[1] https://gitlab.freedesktop.org/NetworkManager/NetworkManager/-/commit/864872e9a8c694ee7472154bf45f869040310b71\n\nCloses-Bug: #2039947\nChange-Id: I23366f4de7842e7c2fe40e431fac76f26e9892de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ee63b3253dedaf62cc321f5b29080d7e7c459a99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ee63b3253dedaf62cc321f5b29080d7e7c459a99"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
