)]}'
{"id":"openstack%2Fkayobe~898916","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~Id3e7913f4ba9d05cec332dab560fc42d44dd4130","project":"openstack/kayobe","branch":"stable/yoga","topic":"bump-interfaces-role-stable/yoga","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-10-20 21:42:20.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-10-23 22:42:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id3e7913f4ba9d05cec332dab560fc42d44dd4130","subject":"Bump MichaelRigart.interfaces to v1.14.4","status":"MERGED","created":"2023-10-20 09:58:12.000000000","updated":"2023-10-23 22:43:07.000000000","submitted":"2023-10-23 22:42:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898916-bump-interfaces-role-stable/yoga","meta_rev_id":"906d8c699ddca2d0c9eff7029728604cda638e66","_number":898916,"virtual_id_number":898916,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-23 22:42:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2023-10-20 21:42:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-10-23 20:43:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-10-23 20:43:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-20 11:27:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-20 16:01:25.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"CC"},{"updated":"2023-10-20 16:18:13.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-10-23 20:43:32.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"7d649b77619095a8faebcc3d021a9eba4fddb725","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-20 09:58:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1eeb7b49d50eb912f1584dbc8f47dae682bca737","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-20 10:34:59.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"6a4909ed51ed1d3e1f270bc65e927e1533d36cf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 11:27:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18ad6e3336ef41d1b8dcc74511a333cf\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/865e0464381b4fd784367f6d59652791 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/753f2d2df3c44aa8a58944053f006987 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7c18291e1d0411c892d8970a6d7a432 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8670aef7ea5d4754ae379f1668b8e3c3 : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3965294bcff04852a88c3d8ee9293d55 : SUCCESS in 9m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/014458a0efae4e1591ad484f947d336e : SUCCESS in 6m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/319bb69866684d058cd5be5543638634 : SUCCESS in 13m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0d4f84c5d3154336892c2d5e95d4fa94 : SUCCESS in 17m 03s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c3ed9972d2f9443783ba7d8534c79b0e : SUCCESS in 59m 13s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/9bad77c005e1478abc94277bd3b00002 : SUCCESS in 50m 09s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/ca62af256e7e4d09b3dae6f58433e3ee : SUCCESS in 1h 23m 49s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a0bbbbc8f8684bcf99234132525d49e2 : SUCCESS in 55m 11s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6967bacce5cf4f59973b021af78e5e00 : SUCCESS in 54m 30s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ca4a68b45de48e6a748f5bebc761245 : SUCCESS in 48m 47s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/4c63d71982574cd6a479aab77bc314b9 : SUCCESS in 27m 35s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/9e915fd308b94448ac4fb37b6dd8f26e : SUCCESS in 24m 31s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/22e7eaa4cd3f41f3abb092bea38a4f89 : SUCCESS in 21m 14s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/62b44e2251ce484fab7ca17d28540fa5 : SUCCESS in 22m 27s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/636a237741f24110ac3ca228bbd95f7e : SUCCESS in 19m 57s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/171802a6db40439c8a6c3297c5fab70a : SUCCESS in 15m 37s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/bf175bbcf5734d19bc8d5b81e9fcfadc : SUCCESS in 1h 13m 15s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/7fba76752e5540a3ba300f42516d000b : SUCCESS in 1h 02m 00s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/847445e6cd2548c8a7614a618b872367 : SUCCESS in 22m 51s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/75bee8fe47bd40eab89b4e98397f885f : SUCCESS in 21m 34s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/939148b8175141e4973a7c994a4b4154 : SUCCESS in 23m 19s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab167affcfbe4009b9b51b4da94bd50c : SUCCESS in 16m 22s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e5a613fa95c34fce8b33fc981db85490 : SUCCESS in 19m 33s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/99675ce5b5c147d0868542b262df10e1 : SUCCESS in 18m 21s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/357a4976ad6245ca86708425201e3611 : SUCCESS in 15m 40s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b5c54b61cf1b473999c44ac2c8a5e6bd : SUCCESS in 17m 53s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7c57fa4fb53407097c94978ac7a6d2b : SUCCESS in 13m 33s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ebc0b5906eb249cabd1e11cdde057713 : SUCCESS in 12m 57s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/e67835b0d4804d7eb190ceac5d01871a : SUCCESS in 21m 59s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/ea5d26c8d723497e86e8e5ad6319f389 : SUCCESS in 22m 07s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a9fdc58a54014b5d8f193ed584376238 : SUCCESS in 16m 21s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7ac8bab6d1441fbbb5a3dc6dc2640fa : SUCCESS in 14m 59s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b15b64b756fb4b75ae88f093741559aa : SUCCESS in 16m 10s","accounts_in_message":[],"_revision_number":1},{"id":"2a47cb520e17654f36e18fbdcd12ea0efa77a79a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-10-20 16:01:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"87d6f4b1dbb592ca786f219a9e26961d64a5668a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-20 16:09:41.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"49f39f60b0a33bd9d49e45d7ae67ac9735fd762b","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-20 16:09:49.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"26720a420b2ddf72fc0f9258d1aed6d958795c38","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-10-20 16:18:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a45c0e370a1d6e865911a715bbeb5dc1cf347859","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-20 20:30:41.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"96169f8046d908c7e5e18841cbf28c4a7cc839ac","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-20 20:30:56.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"97dc80ca7dd57c0562569db762a23ac1f43fafbe","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-10-20 21:42:20.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"776a657f83905eac6c95a1a65f17da1ab48d4f9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 21:53:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73f83e04617b4e908cf823bb9eeb6c08\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2b3ad6a29b7452f9e0327a0b0829cd9 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bf4628a850e4eec85ffe9d1736a1afc : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee825f81f37b4313aa2462c88d431ee8 : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/388a03f374b24eefa6a1bd6c3ae14ae3 : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33d250faf7fe4f23b3a24b1016cf8b86 : SUCCESS in 11m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78f61c0b8b9a46e39d1262ec9767f700 : SUCCESS in 3m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/612b839836ca49e696ea050a3067fe96 : SUCCESS in 6m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9da0405f7fd149998da3ee93b232e527 : SUCCESS in 11m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/18b2037ac2cc41f688e901051d2d54bf : SUCCESS in 17m 19s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7d0861f68b1648d68d9f29cdf154d66b : SUCCESS in 55m 16s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/c5fe1dcd0db04b1b9b714d23a3f01ddf : SUCCESS in 53m 28s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/d1269d7593c640b5b2964723fa86928a : SUCCESS in 1h 14m 04s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/010d05ca0d0045c5b0da06cf188585b3 : SUCCESS in 51m 21s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2ab047aa9494b1f89158d5e0e6fae88 : SUCCESS in 48m 58s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ed2852446434db59f6917c55d383bbf : SUCCESS in 57m 49s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0450be78cb0849d488fa7a525b0283fb : SUCCESS in 20m 45s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/34e28712a51a4f7db5972a661d28a891 : SUCCESS in 17m 01s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/b0bff705efdc41e984846345654a19f5 : SUCCESS in 23m 28s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/cf2779a450c54f8fb0e4a4f2a4afbb8c : SUCCESS in 20m 21s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e487b23d46149afa1932efe93b965de : SUCCESS in 15m 42s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7168f1d172ef4e52beff73a11ffda797 : SUCCESS in 13m 09s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/7c8e65be1d394a67b0a587c9a4929b64 : SUCCESS in 1h 02m 37s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d960911f1cf94dff87a2a2e9623011fe : SUCCESS in 49m 35s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ae7b70f1d33d430388ca1276cf8f064f : SUCCESS in 23m 36s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/aad4213741c040e79b0bc20cf70b3269 : SUCCESS in 21m 28s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e432e830529f45a995097d0a0cff0e86 : SUCCESS in 20m 24s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bcf7f1b7249a43678a9cda34c3a0cccc : SUCCESS in 15m 16s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e3bdd81aec649ec88c7cae08b01d783 : SUCCESS in 18m 54s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e5821d77a8ac49f0b4be39c7c3c4c6db : SUCCESS in 16m 23s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/2933d291d9144001b3651dce05a121a6 : SUCCESS in 17m 08s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ddc6e1921ccf4bc5b4b80e2cc2a7ec96 : SUCCESS in 19m 08s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18a34397c98746f0ac5cd20ed4409c59 : SUCCESS in 14m 03s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d985737f96c34cd68af31c7e5c7fda9b : SUCCESS in 14m 23s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/79cdf211c8fc4b58bc0cf8fbddf2b7a0 : SUCCESS in 18m 36s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/0dd96f1fe5a343aeb95a5390ec9466ef : SUCCESS in 21m 05s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fba13032689048d3b1799435e05b0d5b : SUCCESS in 19m 42s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b2f735fe86a40c28682850e1555d3b8 : SUCCESS in 16m 07s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/373bce5aa7054f7da8a44f5eba987c54 : SUCCESS in 14m 23s","accounts_in_message":[],"_revision_number":2},{"id":"1771ba694f21b26a8b99e912fe4fca0aa877f2dd","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-10-23 20:43:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"57854549e4c99c15ee0797878181f079af87c55a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 20:43:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5a1b7ca61aeb3455a139a481301d748e8a97d6ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 22:42:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/529f2033ab6346f4986c17780409c12f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfd8f8da4b974e9f9b2dbc1765d8880a : SUCCESS in 7m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a2d737617484bee9d9ec98ec461d3f8 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/145ff5a4e45f4d12b01eef28df4431e9 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49e0aa77b6034d2d89d0d3ac993abcb0 : SUCCESS in 8m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52227a83b8a74f1f840dfe3cac0cc971 : SUCCESS in 2m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8ebe3b9b42e3468ca12d84d56aa245c8 : SUCCESS in 5m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7148ff1b0cd04f439fb9117329883d7f : SUCCESS in 13m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/49c26229d8a245a88d763532323c0982 : SUCCESS in 17m 38s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a902f65e95de4cb09a91421f790cb8a2 : SUCCESS in 56m 47s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/8232a16fc9794fe1aa0af4aed150260a : SUCCESS in 1h 11m 03s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/1be0bfe8a7f1479aa27cdafb678669fb : SUCCESS in 53m 32s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d76ba838a5884faa815242e737cb1d94 : SUCCESS in 47m 24s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d9d54b3ad56c4d309812d6cb1ba022dc : SUCCESS in 48m 30s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5922ad914f7d4e558fb00ee14d9b2110 : SUCCESS in 23m 37s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/e1361788dcf34e9ba376949a1813dfb9 : SUCCESS in 22m 33s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4911140831824b86b608447c5a3e1ae3 : SUCCESS in 22m 48s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff7ffd5db681441197c3c010a1e74980 : SUCCESS in 15m 51s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c6f84a539c64af497eff0600a2715f2 : SUCCESS in 17m 14s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/07a5b3fa4b0648f984732cd6a0cd68e1 : SUCCESS in 1h 16m 45s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/7d0ece4a3a564fd39ea1a952cf0fdbcd : SUCCESS in 1h 00m 35s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b64757d9faa94b6b957edf3367ccbc61 : SUCCESS in 22m 48s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/6a12064cc1db4e78ad7c0034e49924f0 : SUCCESS in 21m 58s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/bf677a7fe4ec492a88cbc01892b23f38 : SUCCESS in 20m 59s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21c326f81ec04817a5ea73224ff0e3bc : SUCCESS in 18m 08s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/295c70096e394f6796005f47fb1d6ebf : SUCCESS in 16m 49s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/7e2701cd0eb441b4ba59882baafed19d : SUCCESS in 18m 31s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/76e9dd95623047d8a335ddd14526bc48 : SUCCESS in 15m 51s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8d70f60df65944fc9bd2469873b23a49 : SUCCESS in 15m 55s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c3f5f4e30ad4731933b054fa28e3667 : SUCCESS in 13m 44s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e161cdef2cb4b548b520f21dece7602 : SUCCESS in 15m 45s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/1a297c00b46e472eae7000dcbc1717e7 : SUCCESS in 21m 37s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/c4e789660c9f4a4ea3af20ce0a90d793 : SUCCESS in 23m 09s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/14aa1583bd9040d79858b614457cdcf9 : SUCCESS in 16m 32s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9075282e99794417bfff9e7a6efd398d : SUCCESS in 16m 12s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/811ec4c8ce2047bb898390e9704944cb : SUCCESS in 14m 06s","accounts_in_message":[],"_revision_number":2},{"id":"1ca6a7d2706e5eebb49e2fe48293b501f096170d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 22:42:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"906d8c699ddca2d0c9eff7029728604cda638e66","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 22:43:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/903dd8c319fd470285a939d972d39171\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64e575c42a08404a8f9065650fc213b6 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a71472449ac49e787bc41745cf4d7c8 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b0832bf1ffe4a13a90b62a87a82d918c532f336a","revisions":{"d190e6504ee7f5f3a754ab8cd6ae10c26087d1d3":{"kind":"REWORK","_number":1,"created":"2023-10-20 09:58:12.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/16/898916/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/898916/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/898916/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/898916/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/898916/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/898916/1"}}},"commit":{"parents":[{"commit":"0ff603c18f2e8d86e9b694ea99c1c581b3faca02","subject":"Merge \"bifrost: Populate bifrost host vars on deprovision\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0ff603c18f2e8d86e9b694ea99c1c581b3faca02"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-20 09:56:08.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-20 09:58:01.000000000","tz":120},"subject":"Bump MichaelRigart.interfaces to v1.14.3","message":"Bump MichaelRigart.interfaces to v1.14.3\n\nThis release includes the following fixes:\n\n* Remove nmconnection files generated by cloud-init\n* Remove ens3 also from NetworkManager config\n\nDownload from Git repository until Galaxy import issues are resolved.\n\nChange-Id: Id3e7913f4ba9d05cec332dab560fc42d44dd4130\n(cherry picked from commit 66e7a7f2ee38fcbed9142941b22e2354e2f78c42)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d190e6504ee7f5f3a754ab8cd6ae10c26087d1d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d190e6504ee7f5f3a754ab8cd6ae10c26087d1d3"}]},"branch":"refs/heads/stable/yoga"},"b0832bf1ffe4a13a90b62a87a82d918c532f336a":{"kind":"REWORK","_number":2,"created":"2023-10-20 20:30:41.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/16/898916/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/898916/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/898916/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/898916/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/898916/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/898916/2"}}},"commit":{"parents":[{"commit":"c6d0829576322b74ec812922bf4e3786c26369e3","subject":"Add option to skip kolla docker registry login","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c6d0829576322b74ec812922bf4e3786c26369e3"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-20 12:40:59.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-20 20:30:30.000000000","tz":120},"subject":"Bump MichaelRigart.interfaces to v1.14.4","message":"Bump MichaelRigart.interfaces to v1.14.4\n\nThis release includes the following fixes:\n\n* Remove nmconnection files generated by cloud-init\n* Remove ens3 also from NetworkManager config\n\nCloses-Bug: #2039975\nChange-Id: Id3e7913f4ba9d05cec332dab560fc42d44dd4130\n(cherry picked from commit d4502e6ea268bebce926dcf9320b740272e83dfa)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b0832bf1ffe4a13a90b62a87a82d918c532f336a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b0832bf1ffe4a13a90b62a87a82d918c532f336a"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
