)]}'
{"id":"openstack%2Fkayobe~898951","triplet_id":"openstack%2Fkayobe~stable%2F2023.1~I23366f4de7842e7c2fe40e431fac76f26e9892de","project":"openstack/kayobe","branch":"stable/2023.1","topic":"nm-bridge-vlan-mtu-workaround-stable/2023.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-10-23 16:42:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I23366f4de7842e7c2fe40e431fac76f26e9892de","subject":"Fix MTU of NetworkManager bridge VLAN interfaces","status":"MERGED","created":"2023-10-20 20:32:15.000000000","updated":"2023-10-23 16:43:35.000000000","submitted":"2023-10-23 16:42:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898951-nm-bridge-vlan-mtu-workaround-stable/2023.1","cherry_pick_of_change":898905,"cherry_pick_of_patch_set":5,"meta_rev_id":"8f6bd4d3455853d1003d468985d27cc680b91248","_number":898951,"virtual_id_number":898951,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-23 16:42:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-10-23 14:51:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-10-20 23:34:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-10-23 15:02:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-10-23 15:02:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-20 22:07:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-20 23:34:31.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-10-23 14:51:49.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-10-23 15:02:26.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"23e7fc0b124c60837ad42c56b75f4831196f7f55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-20 20:32:15.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"58e56fabf445d835b2ecc60a118f4b3d41f70b1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 22:07:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1646eb6aed9847d8ad949ea010d51a79\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/182abe0fde7241098a7cbdffc4f9bcf4 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6deb23bdc7a499081beab96fe252841 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f797380fa12a4408961f3b36c2f4e0ee : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a19bf1e2c124065ac09124cd4529296 : SUCCESS in 6m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0aa43b966bb34c7eb277df4885b53e9b : SUCCESS in 3m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/abe3eefcfa4641449f19a275331be8c8 : SUCCESS in 4m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7bd77166d4634a77b0e0e011eb6659e4 : SUCCESS in 12m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/13defa63024446c1b6ce45ea1d53c251 : SUCCESS in 12m 37s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/7873f92e133f4a9f8d1e7ea9a8928599 : SUCCESS in 55m 44s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/32f5bf82ef624191956efb7fce04659f : SUCCESS in 55m 51s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47702c89b01842d8a37d8afb1e5d6a36 : SUCCESS in 52m 48s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/88e33e93c2464a4bb91093a2d2e04134 : SUCCESS in 1h 00m 49s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/543baa103b6241f392ba4e40fdc25252 : SUCCESS in 17m 09s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/aa21592960af4316aa2d8907f11d0fd5 : SUCCESS in 15m 11s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a852a64a91844b20844be1134565a274 : SUCCESS in 13m 21s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7dc1bd61b2024c669f3dc24fc6d3eba3 : SUCCESS in 1h 23m 41s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07fd090bdbd74e4ea4bad496a8f0edb2 : SUCCESS in 1h 16m 50s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ffe3b11a29e34ee0a9b8d4fe2d7c229c : SUCCESS in 18m 44s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/60896247c2f14696a660878fa70c29e6 : SUCCESS in 14m 15s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/b808906247924fceae92ec98a7ebf19e : SUCCESS in 30m 53s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ba0ae46604b94edeaac7a55180edddc0 : SUCCESS in 30m 18s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4aa70571c1ef4d2699ebb3bc14b8f1c6 : SUCCESS in 21m 35s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bccf0bf00ced43dc8d81d87e28cd1939 : SUCCESS in 13m 46s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82233a8231ec41f6abf521d6edec8129 : SUCCESS in 11m 15s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b25ba0b2bcf84342bb515daa9f4e7371 : SUCCESS in 18m 16s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/49c4d7746c264ffd93a8d8087d734602 : SUCCESS in 14m 14s","accounts_in_message":[],"_revision_number":1},{"id":"d2f9501341ab61967b7cf3f66b7a0c111009d6f5","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-10-20 23:34:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2df0bf77a08f8b83d27fbe9231f8d94418f004ba","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-23 14:51:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b7d7ce3ee97826293e66b1c733813d23735f7b6f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-10-23 15:02:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"42df678c8a06dbef316139e9f6252838c3a18636","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 15:02:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"092717078ca694510ecc6f86718ac4cb95b990ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 16:42:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c0b27f09fbc4df296637e3fe4838f9d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a6f81fc8d774ad49f8451264a86aa0f : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be7b78d104d945baa9efb3d848c915f9 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a766d5457be74da79d8d64343fee1846 : SUCCESS in 8m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/081b904edb1a494aaa831b06fb4b8d3a : SUCCESS in 3m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6fce54b87857439ab401a34fe4b75e6c : SUCCESS in 4m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2dbb04e1c1cd438b8dc4578cb8014adb : SUCCESS in 11m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e5afa04d72264d028866b5dd7f81d326 : SUCCESS in 13m 57s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/12344e5c05684bd4a4183902380d4f31 : SUCCESS in 58m 31s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b51d6bb8a0b14f7b8d8040abb8ce65a5 : SUCCESS in 59m 52s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9b4e6e20b0df4df9aba7df92aa801b12 : SUCCESS in 1h 03m 56s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7102833d4e874be1bd1d580515891800 : SUCCESS in 13m 57s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e8f4ee18627f489786864974645c8a39 : SUCCESS in 11m 36s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5febd585482c49dbafb2079836534973 : SUCCESS in 1h 15m 20s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07217017ff794913a5a59532a9add351 : SUCCESS in 1h 22m 57s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a4bbbcc0a9114faeb550efcaea150d64 : SUCCESS in 17m 48s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/332e3e207f6647c3bff58600d882acba : SUCCESS in 11m 07s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c7012503a17f496d9a562bd468b9a21d : SUCCESS in 29m 09s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b1744c15fae744369d5700c54f1f7583 : SUCCESS in 20m 31s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/61af72e6bd534c6f8420f70d41adb457 : SUCCESS in 10m 35s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8cc52d5aad7e42faa0b26a80ab8004b1 : SUCCESS in 9m 19s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bb8ff20527ca4324bb5767697263e898 : SUCCESS in 16m 56s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/964732dcb2be42d5a73c372c4306c89c : SUCCESS in 14m 38s","accounts_in_message":[],"_revision_number":1},{"id":"c86043dfae596a29c01731fe1f2d64b54422fa0c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 16:42:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8f6bd4d3455853d1003d468985d27cc680b91248","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 16:43:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48737514e1784ac1a99885dfb4a12cd8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e91b188420a14217ae271472c5656057 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82e8cb470c674ccbb6d0465e4e3f1671 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6e2a551f8d342c2603da294cfa12ecb001a56240","revisions":{"6e2a551f8d342c2603da294cfa12ecb001a56240":{"kind":"REWORK","_number":1,"created":"2023-10-20 20:32:15.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/51/898951/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/51/898951/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/51/898951/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/51/898951/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/51/898951/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/51/898951/1"}}},"commit":{"parents":[{"commit":"8b433cc01de90b3d18582ec8a134d3e09551d1a8","subject":"Add option to skip kolla docker registry login","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b433cc01de90b3d18582ec8a134d3e09551d1a8"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-20 12:51:37.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-20 20:32:15.000000000","tz":0},"subject":"Fix MTU of NetworkManager bridge VLAN interfaces","message":"Fix MTU of NetworkManager bridge VLAN interfaces\n\nWe are seeing an issue on Rocky Linux 9 where VLAN interfaces on bridges\nhave their MTU set to MTU-1 after rebooting. This is likely related to\nhow MTU is set by NetworkManager for bridges: it first sets the MTU to\nMTU-1 before setting it to the correct value [1].\n\nWe can work around this issue by not explicitly setting the MTU\nconfiguration on the VLAN interface if it is the same as the parent\nbridge. It will be automatically inherited from the parent.\n\n[1] https://gitlab.freedesktop.org/NetworkManager/NetworkManager/-/commit/864872e9a8c694ee7472154bf45f869040310b71\n\nCloses-Bug: #2039947\nChange-Id: I23366f4de7842e7c2fe40e431fac76f26e9892de\n(cherry picked from commit ee63b3253dedaf62cc321f5b29080d7e7c459a99)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e2a551f8d342c2603da294cfa12ecb001a56240"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e2a551f8d342c2603da294cfa12ecb001a56240"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
