)]}'
{"id":"openstack%2Fkayobe~898952","triplet_id":"openstack%2Fkayobe~stable%2Fzed~I23366f4de7842e7c2fe40e431fac76f26e9892de","project":"openstack/kayobe","branch":"stable/zed","topic":"nm-bridge-vlan-mtu-workaround-stable/zed","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-10-23 16:42:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I23366f4de7842e7c2fe40e431fac76f26e9892de","subject":"Fix MTU of NetworkManager bridge VLAN interfaces","status":"MERGED","created":"2023-10-20 20:32:27.000000000","updated":"2023-10-23 16:43:56.000000000","submitted":"2023-10-23 16:42:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898952-nm-bridge-vlan-mtu-workaround-stable/zed","cherry_pick_of_change":898905,"cherry_pick_of_patch_set":5,"meta_rev_id":"7ccc8253ce16d7f6f0b4f4aae0f2c1942641fa86","_number":898952,"virtual_id_number":898952,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-23 16:42:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-10-23 14:51:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-10-20 23:34:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-10-23 15:02:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-10-23 15:02:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-20 22:04:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-20 23:34:23.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-10-23 14:51:42.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-10-23 15:02:15.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"bfa4a5a002d9669da25d0bb794ec75a6e8aac4e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-20 20:32:27.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"daabd0a844689a5b0d6cf5df5583e14a7dc2804d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 22:04:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4af76d39eb074af288137d77d4497a75\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/867b9b8413874191b4092a32d12bdef0 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/949e436bca8f4f409ede7d3811a9a8c7 : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b65e00ec3c94c4a81f0ddfb93794b2f : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b592c1a2c0714e8eb19020569be7afd8 : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5678a1ed8bfc497e83442cfbd92e2175 : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1679a37993424ef8a83fd967bf32385e : SUCCESS in 7m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b7e67150ff74fc1813ed08a64566178 : SUCCESS in 3m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/164a8a7740e642f3be4d6e74c88894bf : SUCCESS in 5m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5a29cb921e784532ab0d5747aba704c5 : SUCCESS in 10m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4de619a432ca4e18bca824805eccf8e1 : SUCCESS in 13m 30s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/759032fd2798471b9eed89f2f254e842 : SUCCESS in 51m 51s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/e7321b48a55a422f85aa08d73dd456a1 : SUCCESS in 46m 17s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6074293c6b5e4a0590ebd3e55d0e7030 : SUCCESS in 54m 44s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/457477a966bf4026bb82481f354fd9e6 : SUCCESS in 44m 50s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4123a12c057d45288234da66420b1117 : SUCCESS in 18m 32s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/17b19c5fd0ad4865936464f09347ab50 : SUCCESS in 18m 39s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52030a2c3a33464d875ccc38c889ecca : SUCCESS in 10m 33s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/812d266fa2e34f1484847ad0c0bbd1ea : SUCCESS in 1h 24m 51s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ad2d8e5590f84dc18a0e4e89b81a768d : SUCCESS in 16m 39s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c6c1f443cbd54ac4b29e1b1bf30a7247 : SUCCESS in 14m 58s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/9e4f82ad3d4e4196ae42341c69fada2d : SUCCESS in 32m 26s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0ff83e70c044362afd1f739538a8ee5 : SUCCESS in 26m 28s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/aa16c8956ee8412c9f32d1dcb9386add : SUCCESS in 12m 50s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c910fc75d62a4ccbbce617ff4ac45dbc : SUCCESS in 11m 50s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fb052f0270fe46d48d6c7ebc5b0eb4ae : SUCCESS in 16m 22s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5be3628b3e1c4a99bf89497154014fa2 : SUCCESS in 14m 02s","accounts_in_message":[],"_revision_number":1},{"id":"a04c380defc35f911f31ddb34931aacba07e7097","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-10-20 23:34:23.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4d1a72af12d7f1b98c7769a2b4d455a101848e4f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-23 14:51:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"44678f48f9c8109ede45a059e30f54274ffc7a7f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-10-23 15:02:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"70b41f02a5e57fedf3a17e0daef2171f34444406","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 15:02:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1657b30ff856c61d13a649f35c645eb6aadf2d78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 16:42:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56d9aafc0792455da0ca8cabdba5b3d4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7925281ffeed4a648993b3edb513064e : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16ee4688719045e691528aa6552119bb : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0d3383448d54bdc88423bfaf547ac2e : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cb45655738d4accbbad89cb74f446e7 : SUCCESS in 7m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e7cd770e9a84f11b61a890e2e881f2c : SUCCESS in 3m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7d899bc3c9b94a19bfce44ae5dd45289 : SUCCESS in 4m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7f858e4e414c4b3abbefde709263280e : SUCCESS in 9m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3a5f3c6beb034facb3fef36d3ad023ed : SUCCESS in 11m 38s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/19311adb2126421c99daa397fb491c7c : SUCCESS in 40m 10s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee4c4b8b8c064210b050ed95e4636f64 : SUCCESS in 37m 49s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/fbed24a883874d8bb74e7bd58cf37379 : SUCCESS in 41m 10s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/07835aa905ff4e2aba322aad62f625ca : SUCCESS in 13m 54s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0664fdfd703c4f2c8730b82626d5f667 : SUCCESS in 12m 56s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a5ec198c7f064448aad43c229aed4298 : SUCCESS in 1h 34m 57s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4d364d3f9f844897baf3767f7bb1a8f3 : SUCCESS in 17m 13s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/964cbec427d745dd8fdea42485bedc4f : SUCCESS in 11m 54s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9bea1e87839741248d17be50d7e65250 : SUCCESS in 19m 38s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6505349f7d3f43b48cfbd1c8cee8f6d1 : SUCCESS in 12m 57s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee1b55430d214c1491247c2e675d2d31 : SUCCESS in 9m 04s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ccbf0979d7f542d697569f841817696d : SUCCESS in 16m 42s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b729777fd3b483b88c6934dcc3cfcd7 : SUCCESS in 11m 00s","accounts_in_message":[],"_revision_number":1},{"id":"4d7358a851290c29de34b859622e83435c13b342","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 16:42:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7ccc8253ce16d7f6f0b4f4aae0f2c1942641fa86","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 16:43:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25087b22f5fa414bbc132653e30c60ce\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26865c3b020e478c8c4c01cb0ed4944a : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f324d4e0b30e480c9fbb9a4bbd144414 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"790427f8ff901611b841b3a6529daec97eeb0909","revisions":{"790427f8ff901611b841b3a6529daec97eeb0909":{"kind":"REWORK","_number":1,"created":"2023-10-20 20:32:27.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/52/898952/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/898952/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/898952/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/898952/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/898952/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/898952/1"}}},"commit":{"parents":[{"commit":"0638862a574613d1b6a87afb70d0d3ba985f24de","subject":"Add option to skip kolla docker registry login","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0638862a574613d1b6a87afb70d0d3ba985f24de"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-20 12:51:37.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-20 20:32:27.000000000","tz":0},"subject":"Fix MTU of NetworkManager bridge VLAN interfaces","message":"Fix MTU of NetworkManager bridge VLAN interfaces\n\nWe are seeing an issue on Rocky Linux 9 where VLAN interfaces on bridges\nhave their MTU set to MTU-1 after rebooting. This is likely related to\nhow MTU is set by NetworkManager for bridges: it first sets the MTU to\nMTU-1 before setting it to the correct value [1].\n\nWe can work around this issue by not explicitly setting the MTU\nconfiguration on the VLAN interface if it is the same as the parent\nbridge. It will be automatically inherited from the parent.\n\n[1] https://gitlab.freedesktop.org/NetworkManager/NetworkManager/-/commit/864872e9a8c694ee7472154bf45f869040310b71\n\nCloses-Bug: #2039947\nChange-Id: I23366f4de7842e7c2fe40e431fac76f26e9892de\n(cherry picked from commit ee63b3253dedaf62cc321f5b29080d7e7c459a99)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/790427f8ff901611b841b3a6529daec97eeb0909"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/790427f8ff901611b841b3a6529daec97eeb0909"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
