)]}'
{"id":"openstack%2Fkayobe~899601","triplet_id":"openstack%2Fkayobe~master~I12ad980ef6d30ea98966c0f4eaf047e8d57a40e7","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-11-01 17:11:51.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}},"hashtags":[],"change_id":"I12ad980ef6d30ea98966c0f4eaf047e8d57a40e7","subject":"Support building Neutron OVN agent image","status":"MERGED","created":"2023-10-30 15:10:18.000000000","updated":"2023-11-01 20:05:17.000000000","submitted":"2023-11-01 20:04:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"899601","meta_rev_id":"42b0e78243c226cc3c2a2a30ab5b24f66fe64757","_number":899601,"virtual_id_number":899601,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-01 20:04:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-10-30 15:27:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-11-01 17:11:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-11-01 17:11:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-30 15:27:27.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-10-30 18:11:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"78a1702ac436598c30016ab5f3a26a0cd18a442b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-10-30 15:10:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e202124b8ea861d83ab903e2cb767d55808f9662","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-30 15:27:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cc25751083dcded87b887df1fdfaab91e317b7dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 18:11:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/768cafc14e184c8b9b99c81a99490bd3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb2823921ed2496797fc649f7610d840 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/239ec91cbe964d8c98b9f4a88f2c2afe : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c771a16a44784ed8aab128b4c1c7403d : SUCCESS in 4m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/83e1d668cacd4205afa373cddcfc9a0d : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb860e9ed0f34a979d7ca143ddbf5226 : SUCCESS in 8m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40de829983d644358f66163e306c31df : SUCCESS in 3m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ed01df07451d48178b8c3f34d8a62322 : SUCCESS in 5m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ac72b091554640609080a8b0f7c8a007 : SUCCESS in 11m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b1e0ddfda67a4fb0b78f2c1c72fb0303 : SUCCESS in 10m 19s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ef97c8a8d9ab49d1bb372f85db2d1074 : SUCCESS in 1h 00m 55s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/993af923167944cd9a4c58c4431f6cdd : SUCCESS in 56m 35s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f160f9e5e6784c99b8c7bb89ce77a046 : SUCCESS in 52m 26s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/66c2158e3be74f79b6e45bb9e4ac2153 : SUCCESS in 1h 00m 01s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f131cc559a024241ba28ccfcca0e719b : SUCCESS in 15m 57s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/01f3aa6dfd644d20a951c4d93feecf2f : SUCCESS in 14m 14s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/36d20601dd5140189b4e633bb12e0919 : SUCCESS in 13m 42s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3ae9af4ed1bd47acb6c07f37bd603ec9 : SUCCESS in 2h 26m 45s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8fb83b97e5d41c3af8875fdb2985121 : SUCCESS in 2h 00m 47s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/61a23f5adb164d379b47bd44a19b2db3 : SUCCESS in 15m 02s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8cdef54fcdf4a72844aea6dc28644bc : SUCCESS in 12m 15s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/d5bc4ce5d382443dbd1d24d7048e95b7 : SUCCESS in 30m 35s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/18190fc6c7b044a890a6ab19409d8ffe : SUCCESS in 24m 40s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9d71f62694d44996bc0eada6dd4b310c : SUCCESS in 24m 20s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bb2625e37941445882f2a2354e5d4e31 : SUCCESS in 13m 24s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb665e7baa964b43a633cb08530dd3c4 : SUCCESS in 11m 31s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4c414635a2b1429fb8138ff4b1d19b78 : SUCCESS in 16m 01s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/12a6ce186e45429fab4a918fc6639032 : SUCCESS in 14m 29s","accounts_in_message":[],"_revision_number":1},{"id":"fc46d726619ecbca7c9e987e399cfd194bcddd4a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-01 17:11:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f957506560dc8e87ca44d9f9e836dd35bc85ccac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 17:12:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b367387e863fbde5c065182ae06fa9ecf33d154a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 20:04:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c21c3e1ecf14fe5a8005ee994ecfcf8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8d5886999514fb29ad8e125611f1251 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb419019038041cf952b9110d18adfa3 : SUCCESS in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/db1d32ff6c6a472da5169d5300641386 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7091bbc911f7408e9328e65526a41952 : SUCCESS in 7m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ede1f1d1b29845b0a2bb80a698b80115 : SUCCESS in 3m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/15526fa72bab4a909247e8527c2250ca : SUCCESS in 4m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/73bd41c56bff41108c7f442762ad431d : SUCCESS in 12m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9e88e9c123664a5481799bd20c1ce02b : SUCCESS in 10m 54s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/65fe6f6464c644d3a560222908a23520 : SUCCESS in 1h 06m 28s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d2541944a604013a3254c9b3db6bd70 : SUCCESS in 56m 38s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/cebe6cca55154bca8a2d412eef0ae10d : SUCCESS in 1h 41m 13s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5d04b14ae0c549f9843297df67caa5b4 : SUCCESS in 16m 27s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba3250a8d27e426a9875bb9d0fc64d04 : SUCCESS in 15m 02s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/91b54ceb8f4f44b68a2be2f822125d75 : SUCCESS in 2h 47m 10s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/856767bef2ab43a599134ce338d23a88 : SUCCESS in 2h 10m 18s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/dc2bee0e80714f38900ecee037411717 : SUCCESS in 14m 10s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9635f81809c04fe99d378b0ed36b8d24 : SUCCESS in 17m 38s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/bcb864c4cf97489fba2f76c3e0917145 : SUCCESS in 27m 23s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/91ebbd990f614b70842359800c945e74 : SUCCESS in 28m 01s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/971525d4a31d49fb8dc794f90f6c91dc : SUCCESS in 10m 33s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8629f85f84f9474d8bb764daf12e99bf : SUCCESS in 12m 02s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/db644838c17a4942b0b42b3912865222 : SUCCESS in 15m 56s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/31d396d331c346b6b91d355048d0406f : SUCCESS in 15m 08s","accounts_in_message":[],"_revision_number":1},{"id":"80d18241c12c3dc3a244b9c2a7a1ee2409baadbd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 20:04:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"42b0e78243c226cc3c2a2a30ab5b24f66fe64757","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 20:05:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/643f262d944943db99c16573f25d1e86\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8e3d1d35a7b491b90ead52df32fcb64 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f092ea8fdf4496095a60367ace9069e : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"82eba749de453d56a970d64b4f98a6848ce6e1cf","revisions":{"82eba749de453d56a970d64b4f98a6848ce6e1cf":{"kind":"REWORK","_number":1,"created":"2023-10-30 15:10:18.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/01/899601/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/899601/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/899601/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/899601/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/899601/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/899601/1"}}},"commit":{"parents":[{"commit":"bd9043dca0af94bd6ed83268ace613de91a3ee0e","subject":"Merge \"Remove unsupported kolla_docker_registry_insecure variable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bd9043dca0af94bd6ed83268ace613de91a3ee0e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-10-30 15:09:38.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-10-30 15:10:13.000000000","tz":0},"subject":"Support building Neutron OVN agent image","message":"Support building Neutron OVN agent image\n\nChange-Id: I12ad980ef6d30ea98966c0f4eaf047e8d57a40e7\nCloses-Bug: #2039558\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/82eba749de453d56a970d64b4f98a6848ce6e1cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/82eba749de453d56a970d64b4f98a6848ce6e1cf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
