)]}'
{"id":"openstack%2Fkayobe~900077","triplet_id":"openstack%2Fkayobe~stable%2F2023.1~I89041a641d8cb66c7848fa9ae9264a51aa1e38c8","project":"openstack/kayobe","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-11-07 14:23:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I89041a641d8cb66c7848fa9ae9264a51aa1e38c8","subject":"Add retries to get_url and uri tasks","status":"MERGED","created":"2023-11-06 10:33:51.000000000","updated":"2023-11-07 14:24:37.000000000","submitted":"2023-11-07 14:23:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900077","cherry_pick_of_change":880931,"cherry_pick_of_patch_set":1,"meta_rev_id":"9b9eee03f0427de6f46986468a1e682662c4a0a1","_number":900077,"virtual_id_number":900077,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-07 14:23:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2023-11-06 10:33:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-06 18:54:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-06 18:54:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-06 12:11:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-06 18:54:01.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"9459dc0e8afe6fb44ea722e42a7c45d07c888ade","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-06 10:33:51.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"cb09823a2676d7e1fd33f828815402ab6f9cfe5c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-06 10:33:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"96a74695b909077db06014fd92ea1d0050864234","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 12:11:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/327f3c7c886743b39c9341e5b0cb7969\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/efd8de200f0b454fa78a0b018cab7b9d : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/241eb19fbfee4b3c97890994a606b8ba : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ce3fb5da4484208bdfdd125c7190bc3 : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8838b127a2b6427a8eeca1d3bf8a19eb : SUCCESS in 7m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/161ce38b7a2c4045b2fcffb0cc3bf720 : SUCCESS in 4m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/43ea81165f784cdd98e4843e315890bd : SUCCESS in 12m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b3a4488ec5bc4aaa9d1092623165707c : SUCCESS in 14m 59s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/14f69a17a4a842cb87d2c3a4d01481e0 : SUCCESS in 1h 06m 36s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/0b4691f96f3b4be08ba7c955ad539a53 : SUCCESS in 54m 49s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9287e9b9fe234a3b888a163112b972fd : SUCCESS in 59m 07s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e00cb4d1e2894e29bd0483da559c74fc : SUCCESS in 53m 52s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e6b19b638c2543ddb65b31e479188a5a : SUCCESS in 16m 47s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/b834c65c900147d998b5c23a59932df1 : SUCCESS in 15m 23s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3ddbe072ab6453e85bd86271c088a7f : SUCCESS in 14m 35s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9e01c0fc61c24985ac8681e9ebd8c59b : SUCCESS in 1h 32m 58s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e7a4f861fd84150a7852a18ef8b7fa3 : SUCCESS in 1h 29m 58s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d0f19f2d5e3b4e59bbd2045ea4051a9b : SUCCESS in 13m 27s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c7883afc8a94874aeaf927587874015 : SUCCESS in 12m 44s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/c22c1b6206d840c399c6b613b5ddb5b1 : SUCCESS in 30m 08s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/bbd5fd93a3f74d2e9368157273d9e740 : SUCCESS in 30m 58s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fdce6d3ac3054e1e9f81c0bc7c973550 : SUCCESS in 25m 01s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c74a1c9fe7074a92867e8e2fde40f58c : SUCCESS in 10m 38s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a614857c73f489c8788874322d9f2ab : SUCCESS in 10m 32s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/93a27bdb82ff46bca48fb4ab5bfc8fb1 : SUCCESS in 16m 58s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82b1147d99214d4ea05525a5d81db0ab : SUCCESS in 14m 46s","accounts_in_message":[],"_revision_number":1},{"id":"47e0c9bd4ce510b244da640b69edbb6e53e90fa3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-11-06 18:54:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bae34a8b9de2359c17cf3c82b2d885d750b308da","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-11-06 18:54:13.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0bd197bd0c526b9138a1536cf00963be1a071379","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 18:54:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"26c0b85c452a7e8b9a4c531909adc34aa3e09baa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 20:54:51.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/323bd1926d1043ca93bd7beff2201361\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/503336dc058f4848ab7a2f0ea52ac0b8 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39cfd2ea4264462dad84ef3186bb8632 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c704b75db5ad45e899b0454f28e93591 : SUCCESS in 7m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f2d94d405bab4adab436a4aae7d37770 : SUCCESS in 4m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/68113f2cd14b48e1828198060fb447a9 : SUCCESS in 13m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c65ce0a46246414fb57deeb8717b9739 : SUCCESS in 16m 28s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a28821c0f6bc448bb7b28251af28665a : POST_FAILURE in 1h 07m 01s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78d9ff7c600a4f26948e01fbb0db251d : POST_FAILURE in 58m 45s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/bc2ea37ddb7d45d28ac5b30c14a8fb43 : SUCCESS in 1h 18m 10s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/2ac89e11f0ef41e293f5d4f85d665cde : POST_FAILURE in 14m 21s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8ca56a09494e4d1ab1f227f6c2cea904 : SUCCESS in 15m 19s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e7c1f3912c7d47cfb5fc835a48782a42 : SUCCESS in 1h 53m 13s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e92377c63ce4d8b8103086ec8577bfc : SUCCESS in 1h 38m 10s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/95345d2e22874884b189dcd5f4d76e83 : SUCCESS in 16m 36s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7f6c2a0354f4b3181f245c9f0fcbee1 : SUCCESS in 13m 06s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d6cf4715537b433e9b60a0a42e18bfe5 : POST_FAILURE in 34m 42s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/50c3821852b149e188a3fda24677db81 : SUCCESS in 23m 15s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/023387eda7924f27ad87f44f6b869fb6 : SUCCESS in 11m 07s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2e6f6ca12f9a448eb3be5ed857497e97 : SUCCESS in 11m 54s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e79b4c807bc3416c86272f9de4436e4b : SUCCESS in 19m 18s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b6a58ecd76324c6eab86e94ee8b479bf : SUCCESS in 14m 15s","accounts_in_message":[],"_revision_number":1},{"id":"8cd725d4b71172b8ae5d2d179141d96e6aa8b9ea","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-11-07 10:24:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2d9cf5809ff6fc6b8f169cb4765a9d6787c2bd20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 12:06:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da0bd23e8cb74822af9d28e3e8305afd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26968b32f3fb4c499826573cf10284d1 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c11a380e292046309861abb01f067c91 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30d69a0ad1974d248b67c41135f93c55 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c972b3be655142428d0c763a60592b36 : SUCCESS in 6m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3cd74c639a934427910b715ce9f560ff : SUCCESS in 3m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e27d35cc74da419bb104be6baca3674d : SUCCESS in 10m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1c5abb41efee4e6e8b9f111b87bfd089 : SUCCESS in 14m 02s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/08a46ade8cc04d4395cf5618df395219 : SUCCESS in 54m 35s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/814d0185353849ddb148e59aa497ccd7 : SUCCESS in 1h 04m 53s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/98f582d93e694938b5e79020b4884e96 : SUCCESS in 48m 54s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/167e1a37322641d98666a21e094feebb : SUCCESS in 1h 05m 47s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/43f7b999d00a4e6ab6841f7b3ebe7a69 : SUCCESS in 13m 48s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/786183171678473d9df1fd0c5ce398c2 : SUCCESS in 17m 06s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d97f59709f2544e78795e1093c449337 : SUCCESS in 13m 58s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/233ab9aefb034ee39e35e0ea4292656e : SUCCESS in 1h 37m 43s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f931b4648e364489a4e00a53c729c8a6 : SUCCESS in 1h 26m 20s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/693eb8eceaf342e5bc6f4656e5963087 : SUCCESS in 17m 01s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/224dd60ec2b14f42a34220375584d2f5 : SUCCESS in 13m 55s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/d307a90a34bc4d35813767467e464dc8 : SUCCESS in 29m 52s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/af6e4f30bb5f43ae96d7b1040a8300d4 : SUCCESS in 26m 41s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba10d2ba1ca94b7e84e2a01f81626bac : SUCCESS in 23m 32s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/080af830d58546f5874ad57de23dd4db : SUCCESS in 15m 07s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2d0fb849ac64476ad543995890e38e1 : SUCCESS in 9m 55s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2bb51ddd646e4a9eae19b5064a45a04e : SUCCESS in 14m 52s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6d261994517b4ca1b772c5215b37bd8a : SUCCESS in 13m 01s","accounts_in_message":[],"_revision_number":1},{"id":"d6e8205980e268a100907ec891edc643d4575fec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 12:07:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5645af8cbaff2c092d5b6f2c47e6f4bc3e1620fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 14:23:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84e2d3d403a244e7a4baf74b47e65d52\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73330f8ce12742bdbad6105db1a74e86 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a610bc935ecd48d897ab0342785afac0 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eebfceee1f1e4798a1a448e15a9ab131 : SUCCESS in 7m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/27c958ded74b471c8ef39dea7c647744 : SUCCESS in 4m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dd8c41215d2043a8ae87e51398d973a9 : SUCCESS in 12m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3b26a0dbfaf9487ba23e2a24ea55eba9 : SUCCESS in 13m 21s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/7cd427a7385c47a298d3dda6146c09bd : SUCCESS in 1h 08m 44s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/41fa836418b94fb692bf183984e6934f : SUCCESS in 53m 13s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/4d8774dc6b004e569f26e473f324b381 : SUCCESS in 1h 11m 20s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/44d03e3198924bf7afad6d8a66bdf63f : SUCCESS in 17m 05s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d72cf934cfe4c75aa3c843f61e9eef9 : SUCCESS in 13m 54s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5661e8c022f448ce9e364a70ca62f0fa : SUCCESS in 1h 50m 17s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/402de453e41f4275a83ab05d8a885c42 : SUCCESS in 1h 39m 32s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4e0442fc0eb64df2b521c2865164ddcd : SUCCESS in 15m 04s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4cc23e250fa64c8baca90d91ef5f24ff : SUCCESS in 16m 00s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/09ca20a443d24bfba6a256492515d84f : SUCCESS in 28m 12s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a13e09aae8d407f89f11839299031ad : SUCCESS in 22m 39s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2bb0387ec3fe45b3ae6642584b80c98f : SUCCESS in 10m 28s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/73ef1286b3f841eea1614ee464d47779 : SUCCESS in 11m 11s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/adfe6eed799d44c1b5d2fc5d8680cf09 : SUCCESS in 17m 43s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c5657a0708245de8aa4d53b63d2e59c : SUCCESS in 16m 17s","accounts_in_message":[],"_revision_number":1},{"id":"073487ce3b8605c9b01ef617fd44512f0f0ed3fd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 14:23:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9b9eee03f0427de6f46986468a1e682662c4a0a1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 14:24:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7370b9ed9a9449829f3f9d4c2937ac39\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef852cece673419a8aeee58ddb3be135 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"186936afaa505b9047006894b821cea932566804","revisions":{"186936afaa505b9047006894b821cea932566804":{"kind":"REWORK","_number":1,"created":"2023-11-06 10:33:51.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/900077/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/900077/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/900077/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/900077/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/900077/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/900077/1"}}},"commit":{"parents":[{"commit":"0c6c7770400127d5d4c02144c76e5695f6422ef0","subject":"Merge \"Resolve slow fact gathering when seed SNAT is off\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0c6c7770400127d5d4c02144c76e5695f6422ef0"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-04-20 09:25:36.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-11-06 10:33:51.000000000","tz":0},"subject":"Add retries to get_url and uri tasks","message":"Add retries to get_url and uri tasks\n\nThese tasks occasionally fail in CI, causing jobs to fail.\n\nChange-Id: I89041a641d8cb66c7848fa9ae9264a51aa1e38c8\n(cherry picked from commit 2951f26ad1d6d79671f5f94016468af3c43e6143)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/186936afaa505b9047006894b821cea932566804"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/186936afaa505b9047006894b821cea932566804"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
