)]}'
{"id":"openstack%2Fkayobe~900461","triplet_id":"openstack%2Fkayobe~stable%2F2023.1~I4813a1f6ab9cb566596e806bd0ada6dff342d119","project":"openstack/kayobe","branch":"stable/2023.1","topic":"kolla-passwords-overrides-issue","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-11-15 21:26:10.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-11-10 10:15:58.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}},"hashtags":[],"change_id":"I4813a1f6ab9cb566596e806bd0ada6dff342d119","subject":"Fix an issue when user forgot combine custom passwords","status":"MERGED","created":"2023-11-08 20:20:14.000000000","updated":"2023-11-16 01:23:47.000000000","submitted":"2023-11-16 01:22:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900461-kolla-passwords-overrides-issue","cherry_pick_of_change":861397,"cherry_pick_of_patch_set":9,"meta_rev_id":"8d9c247c48caa4fa3f053fd341146170396212ce","_number":900461,"virtual_id_number":900461,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-16 01:22:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2023-11-15 14:38:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2023-11-10 10:15:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-09 10:11:10.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2023-11-15 14:38:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-08 20:51:54.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-11-08 22:21:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-09 10:11:10.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2023-11-15 14:38:36.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"02188bc3c0a86f32f496a655cd511071dd78d63f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-08 20:20:14.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"cb2c3ffb70cfac5e002b4fb6856ae90ff58fee4f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-08 20:21:33.000000000","message":"Topic kolla-passwords-overrides-issue-stable/2023.1 removed","accounts_in_message":[],"_revision_number":1},{"id":"df2e0fc67c4094ca652fb7586cda3cd5206649e7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-08 20:21:41.000000000","message":"Topic set to kolla-passwords-overrides-issue","accounts_in_message":[],"_revision_number":1},{"id":"643f9889286c6cefc77bdb6e426c5face6efda64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 22:21:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/22b0d0931e8247aaba742ff719120bfb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a277646a629f42df88912632c829b954 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fe37f329076441b865c85a401707716 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e51a7a8cac0744d8855ab3ca12068a97 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cddee3c1a4d4724aa477571efb450ef : SUCCESS in 6m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f85bdf7423b463fb1d2b60ccbaf9080 : SUCCESS in 3m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/60c320a15beb45b0ba45cf32a81f4c18 : SUCCESS in 4m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5a63666d61f84031ae084882fe62f2d6 : SUCCESS in 10m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/30a9f5d8f3a74f31ba4aa343c9a82017 : SUCCESS in 18m 44s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a4f9545d92374ab9ad7fd6daf478b013 : SUCCESS in 59m 59s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/ebca818fd9834e5e9f3c3226e6047f2f : SUCCESS in 56m 39s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/90b60e6a2232405ea39a12df34248a9d : SUCCESS in 55m 24s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/087d7aba00da461faee13fd390eed1b3 : SUCCESS in 1h 20m 53s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6b0f18f80e7c4546afaf9eafff4ba342 : SUCCESS in 16m 20s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/ae382499f45346c7b5060f22ee1a4f85 : SUCCESS in 17m 12s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/08f992ffd86746dca5f8fb5c6c28dc04 : SUCCESS in 13m 25s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/938a957b437f4a3e9d04976ce6867d3e : SUCCESS in 1h 56m 14s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6fd50e4e321440deb8105931d6d89684 : FAILURE in 53m 06s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/43dde26878b54ce6bda3cd64c69f5b65 : SUCCESS in 17m 42s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b15dd4be62c84beb8a33cf82007ba496 : SUCCESS in 14m 57s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/92f8c82cf40e43c1869be5f75e575ebd : SUCCESS in 31m 14s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/493d0a0815f54317905342aadbd63c57 : SUCCESS in 25m 39s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2036b1a284e346e5b6746986850df4e5 : SUCCESS in 24m 44s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6249d545cac145c39f7148d08296d184 : SUCCESS in 14m 09s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/45d8de5a4eef4aa48357aa7d452353ae : SUCCESS in 13m 44s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/063fda86ab2045648bd2cb952b219f82 : SUCCESS in 17m 49s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47696d670c2544e1afa390f0c1e98f66 : SUCCESS in 16m 27s","accounts_in_message":[],"_revision_number":1},{"id":"7f840311e299c99cf29f3da37d03f25a6b98fdf7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-08 22:42:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"19c0e6a7b7a8f60115a631d31b691a0615904efa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 00:30:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5d8806805a14c9e91bb55adc837bcd6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/261b742cffe4431c80b5974aefc30393 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/964649ac13794563a3e303e17e176572 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79de103a7c5e465d8343bc15dbe5f750 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbec10fd5f9c430587e27895d460abd3 : SUCCESS in 7m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec41524ece7949dc9ba6f84bde3883a3 : SUCCESS in 4m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3ab41cd8ace2401681176d83fcaf6c95 : SUCCESS in 4m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4b3b10ab2f7346e0b5d5d10640b8be96 : SUCCESS in 11m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/869a4815fa0b4ba0b620d17c91cf924c : SUCCESS in 39m 29s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/00bbdb2f95354fb68301fe6d34760a58 : SUCCESS in 53m 50s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/84d89bf963bc4d9aa471b47a8012713e : SUCCESS in 52m 55s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da1f596027ad4116b822cda4646a3f92 : SUCCESS in 54m 46s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/006e5d68c0f64c6a8e9a1e6e5d77e35d : SUCCESS in 1h 17m 06s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/439c6c293c7045e6bcd7d1505b512ffa : SUCCESS in 15m 33s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/6baf98d55e634b5d8ba9ad198b34390b : FAILURE in 14m 20s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c2be84945e34ad4bf23f20fac154460 : SUCCESS in 12m 27s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/25663e3916674542945e6a90bfc85c7e : SUCCESS in 1h 39m 29s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b720101f18fd46a58ca879d9ccbddaff : SUCCESS in 1h 07m 54s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/056acb02892040b783858346b9cc4370 : SUCCESS in 16m 16s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/95ee660e9fde4a9490b943be1c436990 : SUCCESS in 12m 37s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/1a9902ecc69b41dabf053aa443c4fc09 : SUCCESS in 28m 43s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/52f31db42ab44e449e99f4acdd406229 : SUCCESS in 28m 29s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/54a210538eab4070ab0d4c73e43f65ea : SUCCESS in 23m 14s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e1bada18e5094d938ba230cab7fdb45b : SUCCESS in 14m 51s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3550f546e0f64d3aa59d5c0ae8cefeb3 : SUCCESS in 11m 04s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4b316ac6a7d1410f8f323424e1e730f6 : SUCCESS in 15m 29s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c951a6728081424783625d896d4ed7bc : SUCCESS in 12m 37s","accounts_in_message":[],"_revision_number":1},{"id":"c899c527f4351296036d3fff172db2d6bbb8cee2","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-11-09 10:11:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"71dc1ddf9f9de2ac90d192eb276f6b4c7e3961a7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-10 10:15:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"033a43c31eb8e2b7380d3e4fa16d2fe8c192fc1a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-11-15 14:38:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4bee61af35633b46151d76feb12619d74538a390","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 14:39:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5a042ecbbba120f05514ba38886ad202ca2ad6ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 18:11:59.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6f6be5e6ab4d4763b994efd522bf6a07\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/812c04badfeb4f8090486a6196b42de0 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8af07e7feb5146da932f9b1e016e468f : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30308671fbe44d7ca39e4a1b33de5541 : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92e34bde22f445a28e521226abca3542 : SUCCESS in 3m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/711c933f85c647848a4954aeb46eaaed : SUCCESS in 4m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/12ea26c24cf94bdcad375a50638ff930 : SUCCESS in 11m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0cf4d405422d47a68d018b18badb1bd5 : SUCCESS in 14m 40s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/336fcf837dad4912b888b6fbe0c5fca4 : SUCCESS in 1h 03m 42s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a8f372233e84fdfa5be05cd90d54ac1 : SUCCESS in 49m 49s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/f9e7156e4a64432b9ebfac38863ddd92 : SUCCESS in 1h 05m 11s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b19fd03121b74a47ae3c6b430df68d64 : FAILURE in 17m 07s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/49a175c49d7346028f73bd742ce1fcce : SUCCESS in 11m 51s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b81bdb1e1a3542ff9a5d376d7ef434a7 : SUCCESS in 1h 43m 40s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/809d06c63eb84add83579e197de9584e : SUCCESS in 1h 31m 31s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c29a114465ec41f3a3debe77453ec602 : SUCCESS in 16m 08s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f1525072ecb74d4e9ac09851c7dcacb0 : SUCCESS in 15m 00s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e195a9c9ff7540ff845ad60b98498f65 : SUCCESS in 32m 16s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87f2e7cb59794a118291f581eb0a9148 : SUCCESS in 23m 17s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7e6a1b861017464dbb37a692aec09862 : SUCCESS in 12m 49s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e20c5bdb22be471caeaa2d3a82ac4962 : SUCCESS in 10m 17s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/132a85214626495bb0a3b9da951cc7ce : SUCCESS in 15m 48s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ab7876568384ddda47ef77214ffdd8f : SUCCESS in 16m 03s","accounts_in_message":[],"_revision_number":1},{"id":"0b7f8105d06feb1f2f01054dd34d89e947908099","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-15 18:59:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5719c4691d62bd0126574ae44adc93ca990ef060","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 20:54:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/92440650de774c0083ebde217944227f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9c87bc107b64fc382bb6d271df0ef90 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/223e38c4553148c587628aa28c32903e : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa743fb9d3c34a0fb0b6ae499673d309 : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53d2cebcc7e3417eb11d307ffe3e606a : SUCCESS in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02a0755316ca488b867849652fb45b7e : SUCCESS in 3m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/366a26cb657945f8b54301296593567c : SUCCESS in 4m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0977736c610a48218380336f468eaab3 : SUCCESS in 12m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3d40c84a9ba7477dbb3a6d249a7871fa : SUCCESS in 11m 37s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/355f83f455ba45eb9bb088faf5fac5ab : SUCCESS in 1h 10m 05s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/5cd4100bb27e4cb9ab58700d90116eb4 : SUCCESS in 53m 21s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d78acd8eee24f85bd5e0b05014ad149 : SUCCESS in 45m 54s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d57a115c711749d8a3f582154f33c7b9 : FAILURE in 12m 52s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3ab028089dbd42e1955d42b0d8c526a7 : SUCCESS in 17m 20s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/49f163850afe4c41b0c27a92c209d164 : SUCCESS in 16m 54s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d6513392e11410cbc3b07a4e7f5ae84 : SUCCESS in 15m 07s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9d1214ad279944d182bc538c62ff7189 : SUCCESS in 1h 49m 41s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/75c5dd5b248a4744bc1ed2ec635c05d8 : SUCCESS in 1h 39m 14s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f56fdb3eed614cc990b52e32c9c2451a : FAILURE in 14m 05s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c0abcd6015d94da9ba514e0a42f7fe95 : SUCCESS in 14m 50s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/30fc9d21d481421fa1324b5dd8d0b7a5 : FAILURE in 13m 44s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/cd8f3a78e42d4a75a096b4d22c518a26 : SUCCESS in 28m 23s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3829dcf548004c4d8e2be16f3a29515d : SUCCESS in 26m 24s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ebee3b79f97445ce8b4bedff34a82e79 : SUCCESS in 12m 10s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33920569c9a545c297346275fd4bc746 : SUCCESS in 10m 45s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ade925d5c3b748f3a3000124817b8750 : SUCCESS in 19m 39s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e9b452f5a3c40b0a7faaa237a15024d : SUCCESS in 15m 21s","accounts_in_message":[],"_revision_number":1},{"id":"ef199a0341bf80c822fbac4bf68ad9099bf281af","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-15 21:26:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"71ddd6c26f5e5aaf48b2530336d91fa2e67158e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 23:35:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ac424a3a81947af821fc13f71036483\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70830774dbb64f34982c0b6192b49441 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b08aa0bd667746a69b1c8cab1d0f7565 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d74ac30996484a39ade58447ff4caf07 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecba30290293419f9923de40e73aa6f6 : SUCCESS in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09a0420ca73f4ad7b0c79369fb432085 : SUCCESS in 3m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d5776c003fa34a6f844761dd72392041 : SUCCESS in 4m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cff24a758f9c41ee81d382e74b45de68 : SUCCESS in 12m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a4b70d0cf04345d38461dd439fff4ff1 : SUCCESS in 13m 58s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/17789603704a43af9571a7601a9f1122 : SUCCESS in 1h 08m 54s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/90e07a7fbf624643a570c3c65b6cfd74 : SUCCESS in 51m 44s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7f9effd021584cb2ab3714663ffd59a3 : SUCCESS in 46m 20s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/1ce263a141254c7db4083022a864636b : SUCCESS in 1h 19m 06s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/70127579fd6c42c2bfe33f8fb7143bea : SUCCESS in 15m 12s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/22a408e05ae044288a65ffad9cd410a9 : SUCCESS in 16m 38s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/37cc3c52dd484fe68143de9878912b52 : SUCCESS in 14m 43s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9a35fe409ff342e19a3d5da87404323b : SUCCESS in 2h 02m 47s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce17ceb79d474f1b953a874c2da831d4 : SUCCESS in 1h 23m 48s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f7c971324af5436babf93dde2f37413d : SUCCESS in 17m 30s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ad317deb43149cfa88b31897f0d75df : SUCCESS in 14m 17s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/6b13d95ed4c54d498f4a7801725990a9 : SUCCESS in 33m 02s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/af3deb4de7b14e15b4aa2b8d2e26631e : SUCCESS in 27m 03s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d874064d5d1485ea091bf0948845b14 : SUCCESS in 22m 15s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3903f28726cc47ce9058aada582dd9c1 : SUCCESS in 10m 19s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a9069a82e6dc41439be0ba63d0dfee42 : SUCCESS in 13m 13s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/461bd12edf5346bebe7591764cf459f7 : SUCCESS in 17m 32s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/21ec3d1fa888486a8cb922817e4188e8 : SUCCESS in 15m 31s","accounts_in_message":[],"_revision_number":1},{"id":"df1a89ee63e92829f4428a5188e797439bd04a63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 23:36:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a2f232cd99a04b596057a2a72d4e23abd60ff5b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 01:22:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c73f3da848154d90bd905af5e8462264\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f419114d8aa7436fbfaa600de3cce137 : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0335c8b73a264d3a90529be96d7bfff1 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9dcae8a235c423cbd2a3c9c90d19d2a : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e68f4cbfdcc438185ddd3eeb5680ca9 : SUCCESS in 3m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3893c1f9c474483f80e5c3f45d932c03 : SUCCESS in 4m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/929f70764ec3452b93545ef03e679e91 : SUCCESS in 10m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bced5274b04e46369e62e56b69e1c7c1 : SUCCESS in 16m 43s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/3e13536151aa4a2fbee1c38d20c7f61b : SUCCESS in 59m 11s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/95cd9d0f32e94242875ea9470f0a7053 : SUCCESS in 59m 00s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/f71c4edcc66445a6b4c9f836dab9928f : SUCCESS in 1h 16m 33s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5ceb9a9270c148d99879a4c8a7e4ddae : SUCCESS in 16m 56s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e32d9762ea434c178c484b2615d62bcc : SUCCESS in 12m 09s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ab4d9de4cd0147888753dbd032668ccb : SUCCESS in 1h 19m 05s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/802fe8d6cebf494987c4b3182ef4ba2d : SUCCESS in 1h 39m 46s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c060d4ebe4a14f2197f2031523dd0697 : SUCCESS in 15m 47s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b1993268c06c48f0af43bb2c35d7fd78 : SUCCESS in 11m 47s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ae592e89c2dd460c8e86e7965bc2ffeb : SUCCESS in 23m 20s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/740231169b1f4f5ba09890094c23d709 : SUCCESS in 19m 51s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2758a46b11bb4dcb9747d77c05373726 : SUCCESS in 9m 26s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dac6f285725248cbb69910116d0857d3 : SUCCESS in 11m 46s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e4551112837d4afb8c7cc52a6356c38d : SUCCESS in 18m 18s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/106437c0ebd7446a89e5a82fad63cb04 : SUCCESS in 11m 37s","accounts_in_message":[],"_revision_number":1},{"id":"7ea59cd60ece32449777cee1db9d465a3cfb59bf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 01:22:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8d9c247c48caa4fa3f053fd341146170396212ce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 01:23:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/463598da94ef4f8c9db0d81fc6db1794\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4172b175af424e248599e32b0ede6dc4 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/481c556ce9064b0897819748633a6658 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"17c0523edaa2c72e97ea1c956c94ff23782f36c1","revisions":{"17c0523edaa2c72e97ea1c956c94ff23782f36c1":{"kind":"REWORK","_number":1,"created":"2023-11-08 20:20:14.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/61/900461/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/61/900461/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/900461/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/900461/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/900461/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/61/900461/1"}}},"commit":{"parents":[{"commit":"ef5f1c97ad78170233ed18ff147ab2e1647fdb71","subject":"Merge \"Add custom user_data content configuration for Bifrost\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ef5f1c97ad78170233ed18ff147ab2e1647fdb71"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-10-14 14:56:20.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-11-08 20:20:14.000000000","tz":0},"subject":"Fix an issue when user forgot combine custom passwords","message":"Fix an issue when user forgot combine custom passwords\n\nFixes an issue when user forgot to combine\n\u0027kolla_ansible_custom_passwords\u0027,\n\u0027kolla_ansible_default_custom_passwords\u0027 and own dictionary with\ncustom passwords in configuration files. Now\n\u0027kolla_ansible_extra_custom_passwords\u0027 should provide only user\ncustom passwords to add or override in the passwords.yml.\n\nChange-Id: I4813a1f6ab9cb566596e806bd0ada6dff342d119\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 73abf0e57ee3aaeefa53eaf4fb7b1ed5f5c96933)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/17c0523edaa2c72e97ea1c956c94ff23782f36c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/17c0523edaa2c72e97ea1c956c94ff23782f36c1"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
