)]}'
{"id":"openstack%2Fkayobe~900585","triplet_id":"openstack%2Fkayobe~stable%2Fzed~I9e1cc20579cf80525d6ef732a1aac99a65bc171b","project":"openstack/kayobe","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-11-14 11:46:10.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}},"hashtags":[],"change_id":"I9e1cc20579cf80525d6ef732a1aac99a65bc171b","subject":"Fix setting kolla_admin_openrc_cacert","status":"MERGED","created":"2023-11-10 10:06:11.000000000","updated":"2023-11-14 14:30:49.000000000","submitted":"2023-11-14 14:29:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900585","cherry_pick_of_change":868199,"cherry_pick_of_patch_set":2,"meta_rev_id":"6594cbef12c517df44f3813ab989f684a1034e07","_number":900585,"virtual_id_number":900585,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-14 14:29:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2023-11-10 10:10:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2023-11-10 10:12:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2023-11-10 10:12:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-10 10:10:57.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2023-11-10 11:48:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7c23248566a76620b742605c7b271660600ffe19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-10 10:06:11.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"989cf3fecd9c93f651b00e15aa4b6b8a60768746","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-11-10 10:10:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c9e32aa60fddd826a207a087b5c26447c6eea556","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-10 10:12:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e1be4d854f771a4be60e01e7848cd3676bdd8a01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 11:48:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ee9617b095e480fb0dcc37bed8d9f78\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1d2f9a0ff6842d99d1a893110b5eebc : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87fc72ea81d543ffbfa058eb8cfb8094 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5c62c7221e9436bb8d73c4a20e3591d : SUCCESS in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/daaed702053c44cd8698ee121adee8b8 : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/85af79578b5e40fc9f8d7487a9c4c03b : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27b9935d47494496815cbff1853d9ce4 : SUCCESS in 6m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb868033b51e4b92ab3ce988f0a4647e : SUCCESS in 3m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f2ec914c548e4552a81c43de81967bbd : SUCCESS in 4m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bf44811c9a48444ea1138c26cbb462aa : SUCCESS in 10m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/32989f7f19c84cee83f3b7fe19c7ebe9 : SUCCESS in 15m 08s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f60a36bfc29548c39741a0ca018a5844 : SUCCESS in 51m 50s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/75bffcc5e91d46a9a3051079a93a5bc8 : SUCCESS in 57m 35s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a25bd8af29c491c908c8edb590afc19 : SUCCESS in 50m 28s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d828413998544fa992e5215ed106a25a : SUCCESS in 1h 06m 10s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/24fd6254e4d34c5ab4fc931b96919d6e : SUCCESS in 16m 36s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/c5f1fc85ec3f4f2e8370a34a05290d8b : SUCCESS in 18m 06s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/166bb1d2580e46918180c6e40dfde10e : SUCCESS in 14m 54s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd0f1ed2b9a64b8689bba8a7cfb59837 : SUCCESS in 1h 35m 38s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b0dda6f5aa6843a99f9a15b3566a9734 : SUCCESS in 18m 01s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b07055ec09b84da59964db7cd0e5e2c3 : SUCCESS in 13m 17s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/455e6ac406854f4285fd1f55b54f9f10 : SUCCESS in 40m 01s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5518ae56962408e8c28de3410dd949e : SUCCESS in 29m 22s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/97547aa764c5420bac0ff55362d5342d : SUCCESS in 13m 34s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/35f61b68d6f5432ca88747971f5fff59 : SUCCESS in 9m 54s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/92440275157b45c88dbab74b6de5b18f : SUCCESS in 16m 50s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/187fe91f918a43d382ec7d03b80438d8 : SUCCESS in 14m 03s","accounts_in_message":[],"_revision_number":1},{"id":"7213c1da6554469916641cb0f13fa0ed8bb88262","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 11:48:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3e00f706c2dac3691958a92134ad9679a506b5d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 13:50:37.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b14fc707754548f28210ca745604b42e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9460ceecf1124c6596d428d75a2a0f9b : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e15635f0088f4fbd9ce4f84fa3750c43 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21e9e35c670e4077bbeea6a95ffc6972 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fc3da4a283d4db6833070d1ed433994 : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05bb8255b32046fb841f4e4fd3a56a3c : SUCCESS in 3m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dc00c4969975406abd006ce07a6f9287 : SUCCESS in 4m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/da3bfe33c0654d9e800c5bc277da7fb5 : SUCCESS in 10m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d69d73f7a98441b3b661e2b647fab231 : SUCCESS in 16m 59s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0646c2819f2f4e708dbc5a3eb8c4c2c1 : SUCCESS in 56m 19s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b3088b148f24f369b810127de744a85 : FAILURE in 16m 03s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c4d69bbfbed043e4a1981fe1b0bd474b : SUCCESS in 1h 01m 36s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c202ca8401f047d28989bb57a2744168 : SUCCESS in 19m 05s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/999c96efbb544fada2c18c6315960264 : SUCCESS in 12m 54s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53a6af2c8490433b82bae626f33bc2ab : SUCCESS in 1h 30m 50s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/aabe937fd8bb4b3f8e6f027b1151d08c : SUCCESS in 17m 14s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1f2616cc2c54485b90235b3f14eed778 : SUCCESS in 16m 21s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3dbd784a23842a4b4325393f2d7742f : SUCCESS in 27m 18s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/944fabe624fe485b8098ca4971cbaa82 : SUCCESS in 13m 45s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a015acd2fc64108b7e5af5dd3926a67 : SUCCESS in 11m 30s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/71211b0c023342ed828807f5968557cb : SUCCESS in 15m 57s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e04f1211782445c29e6b22250d9da3fe : SUCCESS in 15m 03s","accounts_in_message":[],"_revision_number":1},{"id":"ee8861bc22c6b5bd8d764f2b2e034a826b4e6c25","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-14 11:46:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"458fe642aebb015018c13593e7bafae147fc9ae0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 13:10:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc653fea538c4f2e96de21cecfc9419b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2789eb6d21f54ac889625fe8b7f8d5b6 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38a2b54741ce4538af14282223f56fe0 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c4f7447c2ca416c98f4337173d833fb : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/96f4553765a24a2e975604b79b199b49 : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e621fab69cf04ec0b2e4407c3ebdacfa : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d43e534479bf4d9b8e186109b943c618 : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a55db0ec8b44b888837e81466a24922 : SUCCESS in 3m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/77cacf48997445ba9fd7c0f10f3a5282 : SUCCESS in 3m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/974c1694bcd64b869add135eadaea776 : SUCCESS in 9m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c004a950b84743898410d37c3eceeb7a : SUCCESS in 15m 18s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/6b1a5385c34a4e50a965e997757ba383 : SUCCESS in 58m 34s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/5cd509c5cc0b419cbfe53769a9faf930 : SUCCESS in 58m 21s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7ea2f5fabe84cc59ebfa9d0f4686f7f : SUCCESS in 50m 30s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3087378489974d7db2199a1506c0122e : SUCCESS in 51m 44s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b9c25464d69d470c984bdea5f0f91626 : SUCCESS in 17m 18s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/964c02c2d2b946bcbb016645c42d008a : SUCCESS in 16m 54s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a8545a9b03b4ce28a82ba6cafcaf870 : SUCCESS in 14m 03s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3586371b82bf42e297557f8a9fe30f79 : SUCCESS in 1h 22m 06s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7b9449e0b6444c49a50be4349e96608a : SUCCESS in 16m 40s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/42e7a593e528425696a5db5824f91ad9 : SUCCESS in 14m 02s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/7b2aee0c208b40acbe722f1dc60b32c5 : SUCCESS in 38m 19s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f420ab7824940208838fc185863e9b6 : SUCCESS in 23m 16s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c64b3eef9889479ea6a34fa8e92f0e1f : SUCCESS in 16m 04s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04cee14f97cc47cead84cb1b2d162d08 : SUCCESS in 10m 01s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e0fcefe65c4d48219ecc0aebbd5ebc9f : SUCCESS in 12m 00s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fda373f53254438d91a18675f96fb950 : SUCCESS in 13m 37s","accounts_in_message":[],"_revision_number":1},{"id":"2142bf60672ea51b73c97816a9849bfabe0c6d80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 13:10:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ec6f7cd5e2677a1e9e0e450cdfc483ea716bfbc0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 14:29:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/701e877826554bc391b8554223bddbb1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e93fd6dfb4c0447cbdf6019b37813ab5 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4423b1892e3e4010b1a79a3b97f90941 : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9db2ece9735e4740b4890733a988f303 : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/716edc9b9c8244639473ce5c817d452b : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a2bf337924e4e959f3d559ec0e7e02e : SUCCESS in 3m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/384c1c6cf19940bd806294ef77f5f4d0 : SUCCESS in 3m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5f7b3cfe3249424086f52f9dcbe5d25e : SUCCESS in 9m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e14f3c4553104a5e86bf4db1a1dd1957 : SUCCESS in 15m 07s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/eb20e30934b048129075b27ba53f6450 : SUCCESS in 48m 18s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33449ffc25de4ed4813b71b45b4a4e35 : SUCCESS in 53m 35s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/306ce5918e544418af0536c812e92fee : SUCCESS in 52m 00s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/857d219ed2454867845bf3f2721cd735 : SUCCESS in 15m 14s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5fa2b201b6bf4caeace91dc64fc5888c : SUCCESS in 13m 33s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03e2a968ee1b403da19d23d1fe6d7d5b : SUCCESS in 1h 18m 36s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/61b9d05e9a8647cda8ef83c9ea72cbdb : SUCCESS in 14m 32s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/85e6921c32574f05be6523455e038f08 : SUCCESS in 13m 45s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0dd11beb678649bc8947aa778f0f2be3 : SUCCESS in 21m 15s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b5a91d82152f4a8fbee8f86855d79383 : SUCCESS in 12m 32s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a321a8fb7f6c4a8caa518feb94ab0161 : SUCCESS in 9m 24s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d2ef96ac51124977a51bd6e6717f0855 : SUCCESS in 17m 35s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c5789b0d2e7e49069691a73f1fecc93e : SUCCESS in 11m 45s","accounts_in_message":[],"_revision_number":1},{"id":"4892a9858d6240bc570e5135fcbb4749f9845b65","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 14:29:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6594cbef12c517df44f3813ab989f684a1034e07","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 14:30:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d76ac87566bb4493874ba6db919448d9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c88ec8be4924a39bdc016776248ce7b : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55725505d8bd4494bce0ef166f8e4b19 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5bb5d076c9a35c7d3763973d1197fea5840f86cc","revisions":{"5bb5d076c9a35c7d3763973d1197fea5840f86cc":{"kind":"REWORK","_number":1,"created":"2023-11-10 10:06:11.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/85/900585/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/85/900585/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/900585/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/900585/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/900585/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/85/900585/1"}}},"commit":{"parents":[{"commit":"8663c8f92e4740fc5fc6e466fa8731b0c4dd6dfe","subject":"Merge \"Add retries to get_url and uri tasks\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8663c8f92e4740fc5fc6e466fa8731b0c4dd6dfe"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-12-20 10:54:49.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-11-10 10:06:11.000000000","tz":0},"subject":"Fix setting kolla_admin_openrc_cacert","message":"Fix setting kolla_admin_openrc_cacert\n\nKolla Ansible renamed kolla_internal_fqdn_cacert to\nkolla_admin_openrc_cacert in Victoria, after which we no longer set the\nvariable correctly in globals.yml. This would lead to a missing\nOS_CACERT in admin-openrc.sh and public-openrc.sh.\n\nThis change fixes the issue by renaming the relevant Kayobe variables to\nmatch and passing through the correct variable. Backwards compatibility\nis provided until the end of the deprecation period.\n\nkolla_public_openrc_cacert -\u003e kolla_external_fqdn_cacert\nkolla_admin_openrc_cacert -\u003e kolla_internal_fqdn_cacert\n\nStory: 2010486\nTask: 47054\n\nChange-Id: I9e1cc20579cf80525d6ef732a1aac99a65bc171b\nCo-Authored-By: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 95729405a38e6292a828c26347406e70132136b2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5bb5d076c9a35c7d3763973d1197fea5840f86cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5bb5d076c9a35c7d3763973d1197fea5840f86cc"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
