)]}'
{"id":"openstack%2Fkayobe~900587","triplet_id":"openstack%2Fkayobe~stable%2F2023.1~Ic1dd075076ea186b0928bba1a235605c0cd2ec71","project":"openstack/kayobe","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-11-10 14:18:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic1dd075076ea186b0928bba1a235605c0cd2ec71","subject":"Generate local Kolla Ansible config in check mode","status":"MERGED","created":"2023-11-10 10:13:04.000000000","updated":"2023-11-10 14:19:48.000000000","submitted":"2023-11-10 14:18:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900587","cherry_pick_of_change":869986,"cherry_pick_of_patch_set":2,"meta_rev_id":"06e570d458bed725adba145841c26c4dc393c51a","_number":900587,"virtual_id_number":900587,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-10 14:18:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2023-11-10 11:00:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2023-11-10 10:13:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2023-11-10 11:00:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-10 11:00:45.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2023-11-10 12:08:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"121d1a1d5e35821101fab0c83900819664f8c7bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-10 10:13:04.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"93c303867be26be009a0c697413e0dc121e3291b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-10 10:13:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0af1640ca5f64c20322dffccb4f2757ab93eb603","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-11-10 11:00:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d0b383f85ed954038201980d504904a43cdd4c25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 12:08:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c163ef955b78449f9cae0b0ebb7ebc55\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/546ae4c33f4e46f6bce4a6f303af45be : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c2928644acd4e13bc15e9b61e57db6f : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/494adb91edd64609b8a293767ba45a23 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2c2783723ae40d28ca79a7ffbdf7bf3 : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4664ca7c078541a2af4d7f3a040f65b3 : SUCCESS in 3m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9ca46492ab2241899e40e74ee24dafc8 : SUCCESS in 4m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/77393eef35704c5ebbdc202f5363c463 : SUCCESS in 10m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2e7f1e14a5324543bcffea4fc0cbff6d : SUCCESS in 19m 08s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/1c1683054e4643a5b083e09054fd7f73 : SUCCESS in 1h 03m 03s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/ee5f51f6238b454c9eebef8d6840df12 : SUCCESS in 1h 02m 05s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff5ea12f1fb14f76950a0452c9bf1310 : SUCCESS in 52m 36s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/b7e082391fee41469233e48172882dfc : SUCCESS in 1h 01m 48s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/864a18d2380f47aab49b2bbec75030be : SUCCESS in 16m 21s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/50e16b00a2914f9f8a98acc0c3ff145b : SUCCESS in 19m 47s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c7c89be852cd4dd6ae91310ec0d8a4c8 : SUCCESS in 14m 37s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2eb41b2633234703a1783b1a0b0b57a3 : SUCCESS in 1h 48m 10s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a3156897c2342ecbd4be456f8ad5ff0 : SUCCESS in 1h 47m 05s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/dcd62cecaa5348ec81b62e2f2ad2fa9d : SUCCESS in 16m 52s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e86c30af91b94feabd2b67645f3b4219 : SUCCESS in 14m 06s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/8314e2e049044a859da22787dc440183 : SUCCESS in 37m 01s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/563980120b6a42f68e7ad1e39d6b81e6 : SUCCESS in 30m 19s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8fa848ae4fc64c07aa82a8b99943d43e : SUCCESS in 24m 52s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/40bb214025904bc9b308e7c52f4d1d90 : SUCCESS in 12m 44s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e66d9cf73055497e8860f3e5e8414c4b : SUCCESS in 11m 43s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4429bc0d91db4bebb07543fdabfaf61e : SUCCESS in 13m 13s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dce55a6b3a3348a19510219631b5a5d9 : SUCCESS in 14m 49s","accounts_in_message":[],"_revision_number":1},{"id":"7056c3ae1144620220e12be18a32f2df1f4b449c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 12:08:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0e84e7580889c01389f6cb6116e06005b8e0043e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 14:18:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de21cf8341b041a888dd22d28e0184e6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a14527e97b2471697e69f8f16728cff : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95d937f514b24eb7bd57c71e26428fc6 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d7aca2588ff46bbaa2c3a4c23140ceb : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c849c921cd3f4d1d87ec5958bf6750fa : SUCCESS in 4m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ee688687ae1b4944bc3c950fc99544c6 : SUCCESS in 4m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fc2bef5fbd9548f1bc93ea9559f2e42b : SUCCESS in 13m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/add49a0c78ce4048a0d612ab4f9cc609 : SUCCESS in 16m 25s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0ac603f307144b4f881ce332f63ec621 : SUCCESS in 1h 05m 16s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c6f8d9320934d0086670b3413770511 : SUCCESS in 55m 46s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/455daa2602b8454fb56ece401808169d : SUCCESS in 1h 16m 20s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/40512d6251de4a578931b291d8830f3e : SUCCESS in 18m 58s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a2f9184259f84fb78f0c8035da90d5eb : SUCCESS in 12m 20s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b0fd49a9c43d485c984416a97d856337 : SUCCESS in 1h 49m 55s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/21f8e9875d7a465e8fb61070f2b7dd5d : SUCCESS in 1h 24m 27s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/8a557a70cd6147228610c42f22228253 : SUCCESS in 17m 13s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e60b8e2e48de4437ae257f2865a1cf64 : SUCCESS in 14m 48s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3edb61f3f1c04d2698b5aa1eb16a4549 : SUCCESS in 28m 39s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d895e98419834f9ca1d9557e7cbe6da2 : SUCCESS in 20m 43s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ac53cdf2d94c4e8392064fed131509ff : SUCCESS in 13m 22s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/046c01c439ad4316ac9d6922c9106037 : SUCCESS in 10m 42s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4fc4e2d0ff7647f6a4522506aaeeadb6 : SUCCESS in 18m 00s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d25c8edef81546dcbc1ac25f5c236814 : SUCCESS in 15m 06s","accounts_in_message":[],"_revision_number":1},{"id":"5859df53f7987dfa37f3133dbfffb6e80cff7588","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 14:18:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"06e570d458bed725adba145841c26c4dc393c51a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 14:19:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/111fbe5a2d594b05889c5eea8c305964\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80b12e061f5d4db48e7fee4ddb07472d : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f75540ac591d4cb585a11f902725b617 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a8be8fe89ec6e54603c632400d30b44c4a7bdd2b","revisions":{"a8be8fe89ec6e54603c632400d30b44c4a7bdd2b":{"kind":"REWORK","_number":1,"created":"2023-11-10 10:13:04.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/87/900587/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/87/900587/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/900587/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/900587/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/900587/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/87/900587/1"}}},"commit":{"parents":[{"commit":"ef5f1c97ad78170233ed18ff147ab2e1647fdb71","subject":"Merge \"Add custom user_data content configuration for Bifrost\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ef5f1c97ad78170233ed18ff147ab2e1647fdb71"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-01-12 14:15:11.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-11-10 10:13:04.000000000","tz":0},"subject":"Generate local Kolla Ansible config in check mode","message":"Generate local Kolla Ansible config in check mode\n\nIf running a command in check mode such as\n\n    kayobe overcloud service deploy --check\n\nKayobe does not generate the local configuration for Kolla Ansible. This\ncan lead to an inaccurate result when comparing with the remote\nconfiguration, if there are changes in kayobe-config.\n\nFor example:\n\n* Run kayobe overcloud service deploy\n* Change a file in etc/kayobe/kolla\n* Run kayobe overcloud service deploy --check --diff\n\nWe would expect that the changed file results in a diff against the\nremote config. However there is no diff displayed.\n\nThis change fixes the issue by always generating the local Kolla Ansible\nconfig, even in check mode.\n\nChange-Id: Ic1dd075076ea186b0928bba1a235605c0cd2ec71\nStory: 2010526\nTask: 47132\n(cherry picked from commit 2c0f705e7acab5691591c215e40dea707d12da15)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a8be8fe89ec6e54603c632400d30b44c4a7bdd2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a8be8fe89ec6e54603c632400d30b44c4a7bdd2b"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
