)]}'
{"id":"openstack%2Fkayobe~900591","triplet_id":"openstack%2Fkayobe~stable%2Fzed~I3a5a43fef9c3d68d0db02be12b9f892c437e513d","project":"openstack/kayobe","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-11-10 14:10:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3a5a43fef9c3d68d0db02be12b9f892c437e513d","subject":"dev: Improve error checking for config check functions","status":"MERGED","created":"2023-11-10 10:14:37.000000000","updated":"2023-11-10 14:11:43.000000000","submitted":"2023-11-10 14:10:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900591","cherry_pick_of_change":884202,"cherry_pick_of_patch_set":1,"meta_rev_id":"b5c21bff11c4bc49ba54207df8257346b67a75ad","_number":900591,"virtual_id_number":900591,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-10 14:10:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2023-11-10 11:02:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2023-11-10 10:14:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2023-11-10 11:02:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-10 11:02:51.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2023-11-10 11:54:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9b0d38cf2f0e7a56125928e1fb77ec2bc8433760","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-10 10:14:37.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"4fda529c726dca7bb896419b249879bb9e64effc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-10 10:14:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7a07a8e50ea9b57a460c090c08a21758dc960f9c","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-11-10 11:02:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"907059c63fde5d814d78e8aed99712803961e009","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 11:54:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e5d038bab004b149b48612c5abb33bc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de8434fa18824b9986867249e0d937cc : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50916402cf384270b5c91927af4ff77c : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7659d9fca75414ba83f7dc74b7de64e : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a37db0017ac54a12beabc6e0f7bce247 : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/34904b1b28ae49e6ab8c77022e69e8a8 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e66e7e9b2244fe4946c6ca55bb8c632 : SUCCESS in 7m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4e8df8831e0940a1b7548dcf6d970af0 : SUCCESS in 3m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3a501bef62624e5aad4a1bfb4c2eb80a : SUCCESS in 10m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/91170d4a88df4d4f8796338fc66462e0 : SUCCESS in 15m 30s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ad5b7e4df62a43edba82004191fc930c : SUCCESS in 48m 53s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/f11363f4fb6346ddb864478ec4df43c7 : SUCCESS in 52m 57s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4f828ff5b0d04841ac33dd3f21c649a4 : SUCCESS in 50m 14s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/b14195f7d33c41208fdd2e3fc7f31d81 : SUCCESS in 52m 43s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/16c6d107338e4c27b405c612c3296b58 : SUCCESS in 18m 07s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/7195e52eed2e49d188e9594520f23012 : SUCCESS in 18m 16s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/758089ad27bc4ed9a330ddc600b3bdd6 : SUCCESS in 14m 32s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab53faaf072742ca8117e591ddde1ff6 : SUCCESS in 1h 23m 54s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c3648c3e76d64efeb2cb02cfe5dd394c : SUCCESS in 13m 26s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c8712f5399f48af9288ade7bc046ae7 : SUCCESS in 18m 37s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/20026e0b1ec1465a8c828a4d90277eca : SUCCESS in 33m 15s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e27f6e47fdc4d5792db2697117f1910 : SUCCESS in 27m 17s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/07f775c0ec5245c0aed48a7bf04a07f3 : SUCCESS in 10m 19s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d6e7c4f51d434c97b3e9891241393b3b : SUCCESS in 9m 13s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/63276e9c089847dd85478ea688e7c59e : SUCCESS in 13m 20s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fb279a26dd4f4c88a88c5806df99b2fc : SUCCESS in 13m 36s","accounts_in_message":[],"_revision_number":1},{"id":"91d77d822b3e675502a0752723a5cf87723c6070","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 11:55:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5e4ca7df9b5e6977e125131cb27089fe58a0685a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 14:10:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9baa3a4d81dc44d9936158e0b1028c00\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a887d3347754ef5a7c00e2d1ad33ea2 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46db39eeace2446ca145f18d980469f9 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75292bfece574d4182acd53f58af21e8 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66d1348811734bf3ba62ada9ba26c146 : SUCCESS in 7m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1ef8f49436ac4a36bdd1fb44287c702f : SUCCESS in 4m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/30f48fc25a694ef8b3195adc92482430 : SUCCESS in 10m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cb1329c5829245c18ff2557a79b4f329 : SUCCESS in 15m 20s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/6469f9d9926145b88222aa04e2269e82 : SUCCESS in 39m 20s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e2b47cdb39d448878a6d25e0525eccd1 : SUCCESS in 49m 54s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/bef18c29e2fc4ba29ba5ca5362eb859a : SUCCESS in 1h 23m 08s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/50c64ea2a6684512a58fb75dcece72a2 : SUCCESS in 18m 31s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3086b426d43345c7af83ef08fff7a5f4 : SUCCESS in 14m 00s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8100307d4976477a902280807343e87c : SUCCESS in 1h 33m 56s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/554435aab2d546ba9f9397a19661bb07 : SUCCESS in 16m 09s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f37d5e93d664e7d83a8c57f4f4b7ab8 : SUCCESS in 15m 52s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba9f9647963f484487e461933fb125c4 : SUCCESS in 28m 57s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6f4acfb52a06474ab9dc40090125e7ea : SUCCESS in 14m 01s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/257202d914094673a03a51c66474097a : SUCCESS in 10m 57s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fd688becaae042af816ff7ebee9ac56e : SUCCESS in 14m 02s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/485436847c7940d787fa36529023090d : SUCCESS in 13m 31s","accounts_in_message":[],"_revision_number":1},{"id":"cd4ed4f24a04a95f2e635165be8554c458f52fc4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 14:10:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b5c21bff11c4bc49ba54207df8257346b67a75ad","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 14:11:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6815eb75566945a2bce4a901955f41ff\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99578ea8adfe47f6ac4676a2f5156fda : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c33dcfbee4db715a175ccce4d928036f0c40c94e","revisions":{"c33dcfbee4db715a175ccce4d928036f0c40c94e":{"kind":"REWORK","_number":1,"created":"2023-11-10 10:14:37.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/91/900591/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/91/900591/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/900591/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/900591/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/900591/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/91/900591/1"}}},"commit":{"parents":[{"commit":"8663c8f92e4740fc5fc6e466fa8731b0c4dd6dfe","subject":"Merge \"Add retries to get_url and uri tasks\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8663c8f92e4740fc5fc6e466fa8731b0c4dd6dfe"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-05-24 13:44:55.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-11-10 10:14:37.000000000","tz":0},"subject":"dev: Improve error checking for config check functions","message":"dev: Improve error checking for config check functions\n\nVarious functions in the development/testing scripts rely on \u0027kayobe\nconfiguration dump\u0027 to extract the value of flags. If this command fails\nfor any reason, we should exit the script. Currently, some places we\ncontinue and return 1, since we check the output against the string\n\u0027true\u0027.\n\nThe to_bool helper function handles failure by checking for a valid\nboolean output, so let\u0027s use that everywhere.\n\nChange-Id: I3a5a43fef9c3d68d0db02be12b9f892c437e513d\n(cherry picked from commit 89fc4fa2792ef256ff7624559a8ea8e2ee2f13fc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c33dcfbee4db715a175ccce4d928036f0c40c94e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c33dcfbee4db715a175ccce4d928036f0c40c94e"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
