)]}'
{"id":"openstack%2Fkayobe~900592","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~I3a5a43fef9c3d68d0db02be12b9f892c437e513d","project":"openstack/kayobe","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-11-10 14:00:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3a5a43fef9c3d68d0db02be12b9f892c437e513d","subject":"dev: Improve error checking for config check functions","status":"MERGED","created":"2023-11-10 10:14:59.000000000","updated":"2023-11-10 14:01:52.000000000","submitted":"2023-11-10 14:00:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900592","cherry_pick_of_change":884202,"cherry_pick_of_patch_set":1,"meta_rev_id":"66879abbc5bd5f1ca3ddecb0d77a83cce435bb69","_number":900592,"virtual_id_number":900592,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-10 14:00:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2023-11-10 11:03:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2023-11-10 10:15:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2023-11-10 11:03:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-10 11:03:01.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2023-11-10 11:53:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1ed93f4158ea6eada75f7fb9bb7b3fa2f11f9c2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-10 10:14:59.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"bfcf7419b2a9f9941b3f6cd9f4b40f1232efb22b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-10 10:15:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0a0c15390e5a934547127abb5a83bdd85e028337","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-11-10 11:03:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e03d755ebe129fda5666115fa130d6f1047669aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 11:53:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3383f4ef27f246c0a077c11279f36b4b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02f9052e8e844e4fa8d939b430665321 : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f4cfb062fbc4cee82c2d5dc3ad84b9d : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/275a4e62802546f0936b4e60427e89cc : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d79c3fc1246e4e50bec145532de302dc : SUCCESS in 6m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2e71ce31a2d43b3b26154962221cab6 : SUCCESS in 9m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e4d3cde59f474fa69dde8ae9a58c75b3 : SUCCESS in 7m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ee53cacb143444d79135276926142296 : SUCCESS in 11m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/73f353b7ee424f25b829700e915492e3 : SUCCESS in 27m 46s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e658abbf7f61480bb601124620ca093b : SUCCESS in 1h 06m 54s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/1e59148ee38345cbabff85c7d0ade414 : SUCCESS in 54m 39s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/4325a2dabd434a36be32630d44016043 : SUCCESS in 1h 05m 54s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/fc9f87adeed04620ad3a2a7959856f55 : SUCCESS in 47m 32s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc8c9b2473ba4ef2b1e5061bbc8dccce : SUCCESS in 49m 17s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be627d5d879c4e1787fd98cbd06ac6a1 : SUCCESS in 51m 33s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/133454db72e84cff8e6f284c271be88a : SUCCESS in 24m 07s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/a025e68df60a4b4b92e8ceceb9e1df49 : SUCCESS in 18m 44s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/98b1b1ab218348149fe7ee79e9faac7a : SUCCESS in 21m 17s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e04fc7f66e6d4ee29a00b11356f6d628 : SUCCESS in 21m 10s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c20430af89d04556996c78bf6d5baeb2 : SUCCESS in 15m 14s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fb6e93c7e3e94ece8bcd4e568465a792 : SUCCESS in 14m 38s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/822b20d018e14fb8bf9197a520ae10e8 : SUCCESS in 1h 17m 13s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/507daa95d6ff4c90a4592786ad81f472 : SUCCESS in 1h 00m 24s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7e1b2a2ae64447259d0fb60e5f165e05 : SUCCESS in 24m 07s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/0928bd9c5a53414eb8d0f1f29d583d50 : SUCCESS in 21m 40s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5b0c3e79b5154fc7b307479be7405081 : SUCCESS in 20m 19s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2c77a085af64e3bbc91ca65f00ecaa9 : SUCCESS in 17m 16s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e02e35b0052e46239a18f8a34659f071 : SUCCESS in 16m 11s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/671bc606f70740d6b410e0944308e2bd : SUCCESS in 19m 55s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/e72deaea065f4a3bb4266123bdb21b07 : SUCCESS in 15m 27s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ae97fbf09d374d6a983fd2de2cf37d48 : SUCCESS in 17m 25s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61e6ad52c7ea4b4ca62dadab3b5428ba : SUCCESS in 12m 26s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/58dedbaffa24466d87aaae9a0e019d5d : SUCCESS in 12m 19s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/236e993ec82041e5bc272b1e92715b9e : SUCCESS in 20m 35s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/0178a396904e408bb717f329eebecb54 : SUCCESS in 22m 09s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2713f42b217b48ff979501b8c5e39398 : SUCCESS in 17m 53s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7eca450a7a494fa08a19782e1e840eb2 : SUCCESS in 12m 37s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/92f74f86b87c4117a715a62d7d221ce7 : SUCCESS in 13m 40s","accounts_in_message":[],"_revision_number":1},{"id":"e3ae5f30e2d6d33688b991dffaa104c7bfd7a734","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 11:53:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5c994d100ed5c5a2c77853307f49d10a5f6edabe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 14:00:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c84d77f1fb2421198841aad39f8ba98\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95ea7e4c62b5439f8eea1511fe5516b3 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5467412df7ef4af2aef3577e0c8dd902 : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e87fc588e459484797fa9060124275fc : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c65767beff1240538eda368e376d3c43 : SUCCESS in 8m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0f9c97555a514bce9313d1eb58d836ed : SUCCESS in 5m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9494d16f58e940af9c4d5673fe90daa7 : SUCCESS in 10m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/25281b8c241e42dba8a5c7f4c198ab73 : SUCCESS in 16m 16s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2efe353cbbf24da39687862616575016 : SUCCESS in 55m 09s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/cfcebb67ed204051a9fe5020b8fd983f : SUCCESS in 1h 10m 51s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/cfc552a8feb4479c85ffec2350810c13 : SUCCESS in 37m 44s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca69d14f3c1c4485bc868ffdc109302b : SUCCESS in 42m 14s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff192e18799f4930980b6589cd0cbb00 : SUCCESS in 50m 29s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c1925571d79342b882791f905e3eb8ac : SUCCESS in 26m 15s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/b01418607aca4b5492be036ef72750f5 : SUCCESS in 20m 28s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/fd7d6406390e43f2bfe4ad8d9848f19c : SUCCESS in 20m 36s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6842a3632a6447bfae64bc327245a40b : SUCCESS in 13m 15s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b051073b12ab431c9aab17e7680515f7 : SUCCESS in 15m 14s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9945ac0847d7409ba66641cc911ce2fd : SUCCESS in 1h 15m 47s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/f24926ae2d6a4fcca87b8f4d7bbae9d2 : SUCCESS in 1h 04m 18s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8b6a7610149040d488c974482acf7190 : SUCCESS in 21m 54s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/52681d0b4c164310965f1f6f081d99f8 : SUCCESS in 18m 57s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b10a9b3530bb433f81eb8a179c069fc7 : SUCCESS in 22m 03s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/767aa67d476c420f959cf19aff3c7f9d : SUCCESS in 17m 20s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d33660a9e37f4c38be7d047fcb8298fc : SUCCESS in 18m 00s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/49a331a0a0774357a16a945d842269aa : SUCCESS in 18m 23s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/0a4a203803c44fe2ac6f16d705b1e8a7 : SUCCESS in 16m 24s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/cea9e27920ae4a8fba166f572e6c02f2 : SUCCESS in 17m 51s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90058ec327b547589705103f1d5d4d0e : SUCCESS in 13m 45s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a03acb99c8c94add9e179e513bb9317f : SUCCESS in 13m 14s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/69fa9227107d40599c90f02e58d64da6 : SUCCESS in 22m 05s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/25e6b02d78844d1e9c9e5e71eda5aa93 : SUCCESS in 20m 55s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2fac6ff9ee5140e0b7918a4613436549 : SUCCESS in 17m 20s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f4fe3dc1aff42a390c4d58f567a37c7 : SUCCESS in 15m 03s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/25f366a3134b4c40b75c9f218a914770 : SUCCESS in 15m 38s","accounts_in_message":[],"_revision_number":1},{"id":"d97634cf079e18abca52a8b4781b5429f8ac5e85","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 14:00:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"66879abbc5bd5f1ca3ddecb0d77a83cce435bb69","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 14:01:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d351fa57faeb4330bf279d608909de24\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ef59e686d364badbdd781f4e67fad9e : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"89994fe7fa1d06401e4aef846f667dbb0000fc5c","revisions":{"89994fe7fa1d06401e4aef846f667dbb0000fc5c":{"kind":"REWORK","_number":1,"created":"2023-11-10 10:14:59.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/92/900592/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/92/900592/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/900592/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/900592/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/900592/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/92/900592/1"}}},"commit":{"parents":[{"commit":"49ca368660ea01a24afc13350c62273e6f1acdbb","subject":"Merge \"Add retries to get_url and uri tasks\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49ca368660ea01a24afc13350c62273e6f1acdbb"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-05-24 13:44:55.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-11-10 10:14:59.000000000","tz":0},"subject":"dev: Improve error checking for config check functions","message":"dev: Improve error checking for config check functions\n\nVarious functions in the development/testing scripts rely on \u0027kayobe\nconfiguration dump\u0027 to extract the value of flags. If this command fails\nfor any reason, we should exit the script. Currently, some places we\ncontinue and return 1, since we check the output against the string\n\u0027true\u0027.\n\nThe to_bool helper function handles failure by checking for a valid\nboolean output, so let\u0027s use that everywhere.\n\nChange-Id: I3a5a43fef9c3d68d0db02be12b9f892c437e513d\n(cherry picked from commit 89fc4fa2792ef256ff7624559a8ea8e2ee2f13fc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/89994fe7fa1d06401e4aef846f667dbb0000fc5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/89994fe7fa1d06401e4aef846f667dbb0000fc5c"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
