)]}'
{"id":"openstack%2Fkayobe~900593","triplet_id":"openstack%2Fkayobe~stable%2F2023.1~Ia2ba7414202de6750b18ada47e7a5bb0b636e7b3","project":"openstack/kayobe","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-11-10 14:19:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia2ba7414202de6750b18ada47e7a5bb0b636e7b3","subject":"Improve neutron images regex","status":"MERGED","created":"2023-11-10 10:16:56.000000000","updated":"2023-11-10 14:21:08.000000000","submitted":"2023-11-10 14:19:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900593","cherry_pick_of_change":899915,"cherry_pick_of_patch_set":1,"meta_rev_id":"fa95b26688ae7afae8f921bb79446e15f932f9c6","_number":900593,"virtual_id_number":900593,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-10 14:19:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-11-10 10:18:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-11-10 10:17:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-11-10 10:18:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-10 10:16:56.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2023-11-10 10:18:24.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-11-10 12:08:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"84b18cc72fbaa0fe6c9b34cb5e604d8c28c37103","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-10 10:16:56.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"35712b80127ee859234c859968658e72c92090ef","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-10 10:17:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9456ab0ef738ec9390905aa26f4a20d105181b8b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-10 10:18:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ff3d212bd293045a67da4cfe6c8e8381f4c86332","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 12:08:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fc7540299ca4811be07fc0ffdcb9b71\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d3181f72bf340c1829361a00c14021b : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4f675dba3d448a8af9e4aa6685745e8 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19bf271ef71542f1b43537788d0adc5c : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3d791861c0a4079a12330a645574900 : SUCCESS in 7m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f286cd49916d487d8df9327969a238a2 : SUCCESS in 4m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9a31f8e4161d4498ad6103992a722848 : SUCCESS in 11m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3f04d9fc5a1846089e5028a95aba016a : SUCCESS in 21m 48s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/01e6065944e1499c953c4933aa9447e9 : SUCCESS in 1h 04m 23s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/b3f5baaa5232421cbb44506cce086097 : SUCCESS in 1h 07m 30s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/42b3dd8d82294d058faa6be66a632f62 : SUCCESS in 57m 58s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/da4443bdb56048699e25d938dd85216f : SUCCESS in 1h 04m 50s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3c8e2c9f59bc4467b0629c044f9bd80c : SUCCESS in 15m 48s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/03eb9bcea2024c00bccfd0b3516bdbd3 : SUCCESS in 17m 01s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e23f69a7ba0b4fda87a78655456c0537 : SUCCESS in 13m 15s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b338111af940494eaf012ff1fcd6ef69 : SUCCESS in 1h 36m 58s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a669fdc21b234eac8813b9e80fb1cbed : SUCCESS in 1h 40m 05s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/1a5a50343d89432f978e1fb40fb2dbfa : SUCCESS in 15m 44s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e15ef5a364ba4f4ca72fa2f30c85d028 : SUCCESS in 12m 27s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/b4f830ddf1f64051899585b394ce18b2 : SUCCESS in 37m 26s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/fd1e7aaf753341519eef53e2d9aa13b1 : SUCCESS in 27m 33s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b1f48b51492432dbad380f31ec51412 : SUCCESS in 27m 33s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/440d7ee5a9ce49d3886d8b133a12b44b : SUCCESS in 12m 41s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/896cca2c8f9049b0988eb7abbc2540aa : SUCCESS in 10m 16s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a711b7e7b4fc4385aa1875061db77ce6 : SUCCESS in 20m 05s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/871ef150e5ed42ff9fe6187499595e66 : SUCCESS in 13m 19s","accounts_in_message":[],"_revision_number":1},{"id":"17af99fbe46d9b6f90d12d69ac7067c06481e1af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 12:08:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5656af657ad327ee09658f5c8f60d8358623cf56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 14:19:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e5857b80ff6403ca7b998805954297e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/175e3b4f5ba244b29619434610172b82 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95cdb65981844bd7a4c581087629b960 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6369c715935d4ca58cd968b9c68da6d4 : SUCCESS in 7m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fd4dead0cdee422282682b64866f9097 : SUCCESS in 4m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fc7332fae40a4b83ac54eb4163d103c1 : SUCCESS in 11m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c2900a4d84044d0b8887a7ce292b642c : SUCCESS in 15m 11s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/29a3dc12de7b44d9ad033d14ccab8299 : SUCCESS in 1h 08m 59s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1fd61cb37bc748e09df82e396d784053 : SUCCESS in 46m 02s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/280fa9031a0b4720a2b567541545a657 : SUCCESS in 1h 03m 09s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ef62235a905b41b4ba445bdefdcd2c19 : SUCCESS in 16m 53s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc1ed579baa341dbaed98ef52ed18e6a : SUCCESS in 13m 37s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/599a5b00602d4e8e8400247cca4356ae : SUCCESS in 1h 46m 48s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc83e8714a5e420bacb7db4faa1a7c30 : SUCCESS in 1h 35m 04s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d08c0cc18a50424b8315e9bd800d7f54 : SUCCESS in 13m 32s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eaf1878f4a63407496b00f1fd6d691ed : SUCCESS in 14m 40s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/65182f6e1d90426ab5145dfa166068a6 : SUCCESS in 29m 32s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc37a49ac6f544fba326ad7d530e4b2f : SUCCESS in 24m 39s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/58d464a2cdf34a26855695a2fffc31ea : SUCCESS in 11m 04s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d792c3a6188a4dfaa77366f4e9edcf93 : SUCCESS in 11m 48s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b6212e87c2124da7990dc01b41b1c37f : SUCCESS in 17m 17s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/27f0c669929841dba2f9786a08304450 : SUCCESS in 12m 15s","accounts_in_message":[],"_revision_number":1},{"id":"44d79a397411bdee9be2d9c085bb78c8c6bf86ec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 14:19:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fa95b26688ae7afae8f921bb79446e15f932f9c6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 14:21:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/158d6c4ceef743f1b191a86098ba868f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a79e4fddb5224236a0fb2d1dbc7b6bed : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"be19765a3560b908041b3da0d323f0b73c1f7c86","revisions":{"be19765a3560b908041b3da0d323f0b73c1f7c86":{"kind":"REWORK","_number":1,"created":"2023-11-10 10:16:56.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/93/900593/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/93/900593/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/900593/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/900593/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/900593/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/93/900593/1"}}},"commit":{"parents":[{"commit":"ef5f1c97ad78170233ed18ff147ab2e1647fdb71","subject":"Merge \"Add custom user_data content configuration for Bifrost\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ef5f1c97ad78170233ed18ff147ab2e1647fdb71"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-11-02 06:43:22.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-11-10 10:16:56.000000000","tz":0},"subject":"Improve neutron images regex","message":"Improve neutron images regex\n\nMatch ML2/OVS agents only when OVN is not enabled.\n\nChange-Id: Ia2ba7414202de6750b18ada47e7a5bb0b636e7b3\n(cherry picked from commit d9ee59ea1bdf21f50f5798278b5f6c75bf638a04)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/be19765a3560b908041b3da0d323f0b73c1f7c86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/be19765a3560b908041b3da0d323f0b73c1f7c86"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
