)]}'
{"id":"openstack%2Fkayobe~901060","triplet_id":"openstack%2Fkayobe~master~Id80e7d28c20b1f9ecc52473013c78c28d2ab5532","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-11-17 01:30:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id80e7d28c20b1f9ecc52473013c78c28d2ab5532","subject":"Drop not used grafana-conf from reqs","status":"MERGED","created":"2023-11-15 17:56:21.000000000","updated":"2023-11-17 01:31:20.000000000","submitted":"2023-11-17 01:30:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901060","meta_rev_id":"d6628b33ecaf20f623660d80343e1c1f79111d17","_number":901060,"virtual_id_number":901060,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-17 01:30:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2023-11-16 09:39:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-16 08:48:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2023-11-16 09:39:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-16 20:17:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-15 20:24:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-16 08:48:04.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2023-11-16 09:39:13.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"6491dfb567df79dc505d6699af5ceb5c966ae889","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-15 17:56:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47786c6d2c833a510645e394dc02fe8134680d5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 20:24:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f818499f0fa343fcb84e947d284c7974\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/994c1b789f6e408a958484e278af9bd4 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bc6818228c14b759309458319518dcf : SUCCESS in 8m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bdd74be26766435e84d3ca23ace7964b : SUCCESS in 5m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b2bae6b30db84b028ee6dcea55b0110d : SUCCESS in 12m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f2b9781ea4ee47fb84779fbd500bd128 : SUCCESS in 10m 58s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/1ee950e4de64414ab035a151fa810fd8 : SUCCESS in 1h 04m 47s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/b0e71ca5a46a4247a666fb050ba4481e : SUCCESS in 1h 05m 21s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ec0a9c75932a4a0a86553f1e270e95fc : SUCCESS in 1h 00m 21s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/35ddbf4f5bf14169ac3835e047ba33a8 : SUCCESS in 1h 11m 36s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/344e8ae733424d1a9dbcb74a2c169ebc : SUCCESS in 18m 07s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/1f15870dc0a546fca4df2c38bc92263e : SUCCESS in 17m 51s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0de86091e6ae4e0ba7830b9a2340aca5 : SUCCESS in 15m 07s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/bd1e4a5439e54c80a21ccc430617556b : SUCCESS in 2h 25m 39s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e07452f22c4d43578db618d0c63574ad : SUCCESS in 2h 21m 20s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/95a7d63c940d40158e34776ef73cfe62 : SUCCESS in 17m 17s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e401b9cae64145b2bd603c74bbcd3827 : SUCCESS in 11m 40s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/c7fc2badde0846628ccc3bd30dae0d34 : SUCCESS in 29m 17s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9e9f7d907d474a5887ea1b0abb931525 : SUCCESS in 30m 00s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7aaa49fd1deb4dd78ef1137c87bf4473 : SUCCESS in 24m 05s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/48cc64340cf444bdbacec3bca42e1673 : SUCCESS in 14m 21s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/554920c3739b4773a91c15951eba1575 : SUCCESS in 10m 16s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/144aaa1444fb4af0a5345bbf005568e4 : SUCCESS in 17m 55s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db767e63ddcc4e67956a40e869e00678 : SUCCESS in 13m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9804322a7c54f0d88ba4ddd135d5bb2 : SUCCESS in 5m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a02a9b3df0747d78eefb1d6bafa368d : SUCCESS in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6083dd1aab0a481eb8b739720a103ea2 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":1},{"id":"4ea869e44789859b9b9244eff4f19cb3e803e112","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-11-16 08:48:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"03d3f6440f27d8f69197b035c62774643bd19818","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-16 09:39:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"981fbb2bfe0d61b6321d3c2ef4ccf2ee5677ee68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 10:05:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"350cfe3668b853c20747ca67114571ef82651549","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 13:14:45.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/901059 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"2de9243bc46b058e5fc07e9490220a7596bd5bd4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-11-16 20:17:41.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"985f2a029994a49d0abde453deddd1aa0b2a9faa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 22:53:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6c03e46b0e9459e97ba7afc87ac98eb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e692e9b09d044fdf8dc5711d5304799d : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8c49aaefc684da3b9308003117969e8 : SUCCESS in 8m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2fc88cc92e3d4d0bb1f6ed9b0a8bf3e3 : SUCCESS in 4m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/88c1110e853d48adbdfa6439a7fef0c1 : SUCCESS in 9m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d0913235ad7743e09161716585d105d2 : SUCCESS in 10m 25s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/028f1e9f06d94948a9ec827e7af872be : SUCCESS in 1h 05m 18s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/665ba3e1750e49048219188cae6a66e4 : SUCCESS in 1h 08m 49s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82706dfd630f48c4852abd5c4aa8ad73 : SUCCESS in 57m 04s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8367df6c99d4464a98451c18f5696a05 : SUCCESS in 1h 13m 31s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7e6bd62233ad4f47b9388bbf9f8a9f07 : SUCCESS in 15m 38s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/543324fc148541698c101e472061bbb8 : SUCCESS in 18m 17s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/802311c5d64746fa82549b635134af5f : SUCCESS in 14m 03s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/53a7bd50f06f4574a604180f505b9ea3 : SUCCESS in 2h 30m 51s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/41d64d5cb3e44da79e36ea363363e3c1 : SUCCESS in 2h 14m 18s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/55b88b941bd241189c8326b8a7983cb8 : SUCCESS in 18m 32s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a135a2815934ae3b0a9e4578af0074c : SUCCESS in 13m 36s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/877bcb94f3b14398a8ade3a5617af2b8 : SUCCESS in 34m 10s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b768c19aead1412781fe9f1ca0058c5f : SUCCESS in 24m 44s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/813ba479c6f342319986f7eb4364cea2 : SUCCESS in 26m 54s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/956c476672224e1cbca3b527bbce8280 : SUCCESS in 9m 43s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b9bcb3e0e4c049e89323318ff1d66830 : SUCCESS in 10m 36s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5f41b9dbf67442db8bd7279f91b4b096 : SUCCESS in 19m 22s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2304dfa8dde74aa588f311371532e34e : SUCCESS in 11m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b72dac8f2a114a808d3a4e34e6cae4d8 : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/799e499369c447af9afc97b12c4e441e : SUCCESS in 4m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f739fb749bf146a38cae3bc50e8717f8 : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":1},{"id":"2f6a60926f2f943c8d84fda3fdbddd5033088dbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 22:53:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3c1ecb901f9a0955491bb5d5a2bfdab63765535f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 01:30:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38aa57a8e76d491492f2336a5d6a6801\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bf7b240d1ac408f9d15452c65c6dadb : SUCCESS in 7m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a4dbc268baad4a3ebd2192fe055dd1d2 : SUCCESS in 5m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1a887bdc2a1943ff83404ae5a5922e89 : SUCCESS in 11m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a1f40b970954451286f4f7431991064d : SUCCESS in 8m 31s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/9459abe5702042ae999b4cc0ed69a0c7 : SUCCESS in 47m 51s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd585b2380ba430ab87f4b7ea1998d5f : SUCCESS in 43m 08s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/97f96dc57eb0471185488ed8509b95ef : SUCCESS in 57m 58s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/bd7e829e7c8c439db44a97342bfb3bc1 : SUCCESS in 16m 01s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa71904e38d74b0082e4e82041bdf63b : SUCCESS in 11m 36s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/09d7aadd2ae24e26a6f2d3fb81abf423 : SUCCESS in 2h 32m 28s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65c97c0085644b11b5c7532c0e584e86 : SUCCESS in 1h 56m 30s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/df1f69a93c53449fb2c6ff9454469f5e : SUCCESS in 17m 06s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e94a864c5fce4a92957f4edfaa617561 : SUCCESS in 11m 44s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/95d7ec15163d44bcb1f592b08bee97ae : SUCCESS in 31m 53s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8bc36b35544d469381a3df7cf51c5d25 : SUCCESS in 19m 59s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/442c75ec63e84f36b4503ee9720e328a : SUCCESS in 14m 17s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da8961fb3a7e4f0abb9479dbb2c76145 : SUCCESS in 9m 09s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/06324763a5654c68a418e3f022679a55 : SUCCESS in 18m 48s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c81b2b81a2f04e1ba94bd37d9486cc42 : SUCCESS in 12m 20s","accounts_in_message":[],"_revision_number":1},{"id":"d627aa180dd28b82cf322358a68816a6a4e3173c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 01:30:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d6628b33ecaf20f623660d80343e1c1f79111d17","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 01:31:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38ef35889aeb4a29b2898fb924dec2d8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76c77ed38cfd4893aded73a83cecfcff : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"351e6b317304336a626a53ecd6a4aa9a13b3e275","revisions":{"351e6b317304336a626a53ecd6a4aa9a13b3e275":{"kind":"REWORK","_number":1,"created":"2023-11-15 17:56:21.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/901060/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/901060/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/901060/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/901060/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/901060/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/901060/1"}}},"commit":{"parents":[{"commit":"beb6ec7903bff734c4a55a47379630a07c57d1e3","subject":"Remove Monasca and co remnants","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/beb6ec7903bff734c4a55a47379630a07c57d1e3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-11-15 17:56:02.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-11-15 17:56:02.000000000","tz":60},"subject":"Drop not used grafana-conf from reqs","message":"Drop not used grafana-conf from reqs\n\nChange-Id: Id80e7d28c20b1f9ecc52473013c78c28d2ab5532\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/351e6b317304336a626a53ecd6a4aa9a13b3e275"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/351e6b317304336a626a53ecd6a4aa9a13b3e275"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
